)]}'
{"id":"openstack%2Ftacker-horizon~871222","triplet_id":"openstack%2Ftacker-horizon~master~Idea2712e701488575c3f418f11ecae12516fdc68","project":"openstack/tacker-horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-01-23 00:18:46.000000000","reason":"Change was submitted"},"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2023-01-23 00:00:21.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2023-01-20 18:51:40.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"32102":{"account":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"last_update":"2023-01-20 12:07:09.000000000","reason":"\u003cGERRIT_ACCOUNT_32102\u003e replied on the change","reason_account":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}}},"hashtags":[],"change_id":"Idea2712e701488575c3f418f11ecae12516fdc68","subject":"Address RemovedInDjango40Warning","status":"MERGED","created":"2023-01-20 06:57:06.000000000","updated":"2023-01-23 00:19:48.000000000","submitted":"2023-01-23 00:18:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871222","meta_rev_id":"6828d7bb10e87ad11089ada26f2919f7aa25a30b","_number":871222,"virtual_id_number":871222,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-23 00:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2023-01-23 00:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2023-01-20 18:51:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2023-01-20 12:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":1,"date":"2023-01-23 00:00:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-20 08:30:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-20 09:58:27.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2023-01-20 09:58:27.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2023-01-20 09:58:27.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"}],"messages":[{"id":"894b5e763a60577393111255b7ef6160121b0aab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-20 06:57:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87ec60c9b8ee4b9abeec7849d85727ee654a5b2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 08:30:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96b1d825d2864d28b9b93e6c6502ddbe\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8864fa291e12424c8fdae267698ee3eb : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73779b9ba92c4b1ea3fbfed354f86abc : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a55e093a8ef04e4b93b30d7334e8aa6e : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e29ec24263174fed8d1f86ab6950b27f : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b540068d3f6844b79503b54be1e8b4f5 : SUCCESS in 5m 46s\n- tacker-horizon-integration-tests https://zuul.opendev.org/t/openstack/build/6f1d9bd2204e45db92e6d08e4151f662 : RETRY_LIMIT in 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b178b65e88e5784f4f9601096e228067b696695f","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2023-01-20 10:03:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c25f2121c6bfbb3f208e367a989282026d2ee2f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-20 10:29:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc27879b55f57bcdad89825c0e741a13bf1e91d8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-20 10:29:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc0bf7c0f44ce4eb8d3c382ed6388efe25de65d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 11:52:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c67f9f974be443f18764ca9f4e05d4fa\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ba4079b37a140579e45c43eefec3633 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/870594c5127548a297bf90dcf406b30c : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c06cdafdb214ae491f057bd4897e66f : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fa707c515e14e57965aabcb5326c10b : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e21b1073a6a4cc5b417d7e4fd3780a9 : SUCCESS in 6m 15s\n- tacker-horizon-integration-tests https://zuul.opendev.org/t/openstack/build/824de4d83c7547129ad0e9ab3170ecf2 : RETRY_LIMIT in 33m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1cc3510d49105c360042bbaf6fea08f086144c9","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2023-01-20 12:07:09.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7e186ac9580006b1d1dc8d35d02bb107cdc2ec8","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-01-20 18:51:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0884c1faefefd745a25ed5b5faf6804431ce3076","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-23 00:00:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4917f4fa81173d0b4877760930f97f600dc063b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 00:00:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fff8b7e324827685639f33626bbc68b67fe572b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 00:18:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70680b2341794e26a04b409b45b6c94c\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/510a610cf8284fada4eb6fac355a365f : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b55cfc10a646539ab5bc26090a5efb : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b1e30cffd554331a1e7abcf44504942 : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49c339061312412f945c0fd63e0c04f3 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4d2c61d4814b87aa9cb73cc20defc6 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8c0a704a02e71c277601f1520a76d0a3ff177571","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 00:18:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6828d7bb10e87ad11089ada26f2919f7aa25a30b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 00:19:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d00c09f37541a7813306565eb1bb8a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b146b78663f24f3585cffc50f246abee : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef2bebf58ede4dc53d0b2adf7fe35a389405763f","revisions":{"5ef6fdb6ff5c8063ebb173ed4fec5a91cbf229c8":{"kind":"REWORK","_number":1,"created":"2023-01-20 06:57:06.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/22/871222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/22/871222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/22/871222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/22/871222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/22/871222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/22/871222/1"}}},"commit":{"parents":[{"commit":"1e9a7d3a520745fc677f4b7c040b670cf5a2cca5","subject":"Fix tox4 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/1e9a7d3a520745fc677f4b7c040b670cf5a2cca5"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-20 06:56:23.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-20 06:57:02.000000000","tz":330},"subject":"Address RemovedInDjango40Warning","message":"Address RemovedInDjango40Warning\n\ndjango.utils.translation.ugettext(), ugettext_lazy(), ugettext_noop(),\nungettext(), and ungettext_lazy() are deprecated in favor of the\nfunctions that they’re aliases for: django.utils.translation.gettext(),\ngettext_lazy(), gettext_noop(), ngettext(), and ngettext_lazy().\nFor more info. please refer [1].\n\nhttps://docs.djangoproject.com/en/4.0/releases/3.0/#id3\n\n[1] https://review.opendev.org/c/openstack/horizon/+/827092\n\nChange-Id: Idea2712e701488575c3f418f11ecae12516fdc68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/5ef6fdb6ff5c8063ebb173ed4fec5a91cbf229c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/5ef6fdb6ff5c8063ebb173ed4fec5a91cbf229c8"}]},"branch":"refs/heads/master"},"ef2bebf58ede4dc53d0b2adf7fe35a389405763f":{"kind":"REWORK","_number":2,"created":"2023-01-20 10:29:13.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/22/871222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/22/871222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/22/871222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/22/871222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/22/871222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/22/871222/2"}}},"commit":{"parents":[{"commit":"1e9a7d3a520745fc677f4b7c040b670cf5a2cca5","subject":"Fix tox4 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/1e9a7d3a520745fc677f4b7c040b670cf5a2cca5"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-20 06:56:23.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-20 10:01:34.000000000","tz":330},"subject":"Address RemovedInDjango40Warning","message":"Address RemovedInDjango40Warning\n\ndjango.utils.translation.ugettext(), ugettext_lazy(), ugettext_noop(),\nungettext(), and ungettext_lazy() are deprecated in favor of the\nfunctions that they’re aliases for: django.utils.translation.gettext(),\ngettext_lazy(), gettext_noop(), ngettext(), and ngettext_lazy().\nFor more info. please refer [1].\n\nhttps://docs.djangoproject.com/en/4.0/releases/3.0/#id3\n\n[1] https://review.opendev.org/c/openstack/horizon/+/827092\n\nChange-Id: Idea2712e701488575c3f418f11ecae12516fdc68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/ef2bebf58ede4dc53d0b2adf7fe35a389405763f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/ef2bebf58ede4dc53d0b2adf7fe35a389405763f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}}]}],"submit_requirements":[]}
