)]}'
{"id":"openstack%2Ftacker-horizon~987516","triplet_id":"openstack%2Ftacker-horizon~master~I33f1d1b7edeb8fb995ba3dd2568deb860c2a918f","project":"openstack/tacker-horizon","branch":"master","hashtags":["extract-messages"],"change_id":"I33f1d1b7edeb8fb995ba3dd2568deb860c2a918f","subject":"Update Babel configuration","status":"NEW","created":"2026-05-06 14:05:28.000000000","updated":"2026-05-06 20:40:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4867853cc046625bd6440893ebec1c5e99f381a5","_number":987516,"virtual_id_number":987516,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-06 14:19:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-06 14:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d9c7eb7d0c65a1113d3e20816b21c3153d13afb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 14:05:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f15470a0c289a7b6d55eb0e3cf3c4badf0df8867","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 14:06:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"335ccec094cc5a7105e8dc2d74c5ae735988fb1d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 14:06:23.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"423fe3a97b4888c541764c854aba1e8ed3ade95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 14:19:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f5c0edbca8e468bb0f1b627fdd2c710\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/935558367445447a9412f04cf70894fa : SUCCESS in 3m 57s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c8fd632730ac4034b919e0e88ff77d52 : SUCCESS in 4m 50s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/88559c4600af4107bbb2f8d7469b7607 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02efb7ead8284d71a88a564cbd87246e : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d30984915e7540199a754c21403f623d : SUCCESS in 4m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/afab1e50a597476d907453220ea94593 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9b0305225949f88a487cf91cdbc308 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4867853cc046625bd6440893ebec1c5e99f381a5","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-06 20:40:00.000000000","message":"Hashtag added: extract-messages","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"58e8347b46f7dd4fcb941f8f11f0cd45f0847f61","revisions":{"c809a0a890a7c43f30294fa186a104479ea8b1ac":{"kind":"REWORK","_number":1,"created":"2026-05-06 14:05:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/16/987516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/16/987516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/16/987516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/16/987516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/16/987516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/16/987516/1"}}},"commit":{"parents":[{"commit":"714c5d65dc6f6ccf197a42daf48685d92663e1dc","subject":"Update requirements files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/714c5d65dc6f6ccf197a42daf48685d92663e1dc"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 14:05:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 14:05:28.000000000","tz":0},"subject":"Update Babel configuration","message":"Update Babel configuration\n\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/OSRDVYRX6TFI2VA4LHXDXX7P7CLJCQKM/\n\nChange-Id: I33f1d1b7edeb8fb995ba3dd2568deb860c2a918f\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/c809a0a890a7c43f30294fa186a104479ea8b1ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/c809a0a890a7c43f30294fa186a104479ea8b1ac"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"58e8347b46f7dd4fcb941f8f11f0cd45f0847f61":{"kind":"REWORK","_number":2,"created":"2026-05-06 14:06:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/16/987516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/16/987516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/16/987516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/16/987516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/16/987516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/16/987516/2"}}},"commit":{"parents":[{"commit":"714c5d65dc6f6ccf197a42daf48685d92663e1dc","subject":"Update requirements files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/714c5d65dc6f6ccf197a42daf48685d92663e1dc"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 14:05:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 14:06:20.000000000","tz":0},"subject":"Update Babel configuration","message":"Update Babel configuration\n\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/OSRDVYRX6TFI2VA4LHXDXX7P7CLJCQKM/\n\nChange-Id: I33f1d1b7edeb8fb995ba3dd2568deb860c2a918f\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/58e8347b46f7dd4fcb941f8f11f0cd45f0847f61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/58e8347b46f7dd4fcb941f8f11f0cd45f0847f61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
