)]}'
{"id":"openstack%2Ftacker-specs~591866","triplet_id":"openstack%2Ftacker-specs~master~I0ad50d3ada4237cdf7771559a5332195314ad5b4","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-etsi-nfv-specs","hashtags":[],"change_id":"I0ad50d3ada4237cdf7771559a5332195314ad5b4","subject":"VNF LCM API based on ETSI NFV-SOL specification","status":"MERGED","created":"2018-08-14 23:11:45.000000000","updated":"2020-03-08 02:57:13.000000000","submitted":"2020-03-08 02:56:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":77,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"591866-1583636165398-9c168db8","meta_rev_id":"be559d4b50e68ae9078d66a01d6c71f53336d19b","_number":591866,"virtual_id_number":591866,"owner":{"_account_id":28530,"name":"Aliaksandr Hrechny","email":"grechny.alexander@gmail.com","username":"ahrechny"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-08 02:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":27181,"name":"lijiale","email":"lijiale_yewu@cmss.chinamobile.com","username":"lijiale"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":0,"_account_id":28528,"name":"Kozhevnikov Aleksei","email":"Magistrazh@gmail.com","username":"Magistrazh"},{"value":0,"date":"2020-02-21 04:59:15.000000000","_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},{"value":0,"_account_id":28530,"name":"Aliaksandr Hrechny","email":"grechny.alexander@gmail.com","username":"ahrechny"},{"value":0,"date":"2020-02-21 10:44:12.000000000","_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"date":"2020-03-02 10:19:05.000000000","_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},{"value":0,"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":2,"date":"2020-02-21 02:23:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":27181,"name":"lijiale","email":"lijiale_yewu@cmss.chinamobile.com","username":"lijiale"},{"value":2,"date":"2020-02-18 16:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":0,"_account_id":28528,"name":"Kozhevnikov Aleksei","email":"Magistrazh@gmail.com","username":"Magistrazh"},{"value":0,"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},{"value":0,"_account_id":28530,"name":"Aliaksandr Hrechny","email":"grechny.alexander@gmail.com","username":"ahrechny"},{"value":0,"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},{"value":0,"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":2,"date":"2020-02-18 09:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":1,"date":"2020-03-08 02:50:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":27181,"name":"lijiale","email":"lijiale_yewu@cmss.chinamobile.com","username":"lijiale"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":0,"_account_id":28528,"name":"Kozhevnikov Aleksei","email":"Magistrazh@gmail.com","username":"Magistrazh"},{"value":0,"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},{"value":0,"_account_id":28530,"name":"Aliaksandr Hrechny","email":"grechny.alexander@gmail.com","username":"ahrechny"},{"value":0,"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},{"value":0,"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"},{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"_account_id":27181,"name":"lijiale","email":"lijiale_yewu@cmss.chinamobile.com","username":"lijiale"},{"_account_id":28528,"name":"Kozhevnikov Aleksei","email":"Magistrazh@gmail.com","username":"Magistrazh"},{"_account_id":28530,"name":"Aliaksandr Hrechny","email":"grechny.alexander@gmail.com","username":"ahrechny"},{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-20 13:21:40.000000000","updated_by":{"_account_id":27181,"name":"lijiale","email":"lijiale_yewu@cmss.chinamobile.com","username":"lijiale"},"reviewer":{"_account_id":27181,"name":"lijiale","email":"lijiale_yewu@cmss.chinamobile.com","username":"lijiale"},"state":"REVIEWER"},{"updated":"2018-08-27 04:41:43.000000000","updated_by":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"reviewer":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"state":"REVIEWER"},{"updated":"2018-08-29 09:12:51.000000000","updated_by":{"_account_id":28528,"name":"Kozhevnikov Aleksei","email":"Magistrazh@gmail.com","username":"Magistrazh"},"reviewer":{"_account_id":28528,"name":"Kozhevnikov Aleksei","email":"Magistrazh@gmail.com","username":"Magistrazh"},"state":"REVIEWER"},{"updated":"2019-04-02 11:27:34.000000000","updated_by":{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"},"reviewer":{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"},"state":"REVIEWER"},{"updated":"2019-06-24 07:54:19.000000000","updated_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2020-01-29 02:31:20.000000000","updated_by":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"reviewer":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"state":"REVIEWER"},{"updated":"2020-02-18 09:11:39.000000000","updated_by":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"reviewer":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"state":"REVIEWER"},{"updated":"2020-02-18 16:20:29.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2020-02-21 04:59:15.000000000","updated_by":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"reviewer":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"state":"REVIEWER"},{"updated":"2020-02-21 10:44:12.000000000","updated_by":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"reviewer":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"state":"REVIEWER"},{"updated":"2020-03-02 10:19:05.000000000","updated_by":{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},"reviewer":{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},"state":"REVIEWER"},{"updated":"2020-03-08 02:50:44.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2020-03-08 02:56:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"307b4d104e71228abf64b759f0883fd049384444","author":{"_account_id":28530,"name":"Aliaksandr Hrechny","email":"grechny.alexander@gmail.com","username":"ahrechny"},"date":"2018-08-14 23:11:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37cc6b5bd17c9d4b159a1f108f5dd6da5e1dd6ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-14 23:18:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/66/591866/1/check/build-openstack-sphinx-docs/790682c/html/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/66/591866/1/check/openstack-tox-pep8/cca809d/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d64c5c0382ada2aa9fd1110eb480a5fead32ee8f","author":{"_account_id":27181,"name":"lijiale","email":"lijiale_yewu@cmss.chinamobile.com","username":"lijiale"},"date":"2018-08-20 13:21:40.000000000","message":"Patch Set 1:\n\nit\u0027s a big job. by the way, do you want to split nfvo and vnfm as ETSI ?","accounts_in_message":[],"_revision_number":1},{"id":"bd4f2507aeaf3f5a2d76cae800d66bcdce7fa4d0","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2018-08-27 07:18:01.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThis spec looks good to me. As Lijiale said, we can break the big job to small jobs, such as REST API for vnf_instance, vnf, etc. Then we can implement a part from a big job.\n\nIMO, we can re-organize the code, that can be reused, to support both new REST API and the old one. Then we can deprecate the old API in the future.","accounts_in_message":[],"_revision_number":1},{"id":"c5567bb43db4c2ab98a3648e95c1bb2d30329c18","author":{"_account_id":28528,"name":"Kozhevnikov Aleksei","email":"Magistrazh@gmail.com","username":"Magistrazh"},"date":"2018-08-27 14:39:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73595a7fa05952800feb872c7658c37f16ae7505","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2018-08-28 03:55:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"787d9601914307809e5c168045c677e367494e06","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2018-08-28 09:27:09.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cfbb04887ee844e501325892ae026a39ffc5b6e8","author":{"_account_id":28528,"name":"Kozhevnikov Aleksei","email":"Magistrazh@gmail.com","username":"Magistrazh"},"date":"2018-08-29 09:12:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c614e94dff685aab8b0508e43dd38bad64f31c33","author":{"_account_id":28530,"name":"Aliaksandr Hrechny","email":"grechny.alexander@gmail.com","username":"ahrechny"},"date":"2018-09-20 08:42:01.000000000","message":"Patch Set 1:\n\n\u003e it\u0027s a big job. by the way, do you want to split nfvo and vnfm as\n \u003e ETSI ?\n\nIn my opinion we shouldn\u0027t split them as etsi for now, but it makes sense to discuss","accounts_in_message":[],"_revision_number":1},{"id":"51827b7b6efdb672bf04f477c46a8f8c4d7ca155","author":{"_account_id":28530,"name":"Aliaksandr Hrechny","email":"grechny.alexander@gmail.com","username":"ahrechny"},"date":"2018-09-20 09:32:04.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c4fda90e15b280b2dac884a6335e3acdd54e6d99","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2018-09-20 12:58:30.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"673fb8fdfa6696b23e54479187dbd7d4fdd9f9a8","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2018-10-26 10:06:11.000000000","message":"Patch Set 1: Code-Review-1\n\n@Aliaksandr\n\nCould you please update the patch for open queries and TODO items.","accounts_in_message":[],"_revision_number":1},{"id":"fc1be3e0a87dd7507a68211222b3e625ef5bde91","author":{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"},"date":"2019-04-02 11:27:34.000000000","message":"Patch Set 1:\n\nHi Aliaksandr,\n\nCan you please provide a sample VNFD template to better understand the parameters that will in VNFD.","accounts_in_message":[],"_revision_number":1},{"id":"ec4e263d0468c6071b3aa7f6891f4d2388db92bd","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-04-04 05:22:22.000000000","message":"Patch Set 1:\n\nHi Aliaksandr,\ndo you still work on this spec? If not, may I take over this spec?","accounts_in_message":[],"_revision_number":1},{"id":"9920c22b1b5218ce178f5f1794aa25b2810402c5","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2019-04-04 08:59:18.000000000","message":"Patch Set 1:\n\n\u003e Hi Aliaksandr,\n \u003e do you still work on this spec? If not, may I take over this spec?\n\nI am happy if someone take over this feature","accounts_in_message":[],"_revision_number":1},{"id":"1f2ccafc0a4e83e60d31ba1a4eae4b9d43970699","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-04-15 07:31:53.000000000","message":"Patch Set 1:\n\n\u003e \u003e Hi Aliaksandr,\n \u003e \u003e do you still work on this spec? If not, may I take over this\n \u003e spec?\n \u003e \n \u003e I am happy if someone take over this feature\n\n+1","accounts_in_message":[],"_revision_number":1},{"id":"cd3c636f7e308518e6355cb7e87f293fb40641d9","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-04-25 08:18:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"216c02ef1f1638f9671387bb12a121fe93ea01ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 08:28:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/66/591866/2/check/openstack-tox-docs/b531521/html/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/66/591866/2/check/openstack-tox-pep8/4bd2c84/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"ad28ea8f6ddeb5064271c0d32349375f601798ef","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2019-05-04 02:34:18.000000000","message":"Patch Set 2: Code-Review+1\n\n(4 comments)\n\nThis spec almost look good to me. Describing step-to-step to finish this spec will be good.","accounts_in_message":[],"_revision_number":2},{"id":"aa03975a8b9478cde74288482c52a067094377ae","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-06 09:25:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b8c586b72ca9ec5e3a5cdf9c53a011271b23bda0","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2019-05-07 12:52:19.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nSorry if it is a basic question...","accounts_in_message":[],"_revision_number":2},{"id":"14c365726341ab741bcaa6bf57f78ecd9a09ba15","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2019-05-08 02:52:10.000000000","message":"Patch Set 2:\n\nSorry, \"just creating ID of VNF\" is a little wrong.\nProbably, VNFM parse VNF Descriptor and have the information in the procedure.\n\nWhat I want to say is \"createion of a VNF instance resource\" do not include instantiation, and Tacker\u0027s create-vnf is different from such feature.","accounts_in_message":[],"_revision_number":2},{"id":"5bc2f257df15587f752fd0fcbc7b9f2ee9e14d7a","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-05-14 07:29:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bc42410547fa99fc086a644f908607737b42d80c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 07:38:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/66/591866/3/check/openstack-tox-docs/eb195a2/html/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/66/591866/3/check/openstack-tox-pep8/67af47e/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"6aca77309c6e8e87b9987b3aa023117667d0c904","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-05-14 07:53:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"18a55d6dc5346a412a657add1ad1d9270d7860b4","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-05-14 07:53:54.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f8cd70922f265b01b9869c2fecc90055658d666c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 08:02:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/66/591866/4/check/openstack-tox-docs/41b66ac/ : FAILURE in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/66/591866/4/check/openstack-tox-pep8/98dbc65/ : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":4},{"id":"61d35e4d182485fbc6bd8f7c7e3c5b80cc9154d8","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-05-15 01:24:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1c4ee7754f8b310468d13c44a5e10081fafda154","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 01:37:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/66/591866/5/check/openstack-tox-docs/7194572/html/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/66/591866/5/check/openstack-tox-pep8/5bc0814/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":5},{"id":"930652400c3cebbc53ce4db5967e6e3ec57ab72e","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2019-05-31 00:51:29.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nThank you for update.\nI have some additional comments.","accounts_in_message":[],"_revision_number":5},{"id":"6710bc153d9f36cc6d80cbabaf8d1e87a04886ce","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-06-04 08:12:37.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"487ed952ae372541a99152245f4f055b33326bc5","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-06-05 00:48:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2ce931bbde1c0475590119d5f5321b7e6f9b5d96","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-06-05 01:36:25.000000000","message":"Patch Set 5:\n\n\u003e (1 comment)\n\nI got it! Thanks Jo:)","accounts_in_message":[],"_revision_number":5},{"id":"9a747a074ebe61ae8183196b2bff9a373110a865","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-06-07 02:54:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6cf6902bc8c92722456eb24baa3b12ae8080f5e3","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-06-07 02:55:08.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b3f297c32b1951e4aaa70611dfbd1d90a7e8cef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 03:08:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/66/591866/6/check/openstack-tox-docs/e202771/html/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/66/591866/6/check/openstack-tox-pep8/8de6e67/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":6},{"id":"5f56bc56a3cd07ff7681c6fbf5d8714e8d6abf14","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2019-06-24 07:54:19.000000000","message":"Patch Set 6: Code-Review+1\n\nI\u0027m so sorry for the late reply, LGTM.","accounts_in_message":[],"_revision_number":6},{"id":"918e38bad23c4b4e7a3457ea6eb7c1a5c13e64d1","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2019-06-24 16:51:35.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"a8f6419f04faaad7caa0b7b7f8d092962cce3a67","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2019-11-06 05:14:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"597d2e9fae6884b98809fba4e850b22550af99ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 05:20:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba57c51cdabd4495a2add9e3b418751a : FAILURE in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d91ce988cce748eea7cba6eabc3c5e03 : FAILURE in 2m 11s","accounts_in_message":[],"_revision_number":7},{"id":"047240344c4958ff169b71c7bc765714e96e56d9","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2019-11-07 08:05:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"70714493f1ca72c588186e4e4c15eb1880fefe88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 08:19:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13126b18a7ad49a18d34fe7e3bfb8aa0 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/501e3afe114048e9aea606c5e717f5fd : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":8},{"id":"ed9de53a5b8365ee33eb92ab733769695d722ba2","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-11-19 08:05:07.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8e4ece4913d0f02647adf0fa9df97976da5ec17c","author":{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},"date":"2019-12-18 03:04:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8e49046195fea4cb43793da5d9c0f68ec443a74a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:09:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1145528fa14510b43568ac7d750486 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7fd5383cf0948bba715646eaeaa883b : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":9},{"id":"528ffa71300727c8bcb6675ab219e09b8b4e8699","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-01-09 09:38:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"13eb553e21ecbd3d9f914ff9efc6ec425f510a86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 09:44:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec681ca8990432d97369a446e3924de : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12fbb8e26b15499abe278ef9fd5a9649 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":10},{"id":"d7252b58dbf6994bc2682d511aeb1ffaec0a098a","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2020-01-14 08:26:23.000000000","message":"Patch Set 10:\n\n(8 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f4d0934307c8fc539715c37a6a1013eef9390dfb","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-01-29 02:29:44.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"509047c377fb245cd98297ff46fd1d19b076b9dd","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-01-29 02:31:20.000000000","message":"Patch Set 11:\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"45c9ac5adb52ec8c75ee951843444dbf771bca3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 02:33:51.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4aad0df48544779095228b03653225 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad9ea511e744962a73864cb096b10a0 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":11},{"id":"4b07d7859d043c3615b3a724edeb5436598d9a1d","author":{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},"date":"2020-02-07 09:51:34.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b89d77c3fd28b0fb34e15ac3af5bd434b6ea54f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 10:00:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313066e70d5c421aafbc2532f5009815 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fadf7b5c94d341ecb55d1c7cda25a76b : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":12},{"id":"75f9f8c8085c61b83b2ed71a7335723041eabc09","author":{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},"date":"2020-02-10 01:32:01.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f51900a42ff1ce43c226f78ebe0e67b14f8ae2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 01:36:13.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bde2742b28240ae87b2bb8b28853733 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d1cfbd9b4b043e994117581d1af6587 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":13},{"id":"eba687541d3ce4fb980548e6fabb4ab46178b48a","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-02-18 08:10:39.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f72636e8ce16423b416f4daf773ee5c5a5d7b97f","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-02-18 08:16:44.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"4f277eec0e166a04b9f4a80eb5593341ddad8c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 08:23:55.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc7faf6414144598cd7a60dff39acf4 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/422b0ae42d674bab9fe3187976d45392 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":15},{"id":"851f07890151c25b0b3328079e97601266845e9e","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2020-02-18 09:11:39.000000000","message":"Patch Set 15: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":15},{"id":"4fcf8776e0018cd6a3b84bf85191d9275e3dcbaf","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2020-02-18 16:20:29.000000000","message":"Patch Set 15: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":15},{"id":"8934fd09f4e1e89c33873f3b9efdcf0aafd27cd0","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-02-21 02:23:43.000000000","message":"Patch Set 15: Code-Review+2\n\n(4 comments)\n\nLooks good.\nBefore merging it, I request you to please  reply on comments.","accounts_in_message":[],"_revision_number":15},{"id":"e42802467e12db86e548eedcce6266957ffc91bc","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2020-02-21 04:59:15.000000000","message":"Patch Set 15:\n\n(3 comments)\n\nI have a few simple question for this.:)","accounts_in_message":[],"_revision_number":15},{"id":"a3b5e853f14b519ebfae7ec7655fc97cc9c3fce3","author":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"date":"2020-02-21 10:44:12.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"5a01d3831878486deb739c02580d5507b28a48c9","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-02-25 00:52:25.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"a9f5db3c6868a2b39572411a1cf3361ccd11ac94","author":{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},"date":"2020-03-02 10:19:05.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d0c76869425c296c3b1829cebccd01e8a7939cf4","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-03-08 02:50:44.000000000","message":"Patch Set 15: Workflow+1\n\n(1 comment)\n\nLets merge this spec.\nPlease make sure to keep trak for further enhancements.","accounts_in_message":[],"_revision_number":15},{"id":"c470e490a67fb6aff982b29363f04a0fafa63a73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 02:50:55.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"59dc75d2505b6f72e62e8f805132f8d58542a17f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 02:56:05.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9625c6e3b45f44048ec6bf7f24a58365 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eebbe00326c740b383bf13674f0030fe : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":15},{"id":"98b825a93874c1e5b7924aa6d35d19c92718cbd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 02:56:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"d0e42dc18f0fb293a8a1a2f080ff4f214e69286c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 02:57:13.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/98a5d160b13c4e3b917bf4565ff8d909 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"8c5dc60b7d096ea1f2b6c86f4350bee8a9a5b7b5","revisions":{"e6cd8f37c7e9f250523a6c981e82d2624568c555":{"kind":"REWORK","_number":1,"created":"2018-08-14 23:11:45.000000000","uploader":{"_account_id":28530,"name":"Aliaksandr Hrechny","email":"grechny.alexander@gmail.com","username":"ahrechny"},"ref":"refs/changes/66/591866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/1"}}},"commit":{"parents":[{"commit":"2184b85a0dd8f20eab899a85ee40535b80abf070","subject":"Merge \"Support shared vim for policy action execution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/2184b85a0dd8f20eab899a85ee40535b80abf070"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 23:08:03.000000000","tz":120},"subject":"REST API for VNF based on ETSI NFV-SOL specification","message":"REST API for VNF based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e6cd8f37c7e9f250523a6c981e82d2624568c555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e6cd8f37c7e9f250523a6c981e82d2624568c555"}]},"branch":"refs/heads/master"},"a24c75518cd95dc1803b655dce8cf0f981170968":{"kind":"REWORK","_number":2,"created":"2019-04-25 08:18:52.000000000","uploader":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"ref":"refs/changes/66/591866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/2"}}},"commit":{"parents":[{"commit":"744b055da51c4c51e0f61024f3b614ae0a5bfeaf","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/744b055da51c4c51e0f61024f3b614ae0a5bfeaf"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Hiroyuki JO","email":"jo.hiroyuki@lab.ntt.co.jp","date":"2019-04-24 05:55:54.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/a24c75518cd95dc1803b655dce8cf0f981170968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/a24c75518cd95dc1803b655dce8cf0f981170968"}]},"branch":"refs/heads/master"},"b064a4a84b5c756b6f292427853b00017916318f":{"kind":"REWORK","_number":3,"created":"2019-05-14 07:29:45.000000000","uploader":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"ref":"refs/changes/66/591866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/3"}}},"commit":{"parents":[{"commit":"744b055da51c4c51e0f61024f3b614ae0a5bfeaf","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/744b055da51c4c51e0f61024f3b614ae0a5bfeaf"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Hiroyuki JO","email":"jo.hiroyuki@lab.ntt.co.jp","date":"2019-05-14 07:28:56.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b064a4a84b5c756b6f292427853b00017916318f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b064a4a84b5c756b6f292427853b00017916318f"}]},"branch":"refs/heads/master"},"6d2ce1002fd40f2d652b482de5bce659eb05228a":{"kind":"REWORK","_number":4,"created":"2019-05-14 07:53:27.000000000","uploader":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"ref":"refs/changes/66/591866/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/4"}}},"commit":{"parents":[{"commit":"744b055da51c4c51e0f61024f3b614ae0a5bfeaf","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/744b055da51c4c51e0f61024f3b614ae0a5bfeaf"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Hiroyuki JO","email":"jo.hiroyuki@lab.ntt.co.jp","date":"2019-05-14 07:53:19.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6d2ce1002fd40f2d652b482de5bce659eb05228a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6d2ce1002fd40f2d652b482de5bce659eb05228a"}]},"branch":"refs/heads/master"},"573146199d21f1e477e3209b182d7c4d7ca7a877":{"kind":"REWORK","_number":5,"created":"2019-05-15 01:24:46.000000000","uploader":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"ref":"refs/changes/66/591866/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/5"}}},"commit":{"parents":[{"commit":"744b055da51c4c51e0f61024f3b614ae0a5bfeaf","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/744b055da51c4c51e0f61024f3b614ae0a5bfeaf"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Hiroyuki JO","email":"jo.hiroyuki@lab.ntt.co.jp","date":"2019-05-15 01:24:34.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/573146199d21f1e477e3209b182d7c4d7ca7a877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/573146199d21f1e477e3209b182d7c4d7ca7a877"}]},"branch":"refs/heads/master"},"da21520485e80382465db8834c4c0976101854b8":{"kind":"REWORK","_number":6,"created":"2019-06-07 02:54:49.000000000","uploader":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"ref":"refs/changes/66/591866/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/6"}}},"commit":{"parents":[{"commit":"744b055da51c4c51e0f61024f3b614ae0a5bfeaf","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/744b055da51c4c51e0f61024f3b614ae0a5bfeaf"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Hiroyuki JO","email":"jo.hiroyuki@lab.ntt.co.jp","date":"2019-06-07 01:55:46.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/da21520485e80382465db8834c4c0976101854b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/da21520485e80382465db8834c4c0976101854b8"}]},"branch":"refs/heads/master"},"e73f2ac52b454253f8ce80188c675b344423c9cf":{"kind":"REWORK","_number":7,"created":"2019-11-06 05:14:05.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/66/591866/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/7"}}},"commit":{"parents":[{"commit":"906e411895dfc6fff86b44fe5418da430c12b0cb","subject":"add Ussuri specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/906e411895dfc6fff86b44fe5418da430c12b0cb"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2019-11-06 04:31:52.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e73f2ac52b454253f8ce80188c675b344423c9cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e73f2ac52b454253f8ce80188c675b344423c9cf"}]},"branch":"refs/heads/master"},"a4335383932cbd1e9e40ceb39f3d507c797b6257":{"kind":"REWORK","_number":8,"created":"2019-11-07 08:05:32.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/66/591866/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/8"}}},"commit":{"parents":[{"commit":"906e411895dfc6fff86b44fe5418da430c12b0cb","subject":"add Ussuri specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/906e411895dfc6fff86b44fe5418da430c12b0cb"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2019-11-06 14:06:23.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/a4335383932cbd1e9e40ceb39f3d507c797b6257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/a4335383932cbd1e9e40ceb39f3d507c797b6257"}]},"branch":"refs/heads/master"},"440fd205672851f7ee031e728c5f35596afe7bfc":{"kind":"REWORK","_number":9,"created":"2019-12-18 03:04:12.000000000","uploader":{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},"ref":"refs/changes/66/591866/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/9"}}},"commit":{"parents":[{"commit":"906e411895dfc6fff86b44fe5418da430c12b0cb","subject":"add Ussuri specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/906e411895dfc6fff86b44fe5418da430c12b0cb"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Nitin Uikey","email":"nitin.uikey@nttdata.com","date":"2019-12-18 02:51:21.000000000","tz":0},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/440fd205672851f7ee031e728c5f35596afe7bfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/440fd205672851f7ee031e728c5f35596afe7bfc"}]},"branch":"refs/heads/master"},"7306c0a32592b8aa8b5b204959ec6f256d8c648e":{"kind":"REWORK","_number":10,"created":"2020-01-09 09:38:20.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/66/591866/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/10"}}},"commit":{"parents":[{"commit":"906e411895dfc6fff86b44fe5418da430c12b0cb","subject":"add Ussuri specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/906e411895dfc6fff86b44fe5418da430c12b0cb"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-01-09 08:34:02.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nCo-Authored-By: Keiko Kuriu \u003ckeiko.kuriu.wa@hco.ntt.co.jp\u003e\nCo-Authored-By: Tushar Patil \u003ctushar.vitthal.patil@gmail.com\u003e\nCo-Authored-By: Nitin Uikey \u003cnitin.uikey@nttdata.com\u003e\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/7306c0a32592b8aa8b5b204959ec6f256d8c648e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/7306c0a32592b8aa8b5b204959ec6f256d8c648e"}]},"branch":"refs/heads/master"},"395604ed3baa8d2b86b914a7dce9a01ae4991d6f":{"kind":"REWORK","_number":11,"created":"2020-01-29 02:29:44.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/66/591866/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/11"}}},"commit":{"parents":[{"commit":"906e411895dfc6fff86b44fe5418da430c12b0cb","subject":"add Ussuri specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/906e411895dfc6fff86b44fe5418da430c12b0cb"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-01-29 02:26:50.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nCo-Authored-By: Keiko Kuriu \u003ckeiko.kuriu.wa@hco.ntt.co.jp\u003e\nCo-Authored-By: Tushar Patil \u003ctushar.vitthal.patil@gmail.com\u003e\nCo-Authored-By: Nitin Uikey \u003cnitin.uikey@nttdata.com\u003e\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/395604ed3baa8d2b86b914a7dce9a01ae4991d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/395604ed3baa8d2b86b914a7dce9a01ae4991d6f"}]},"branch":"refs/heads/master"},"0015c65c378f77fce003ff4ec9b907b6e3cf3b69":{"kind":"REWORK","_number":12,"created":"2020-02-07 09:51:34.000000000","uploader":{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},"ref":"refs/changes/66/591866/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/12"}}},"commit":{"parents":[{"commit":"906e411895dfc6fff86b44fe5418da430c12b0cb","subject":"add Ussuri specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/906e411895dfc6fff86b44fe5418da430c12b0cb"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Nitin Uikey","email":"nitin.uikey@nttdata.com","date":"2020-02-07 09:50:14.000000000","tz":0},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0015c65c378f77fce003ff4ec9b907b6e3cf3b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0015c65c378f77fce003ff4ec9b907b6e3cf3b69"}]},"branch":"refs/heads/master"},"73f32a13b717276bb4f02507e4d7dadd4874e3b0":{"kind":"REWORK","_number":13,"created":"2020-02-10 01:32:01.000000000","uploader":{"_account_id":30650,"name":"nitin uikey","email":"nitin.uikey@nttdata.com","username":"nitin.uikey"},"ref":"refs/changes/66/591866/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/13"}}},"commit":{"parents":[{"commit":"906e411895dfc6fff86b44fe5418da430c12b0cb","subject":"add Ussuri specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/906e411895dfc6fff86b44fe5418da430c12b0cb"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Nitin Uikey","email":"nitin.uikey@nttdata.com","date":"2020-02-10 01:28:57.000000000","tz":0},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/73f32a13b717276bb4f02507e4d7dadd4874e3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/73f32a13b717276bb4f02507e4d7dadd4874e3b0"}]},"branch":"refs/heads/master"},"2651874e0a5a17fdab1906de0aa66c02990ecf2f":{"kind":"REWORK","_number":14,"created":"2020-02-18 08:10:39.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/66/591866/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/14"}}},"commit":{"parents":[{"commit":"906e411895dfc6fff86b44fe5418da430c12b0cb","subject":"add Ussuri specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/906e411895dfc6fff86b44fe5418da430c12b0cb"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-02-18 08:08:35.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nCo-Authored-By: Keiko Kuriu \u003ckeiko.kuriu.wa@hco.ntt.co.jp\u003e\nCo-Authored-By: Tushar Patil \u003ctushar.vitthal.patil@gmail.com\u003e\nCo-Authored-By: Nitin Uikey \u003cnitin.uikey@nttdata.com\u003e\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/2651874e0a5a17fdab1906de0aa66c02990ecf2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/2651874e0a5a17fdab1906de0aa66c02990ecf2f"}]},"branch":"refs/heads/master"},"8c5dc60b7d096ea1f2b6c86f4350bee8a9a5b7b5":{"kind":"REWORK","_number":15,"created":"2020-02-18 08:16:44.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/66/591866/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/66/591866/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/66/591866/15"}}},"commit":{"parents":[{"commit":"906e411895dfc6fff86b44fe5418da430c12b0cb","subject":"add Ussuri specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/906e411895dfc6fff86b44fe5418da430c12b0cb"}]}],"author":{"name":"ahrechny","email":"grechny.alexander@gmail.com","date":"2018-08-14 22:02:56.000000000","tz":120},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-02-18 08:16:22.000000000","tz":540},"subject":"VNF LCM API based on ETSI NFV-SOL specification","message":"VNF LCM API based on ETSI NFV-SOL specification\n\nIntroduce new VNF Lifecycle and REST API endpoints for\nVNF management instead of existing ones.\n\nBlueprint: support-etsi-nfv-specs\n\nChange-Id: I0ad50d3ada4237cdf7771559a5332195314ad5b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8c5dc60b7d096ea1f2b6c86f4350bee8a9a5b7b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8c5dc60b7d096ea1f2b6c86f4350bee8a9a5b7b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
