)]}'
{"id":"openstack%2Ftacker-specs~636272","triplet_id":"openstack%2Ftacker-specs~master~I83347cdc7c7ed4c560d8cee412edf575696eb038","project":"openstack/tacker-specs","branch":"master","topic":"bp/multi-interface-container","hashtags":[],"change_id":"I83347cdc7c7ed4c560d8cee412edf575696eb038","subject":"Adding Multi-Interface for Containerized VNF in Tacker","status":"MERGED","created":"2019-02-12 06:52:05.000000000","updated":"2019-06-25 10:24:32.000000000","submitted":"2019-06-25 10:24:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"636272-1561458272999-f027147f","meta_rev_id":"4a3182434c873ae339198b038f1345a2e4c8c071","_number":636272,"virtual_id_number":636272,"owner":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":2,"date":"2019-06-25 10:24:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":0,"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},{"value":2,"date":"2019-05-29 05:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":2,"date":"2019-06-25 10:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"all":[{"value":0,"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":1,"date":"2019-06-25 10:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 00:59:21.000000000","updated_by":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"reviewer":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"state":"REVIEWER"},{"updated":"2019-05-28 01:36:26.000000000","updated_by":{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},"reviewer":{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},"state":"REVIEWER"},{"updated":"2019-05-29 05:54:15.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-06-25 10:13:43.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2019-06-25 10:24:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aac3e94847e3f23a4f422233be2c67ffffd7acda","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-02-12 06:52:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43ae347dfd387c1592134adcafa3cbc6d7c2b90b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 07:02:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/1/check/openstack-tox-docs/f17148d/ : FAILURE in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/1/check/openstack-tox-pep8/1a85dba/ : FAILURE in 8m 42s","accounts_in_message":[],"_revision_number":1},{"id":"33b08486df34cb6c6a277d8001b8e4cb1eb92851","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-02-12 07:07:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"068a8bb2d93ef9b14a42c172ff9dc7239ba5f44d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 07:12:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/2/check/openstack-tox-docs/1e72e4a/ : FAILURE in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/2/check/openstack-tox-pep8/1591413/ : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"448337a85bdfa54fbad561e168d159fe41b116e8","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-02-12 08:13:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c26ed9793a80d6a01b6742e3fdb03b32c7a564a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 08:18:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/3/check/openstack-tox-docs/0267fcc/ : FAILURE in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/3/check/openstack-tox-pep8/3e2d8eb/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"ca496d19c19fc77554c998bc180081aa8c3e86cc","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-02-12 08:54:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7887303002fd096dd6a4e8719311606aab1cabac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 09:03:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/4/check/openstack-tox-docs/14ec2e9/ : FAILURE in 7m 58s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/4/check/openstack-tox-pep8/2a76210/ : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":4},{"id":"1a03eef87cf96dbb1e6fa4337c7d099d36a71b89","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-02-12 09:41:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5cc2767f0fc74b5fbce4711e3725f2b5a972f8f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 09:49:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/5/check/openstack-tox-docs/51a3d42/ : FAILURE in 5m 23s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/5/check/openstack-tox-pep8/3982a4b/ : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":5},{"id":"184beeba093909f3e9398656504d062a3bea4199","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-02-12 10:29:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9f93926ca6159f4611e37d957e99379ab6999600","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 10:39:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/6/check/openstack-tox-docs/ad4636a/html/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/6/check/openstack-tox-pep8/1aa4d9d/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":6},{"id":"e48707b888c154615ceddb6b8e18496a4f2c480b","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-03-16 06:54:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"31d2235bc28895d0fd5e898b36438e307cbce926","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-16 07:03:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/7/check/openstack-tox-docs/b477507/html/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/7/check/openstack-tox-pep8/1bdea79/ : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":7},{"id":"1900ff0974df60dac3653b132fc67611828d006d","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-03-16 07:09:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cdd13f651d2b68db952ce70d0eed8420df2a5496","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-16 07:17:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/8/check/openstack-tox-docs/034f5be/html/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/8/check/openstack-tox-pep8/b1b04a3/ : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":8},{"id":"e7da3a5218ff5b460162764e2144623b0ae22c16","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-03-19 05:54:28.000000000","message":"Patch Set 8: Code-Review-1\n\n(5 comments)\n\nLooks good initial draft.\nI will suggest to add more technical/architectural details.","accounts_in_message":[],"_revision_number":8},{"id":"5299a1dff36230cab6d0f791e71565bd0de21d3a","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-03-19 12:53:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"71ec0167f4c09fb107816ca8cc5dabdbf06712b0","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-03-19 12:53:12.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3352536e2df01a04ab72b8ef680ddb27dd23ba8a","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-03-19 12:55:56.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"0e88a3792e06b9a4c50e0c3f7737e27486e523eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-19 13:05:09.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/10/check/openstack-tox-docs/67fecb2/html/ : SUCCESS in 6m 40s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/10/check/openstack-tox-pep8/26329df/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":10},{"id":"4fbf7d4b5b64a6b460d0b1f99f49b5bf378129b0","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-04-01 02:32:23.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9167bf93583787ca0e65202ff78de11774b9ef85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 02:37:33.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/11/check/openstack-tox-docs/b34e4db/html/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/11/check/openstack-tox-pep8/13d0c87/ : FAILURE in 3m 59s","accounts_in_message":[],"_revision_number":11},{"id":"318bdff31a5aa49a5460998e57c3b13c4bc8677e","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-04-01 02:41:06.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"1021107d632ab4ac937704d43b234bc004f1b272","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 02:46:10.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/12/check/openstack-tox-docs/c4ec49c/html/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/12/check/openstack-tox-pep8/1778a51/ : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":12},{"id":"87353602eb7d14d7496838d517a4f6115686e376","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-04-01 02:46:34.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"441b3e210348d93294808bc92103ec2bc7497e12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 02:55:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/13/check/openstack-tox-docs/f1af8c5/html/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/13/check/openstack-tox-pep8/18bcb0b/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":13},{"id":"e958b24332696e78834b558f883db8509b80948c","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-04-01 04:30:23.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"e1288b0e202530bf30d7e8cb937e46d39dffac0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 04:37:57.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/14/check/openstack-tox-docs/5656ad0/html/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/14/check/openstack-tox-pep8/c95a191/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":14},{"id":"3688015ad92f9b0375d0760d76c8702af17fc40b","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-04-01 05:24:04.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"6a4e95f9d32e224d369ed48b07861d6a4e94fcf9","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-04-01 05:45:08.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"4a5fa8dc8f02b23610297a2a380eff8772c3194a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 05:49:10.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/16/check/openstack-tox-docs/c4f74f6/html/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/16/check/openstack-tox-pep8/aa69b0e/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":16},{"id":"75c55cd9fa27508c15022289687815d4d403d17f","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-04-02 05:09:16.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"cb54bd8913e37acb5c7bd10f60fa8e70b532dbe0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 05:25:28.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/17/check/openstack-tox-docs/4fb5df5/html/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/17/check/openstack-tox-pep8/4cae7a3/ : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":17},{"id":"9db54bdc0ca8af7913b13e034738d5b372d39529","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-04-02 05:29:33.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"d2162289efcbe62dc41d2a0bfc29d2ccb2bcb55f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 05:38:20.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/18/check/openstack-tox-docs/2a79b23/html/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/18/check/openstack-tox-pep8/ac39a65/ : FAILURE in 3m 45s","accounts_in_message":[],"_revision_number":18},{"id":"7a3e29428dc70b47e54bae26a1373dae50e40514","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-04-02 06:13:00.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"118e52ac9ed08f54374b95b0c2b9b20b419448ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 06:27:03.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/19/check/openstack-tox-docs/96c57d2/html/ : SUCCESS in 6m 22s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/19/check/openstack-tox-pep8/ea8a257/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":19},{"id":"a3cd19a094bc218e42beb99efb774cac4d138c7c","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-04-02 06:34:04.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"9c837c2cbcbfeae0c96e162dfb7ac1f74054a6be","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-09 10:19:15.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e950ced216d21538a6e3cc2c0e93d054e88b7003","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-09 10:22:29.000000000","message":"Patch Set 19: Code-Review-1\n\nKindly take care of previous comments as well.","accounts_in_message":[],"_revision_number":19},{"id":"2cbd3aeead8e47f2ffd5664dc7c501147dba134c","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-05-09 12:32:19.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"8157b8bb70c84e7f9af284ee82d6dae9d7de852a","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-05-09 12:34:20.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"13dacaffb6bde64162252962bb394afd87747ef0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:45:15.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/20/check/openstack-tox-docs/0b866ec/html/ : SUCCESS in 5m 59s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/20/check/openstack-tox-pep8/dca7941/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":20},{"id":"77010e8e91b92cab1ec4917648d2bdeca4990972","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-27 15:32:49.000000000","message":"Patch Set 20: Code-Review+1\n\n(2 comments)\n\nFew nits.","accounts_in_message":[],"_revision_number":20},{"id":"cf8aadbebc13c4e52317017f2308f587a4aa75f3","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-05-28 00:59:21.000000000","message":"Patch Set 20:\n\n(9 comments)\n\nThanks for good spec, hyunsik. Please kindly have a look at my comments.","accounts_in_message":[],"_revision_number":20},{"id":"5288169fb52acf7bce1d62899a55e1fec24b2f76","author":{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},"date":"2019-05-28 01:36:26.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"c8a45a9845b0593e0bf5f2db4c600da4df999c42","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-05-28 04:47:15.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"cc457fa2b60c75a2c92f29799ad7006692d54126","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 04:52:33.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/21/check/openstack-tox-docs/e33d8ed/html/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/21/check/openstack-tox-pep8/c7a0491/ : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":21},{"id":"94edc17bfb094a3c2a87c4192afd079efc31f620","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2019-05-28 04:54:09.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"726f85d247a2c8559175c851a524c736ff81aa54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 05:03:21.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/22/check/openstack-tox-docs/224616e/html/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/22/check/openstack-tox-pep8/cb9aed1/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":22},{"id":"f953d912ba595ea72ea4c03c06b1f2f7d91c36f2","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-29 05:54:15.000000000","message":"Patch Set 22: Code-Review+2\n\nPlease start on code side.","accounts_in_message":[],"_revision_number":22},{"id":"d9aa1f9c1115e8229a8e11ed44b93754e07039ee","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2019-06-25 10:13:43.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":22},{"id":"6724b8b15e38fbaa428ee0be81cfc1b24d88373f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 10:13:56.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"de8b33c479c7c54464737804ab8f8fd5259b4f43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 10:24:32.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/636272/22/gate/openstack-tox-docs/aace053/html/ : SUCCESS in 2m 24s\n- openstack-tox-pep8 http://logs.openstack.org/72/636272/22/gate/openstack-tox-pep8/6013a80/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":22},{"id":"02dd8c8af5ef387105d166f23284c09a521a819b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 10:24:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"60cbeb210c6261d70c451dedf2954e8b017e2527","revisions":{"8f0939014ae1be4e50b8a04befe848117b006ccf":{"kind":"REWORK","_number":1,"created":"2019-02-12 06:52:05.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/1"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:43:19.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew featrue to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8f0939014ae1be4e50b8a04befe848117b006ccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8f0939014ae1be4e50b8a04befe848117b006ccf"}]},"branch":"refs/heads/master"},"84c11e4000994791c38fa7551f0dc99c614fa923":{"kind":"REWORK","_number":2,"created":"2019-02-12 07:07:43.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/2"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 07:03:04.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/84c11e4000994791c38fa7551f0dc99c614fa923"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/84c11e4000994791c38fa7551f0dc99c614fa923"}]},"branch":"refs/heads/master"},"19c111b9cb352ee73385f8f2056b86b37645c1de":{"kind":"REWORK","_number":3,"created":"2019-02-12 08:13:11.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/3"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 08:08:45.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/19c111b9cb352ee73385f8f2056b86b37645c1de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/19c111b9cb352ee73385f8f2056b86b37645c1de"}]},"branch":"refs/heads/master"},"6286f149169cf8f1095e849af69f2256213ebb40":{"kind":"REWORK","_number":4,"created":"2019-02-12 08:54:10.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/4"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 08:49:41.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6286f149169cf8f1095e849af69f2256213ebb40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6286f149169cf8f1095e849af69f2256213ebb40"}]},"branch":"refs/heads/master"},"bdb1293cc30cd81599de2a3216279658571ce117":{"kind":"REWORK","_number":5,"created":"2019-02-12 09:41:46.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/5"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 09:37:21.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/bdb1293cc30cd81599de2a3216279658571ce117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/bdb1293cc30cd81599de2a3216279658571ce117"}]},"branch":"refs/heads/master"},"7feadebfc1680995b9a5e0b2106692830c2b0490":{"kind":"REWORK","_number":6,"created":"2019-02-12 10:29:25.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/6"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 10:24:59.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/7feadebfc1680995b9a5e0b2106692830c2b0490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/7feadebfc1680995b9a5e0b2106692830c2b0490"}]},"branch":"refs/heads/master"},"132e9336a4996d657ee5f2c300a99e13d822f53c":{"kind":"REWORK","_number":7,"created":"2019-03-16 06:54:39.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/7"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-03-16 06:48:31.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/132e9336a4996d657ee5f2c300a99e13d822f53c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/132e9336a4996d657ee5f2c300a99e13d822f53c"}]},"branch":"refs/heads/master"},"21e0d4652bc80b153e2a9b3426d6bd43292868a9":{"kind":"REWORK","_number":8,"created":"2019-03-16 07:09:11.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/8"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-03-16 07:09:08.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/21e0d4652bc80b153e2a9b3426d6bd43292868a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/21e0d4652bc80b153e2a9b3426d6bd43292868a9"}]},"branch":"refs/heads/master"},"22c3f5bbd0ffb038e583fd23dcd633d02af2a5f1":{"kind":"REWORK","_number":9,"created":"2019-03-19 12:53:10.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/9"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-03-19 12:52:58.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/22c3f5bbd0ffb038e583fd23dcd633d02af2a5f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/22c3f5bbd0ffb038e583fd23dcd633d02af2a5f1"}]},"branch":"refs/heads/master"},"7cc406a8511ac7ef5611b4114e70a1d99b588abc":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-03-19 12:55:56.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/10"}}},"commit":{"parents":[{"commit":"6dc97cc61f013f3785f92f4dcd9f051c798d3cc1","subject":"Postpone tacker-studio for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6dc97cc61f013f3785f92f4dcd9f051c798d3cc1"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","date":"2019-03-19 12:55:56.000000000","tz":0},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/7cc406a8511ac7ef5611b4114e70a1d99b588abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/7cc406a8511ac7ef5611b4114e70a1d99b588abc"}]},"branch":"refs/heads/master"},"c90c98039844e52a1cbbb06d034b6ad1cbc47762":{"kind":"REWORK","_number":11,"created":"2019-04-01 02:32:23.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/11"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 02:31:15.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\nstoryboard https://storyboard.openstack.org/#!/story/2005220\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c90c98039844e52a1cbbb06d034b6ad1cbc47762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c90c98039844e52a1cbbb06d034b6ad1cbc47762"}]},"branch":"refs/heads/master"},"c60070e9ca0132e99b4fbb78ae9c9b0c7715ceab":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2019-04-01 02:41:06.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/12"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 02:39:46.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c60070e9ca0132e99b4fbb78ae9c9b0c7715ceab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c60070e9ca0132e99b4fbb78ae9c9b0c7715ceab"}]},"branch":"refs/heads/master"},"5a1c94a81a83116ef02a665b1ae9e26b08fd914d":{"kind":"REWORK","_number":13,"created":"2019-04-01 02:46:34.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/13"}}},"commit":{"parents":[{"commit":"37e35389ebcda9344c8408e1a2f07028085a3307","subject":"Add reservation-vnfm spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37e35389ebcda9344c8408e1a2f07028085a3307"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 02:46:22.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/5a1c94a81a83116ef02a665b1ae9e26b08fd914d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/5a1c94a81a83116ef02a665b1ae9e26b08fd914d"}]},"branch":"refs/heads/master"},"92824664503719eb310262a55c363e0f04dc25c1":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-04-01 04:30:23.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/14"}}},"commit":{"parents":[{"commit":"6dc97cc61f013f3785f92f4dcd9f051c798d3cc1","subject":"Postpone tacker-studio for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6dc97cc61f013f3785f92f4dcd9f051c798d3cc1"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 04:30:23.000000000","tz":0},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/92824664503719eb310262a55c363e0f04dc25c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/92824664503719eb310262a55c363e0f04dc25c1"}]},"branch":"refs/heads/master"},"8ff53ef546c86c0903825a8ce05fb35166aee866":{"kind":"NO_CHANGE","_number":15,"created":"2019-04-01 05:24:04.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/15"}}},"commit":{"parents":[{"commit":"6dc97cc61f013f3785f92f4dcd9f051c798d3cc1","subject":"Postpone tacker-studio for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6dc97cc61f013f3785f92f4dcd9f051c798d3cc1"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-02-12 06:42:46.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 05:23:51.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8ff53ef546c86c0903825a8ce05fb35166aee866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8ff53ef546c86c0903825a8ce05fb35166aee866"}]},"branch":"refs/heads/master"},"aca78c951906bd395bce8d73e96b03c40cdbef7c":{"kind":"NO_CHANGE","_number":16,"created":"2019-04-01 05:45:08.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/16"}}},"commit":{"parents":[{"commit":"6dc97cc61f013f3785f92f4dcd9f051c798d3cc1","subject":"Postpone tacker-studio for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6dc97cc61f013f3785f92f4dcd9f051c798d3cc1"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 05:44:38.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 05:44:38.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/aca78c951906bd395bce8d73e96b03c40cdbef7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/aca78c951906bd395bce8d73e96b03c40cdbef7c"}]},"branch":"refs/heads/master"},"730ddef263ce3c397ca3a33b64bf88d8c300af56":{"kind":"REWORK","_number":17,"created":"2019-04-02 05:09:16.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/17"}}},"commit":{"parents":[{"commit":"6dc97cc61f013f3785f92f4dcd9f051c798d3cc1","subject":"Postpone tacker-studio for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6dc97cc61f013f3785f92f4dcd9f051c798d3cc1"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 05:44:38.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-02 05:09:04.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/730ddef263ce3c397ca3a33b64bf88d8c300af56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/730ddef263ce3c397ca3a33b64bf88d8c300af56"}]},"branch":"refs/heads/master"},"45b84c3324afd519328ab8be4c054357947978d2":{"kind":"REWORK","_number":18,"created":"2019-04-02 05:29:33.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/18"}}},"commit":{"parents":[{"commit":"6dc97cc61f013f3785f92f4dcd9f051c798d3cc1","subject":"Postpone tacker-studio for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6dc97cc61f013f3785f92f4dcd9f051c798d3cc1"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 05:44:38.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-02 05:29:15.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/45b84c3324afd519328ab8be4c054357947978d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/45b84c3324afd519328ab8be4c054357947978d2"}]},"branch":"refs/heads/master"},"765ba172cc3e89feb74c7d1e6cdf393e1a615777":{"kind":"REWORK","_number":19,"created":"2019-04-02 06:13:00.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/19"}}},"commit":{"parents":[{"commit":"6dc97cc61f013f3785f92f4dcd9f051c798d3cc1","subject":"Postpone tacker-studio for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6dc97cc61f013f3785f92f4dcd9f051c798d3cc1"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 05:44:38.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-02 06:12:50.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/765ba172cc3e89feb74c7d1e6cdf393e1a615777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/765ba172cc3e89feb74c7d1e6cdf393e1a615777"}]},"branch":"refs/heads/master"},"8e9c508bdc88bbf45910032e344345cbf23ff753":{"kind":"REWORK","_number":20,"created":"2019-05-09 12:34:20.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/20"}}},"commit":{"parents":[{"commit":"43f4c7540e8b23a91ae27932d24d5d826ba026aa","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/43f4c7540e8b23a91ae27932d24d5d826ba026aa"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 05:44:38.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-05-09 12:34:07.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8e9c508bdc88bbf45910032e344345cbf23ff753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8e9c508bdc88bbf45910032e344345cbf23ff753"}]},"branch":"refs/heads/master"},"4a7b4c24698432720ca574684702cfd2eab67d69":{"kind":"REWORK","_number":21,"created":"2019-05-28 04:47:15.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/21"}}},"commit":{"parents":[{"commit":"43f4c7540e8b23a91ae27932d24d5d826ba026aa","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/43f4c7540e8b23a91ae27932d24d5d826ba026aa"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 05:44:38.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-05-28 04:46:57.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/4a7b4c24698432720ca574684702cfd2eab67d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/4a7b4c24698432720ca574684702cfd2eab67d69"}]},"branch":"refs/heads/master"},"60cbeb210c6261d70c451dedf2954e8b017e2527":{"kind":"REWORK","_number":22,"created":"2019-05-28 04:54:09.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/72/636272/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/72/636272/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/72/636272/22"}}},"commit":{"parents":[{"commit":"43f4c7540e8b23a91ae27932d24d5d826ba026aa","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/43f4c7540e8b23a91ae27932d24d5d826ba026aa"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-04-01 05:44:38.000000000","tz":540},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2019-05-28 04:53:55.000000000","tz":540},"subject":"Adding Multi-Interface for Containerized VNF in Tacker","message":"Adding Multi-Interface for Containerized VNF in Tacker\n\nNew feature to support multi-interface for containeriezed VNF.\n\nBlueprint multi-interface-container\n\nChange-Id: I83347cdc7c7ed4c560d8cee412edf575696eb038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/60cbeb210c6261d70c451dedf2954e8b017e2527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/60cbeb210c6261d70c451dedf2954e8b017e2527"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
