)]}'
{"id":"openstack%2Ftacker-specs~731658","triplet_id":"openstack%2Ftacker-specs~master~Icc79264fa7bfd207a51f83319cfbaa5b1bb87a21","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-etsi-nfv-specs","hashtags":[],"change_id":"Icc79264fa7bfd207a51f83319cfbaa5b1bb87a21","subject":"Enhance VNF LCM API based on ETSI NFV-SOL specification","status":"MERGED","created":"2020-05-29 06:15:20.000000000","updated":"2020-07-01 12:33:00.000000000","submitted":"2020-07-01 12:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731658-1593606642383-b4231495","meta_rev_id":"ac6a1b819795644e8ffdc00e1ed7b92255d758e3","_number":731658,"virtual_id_number":731658,"owner":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-01 12:30:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2020-06-19 06:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":2,"date":"2020-07-01 12:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":1,"date":"2020-07-01 12:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 06:28:23.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-07-01 12:24:29.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2020-07-01 12:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5d90f02fffee83fd85bc69816e59692d50e5e04","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-05-29 06:15:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11fd5b1e7fb5cebe2b131d15d77a2800911f28f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 06:23:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4541c9cb96da42c5b4bc5a7c1accdefe : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac2787ab7af94240aedbd63af8a1d52c : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"11631f5282cce23ef4493263f608b22c57044650","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-05-29 06:27:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76fd1a9e851e1c05f32f76c3d2cd0eaf21f05bc0","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-05-29 06:51:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5796e2e30716f17dd73ddfccd837bcf037d56c3c","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-05-29 06:52:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"15b61522024f30ff86d8fb69b57a4ac88bf30523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 06:57:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd0fe20a4bc4b90806d3df7d230da41 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee6408c47224453499a6294771847b45 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"120945689aabcadac321f8896e0dd2226b5ccd8a","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-05-29 10:26:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cfe7bc78bdf745ca11d0db37d451a7f8e86a414a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 10:31:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df2dd66df4f45f0ab8d69bd659f594e : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d75de8df2364237bf606d34a3a7401c : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"dea90ff8c53cb1be4e03f93b69b14a5f067a44d9","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-06-18 18:51:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"72b4778eda3b123443e53359cf4a9ac49ad9d084","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-06-19 02:19:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a391df10e7318f284a9602f7aac093915e07bdaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 02:24:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a32225f7d540919a44b32c3d55bd88 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381e868d416841939545f501fa0e3428 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":4},{"id":"84d0611baf831fd0bb35ee3dae163d819e0692a8","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-06-19 06:28:23.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6659e1cec1a1d9af14f198164a07a14454e8c862","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2020-07-01 12:24:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"71f94644850cadad0f12730eb1ba7ea4fe968c6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 12:24:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cc5203505ca7f265bde2905ad0d59af98dde75a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 12:30:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b5c7d60944956838f4b62c7f0eafd1f3c7190c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 12:30:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439e453582a74e0b941173200dce91aa : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419805c69fbe4178987dd7e8638c936b : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":4},{"id":"a75b8f2564f57e1466bcc06840536b1e0dacc3e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 12:33:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/6e5704922c004b538024aab994285e5d : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"af974a0a42edf9dd63b6beaa7eea94f000e867e0","revisions":{"837077af10c9436df30696e180bd0ae8cabb8c49":{"kind":"REWORK","_number":1,"created":"2020-05-29 06:15:20.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/58/731658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/58/731658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/1"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 05:59:00.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 06:15:16.000000000","tz":540},"subject":"Enhance VNF LCM API based on ETSI NFV-SOL specification","message":"Enhance VNF LCM API based on ETSI NFV-SOL specification\n\nThis spec aim to implement remaining VNF LCM API features\nthat are not implemented in Ussuri release.\n\n* Support remaining API message-body attributes\n* Support \"vnfPkgId\" to related LCM APIs\n* Enable URI query parameters based on NFV-SOL v2.4.1\n\nChange-Id: Icc79264fa7bfd207a51f83319cfbaa5b1bb87a21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/837077af10c9436df30696e180bd0ae8cabb8c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/837077af10c9436df30696e180bd0ae8cabb8c49"}]},"branch":"refs/heads/master"},"ef14ef1954b83c37413fae49e002cb02afd6faf7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-29 06:51:56.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/58/731658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/58/731658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/2"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 05:59:00.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 06:51:32.000000000","tz":540},"subject":"Enhance VNF LCM API based on ETSI NFV-SOL specification","message":"Enhance VNF LCM API based on ETSI NFV-SOL specification\n\nThis spec aim to implement remaining VNF LCM API features\nthat are not implemented in Ussuri release.\n\n* Support remaining API message-body attributes\n* Support \"vnfPkgId\" to related LCM APIs\n* Enable URI query parameters based on NFV-SOL v2.4.1\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: Icc79264fa7bfd207a51f83319cfbaa5b1bb87a21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/ef14ef1954b83c37413fae49e002cb02afd6faf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/ef14ef1954b83c37413fae49e002cb02afd6faf7"}]},"branch":"refs/heads/master"},"840ecd4192d9be15969f64c84d2e07291b6b49ac":{"kind":"REWORK","_number":3,"created":"2020-05-29 10:26:38.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/58/731658/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/58/731658/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/3"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 05:59:00.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:26:11.000000000","tz":540},"subject":"Enhance VNF LCM API based on ETSI NFV-SOL specification","message":"Enhance VNF LCM API based on ETSI NFV-SOL specification\n\nThis spec aim to implement remaining VNF LCM API features\nthat are not implemented in Ussuri release.\n\n* Support remaining API message-body attributes\n* Support \"vnfPkgId\" to related LCM APIs\n* Enable URI query parameters based on NFV-SOL v2.4.1\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: Icc79264fa7bfd207a51f83319cfbaa5b1bb87a21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/840ecd4192d9be15969f64c84d2e07291b6b49ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/840ecd4192d9be15969f64c84d2e07291b6b49ac"}]},"branch":"refs/heads/master"},"af974a0a42edf9dd63b6beaa7eea94f000e867e0":{"kind":"REWORK","_number":4,"created":"2020-06-19 02:19:28.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/58/731658/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/58/731658/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/58/731658/4"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 05:59:00.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-06-19 02:18:43.000000000","tz":540},"subject":"Enhance VNF LCM API based on ETSI NFV-SOL specification","message":"Enhance VNF LCM API based on ETSI NFV-SOL specification\n\nThis spec aim to implement remaining VNF LCM API features\nthat are not implemented in Ussuri release.\n\n* Support remaining API message-body attributes\n* Support \"vnfPkgId\" to related LCM APIs\n* Enable URI query parameters based on NFV-SOL v2.4.1\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: Icc79264fa7bfd207a51f83319cfbaa5b1bb87a21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/af974a0a42edf9dd63b6beaa7eea94f000e867e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/af974a0a42edf9dd63b6beaa7eea94f000e867e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
