)]}'
{"id":"openstack%2Ftacker-specs~731715","triplet_id":"openstack%2Ftacker-specs~master~I851df2373bd4df01aa45c40506db47fa572723c7","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-etsi-nfv-specs","hashtags":[],"change_id":"I851df2373bd4df01aa45c40506db47fa572723c7","subject":"Support error handling based on ETSI NFV-SOL specification","status":"MERGED","created":"2020-05-29 10:10:09.000000000","updated":"2020-09-07 15:53:19.000000000","submitted":"2020-09-07 15:50:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731715-1599493859338-69a35627","meta_rev_id":"738088bbdc99f94c473583c64e76209707e1f2b2","_number":731715,"virtual_id_number":731715,"owner":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-07 15:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2020-09-01 17:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2020-08-25 10:16:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":2,"date":"2020-09-07 15:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"date":"2020-09-03 02:53:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":1,"date":"2020-09-07 15:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 10:16:42.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2020-09-01 17:59:25.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-09-03 02:53:11.000000000","updated_by":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"reviewer":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"state":"REVIEWER"},{"updated":"2020-09-07 15:39:45.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2020-09-07 15:50:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"312d9f4b92ea448fdbf72c38442f69b6a011a2f1","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-05-29 10:10:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29159d2dba5aa98d0edfb4ee372ee29226d24228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 10:15:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e76b6843463243c1a9f0eebdde5c81c8 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1026029e461240f3a9fcf660a10055bf : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b516b7a8cbad5c039d3c4f88dbeb8e03f9220360","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-05-29 10:29:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4becde77df372b51f9786483f004b5f76781a5be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 10:33:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea78113419d4848b5c055fe41f74900 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405442018ebd41679b3596775296a970 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ae916c20df1e02907ff0a9900fce9fe1be76a778","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-06-18 18:52:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b1e64249bafc3b88b066dac4c1d9d2bc6d531578","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2020-07-18 15:33:50.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ed52daf8d1eab792dfcc429f10c66336b4fbd8c0","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-07-28 11:47:24.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ec001c6535ea0bfd0203fbe421d6c2c25cb388a4","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-07-28 11:48:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5fb361737f4c7ab1e23d70669e704338c98137ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 11:55:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb31140212a41bebb95e16494b2a4d2 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e21fa4ee6a480a9ae212adcbc25b43 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"2598e2d52c0ed5cd7c18469aa1ef056355648274","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-08-06 06:24:54.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nI have a naive query, as per the documentation tacker will be supporting on-demand rollback. \nIs this understanding correct? \nIf yes then are we considering automatic rollback as well in tacker.","accounts_in_message":[],"_revision_number":3},{"id":"3e769efb7be8b02aa2cb69cb00ef0cec54dfdcd7","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-08-07 12:12:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2017bff2a87daf6dc7973a8a56ab5d4bf400611c","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2020-08-09 14:38:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7baa1d6ff220eff16322af9784ed6597f4eb4d41","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-08-19 07:51:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3ebfbfd516157afa8c4d6dba1cec4acc5387dc8","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-08-25 05:37:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9c677d8115d0062109f4eb5bbf74babc27c6a862","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-08-25 06:10:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"67474d83aa16d4a71bca5690374eea972626a2d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 06:34:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602776707f99470eb6cddd6a3e35e7a1 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed44532f25a44319ce0107a943b173a : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":4},{"id":"7d33fa5162acf39108782993bc8abffe59db1db6","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-08-25 10:16:42.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":4},{"id":"3a90f64dbf60bbca1712522f675cd00e56a585f2","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-01 17:59:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c198e0698181f4300e5383442197490b60f53995","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2020-09-07 15:39:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9bd532008d0045cff63a69837da82e31ca9d94a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 15:39:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a70cba21f7c063ad8412606a40cdad39c036d9ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 15:50:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c37a8bfed34c13afcb9555964bcd48 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a856dcb7e5b4f43be603fec6ef9486e : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":4},{"id":"3fea47c4bac000689e2ee230b8715c1b254a0f5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 15:50:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8837bf60fad15dfcc2ed78d9af64fcc9570d4b1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 15:53:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/a9924f11211943578a2de5b2ce8533a5 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"87ca8d00ab8ea4b279c334d061a545b3f61f351a","revisions":{"22377ee2f8ac6a4730d7a524c4f40e22a82998f3":{"kind":"REWORK","_number":1,"created":"2020-05-29 10:10:09.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/15/731715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/15/731715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/1"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:04:38.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:09:24.000000000","tz":540},"subject":"Support error handling based on ETSI NFV-SOL specification","message":"Support error handling based on ETSI NFV-SOL specification\n\nIntroduces a new interface to execute rollback operation for VNF\nlifecycle management operations.\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: I851df2373bd4df01aa45c40506db47fa572723c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/22377ee2f8ac6a4730d7a524c4f40e22a82998f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/22377ee2f8ac6a4730d7a524c4f40e22a82998f3"}]},"branch":"refs/heads/master"},"376c439cc292a3bdb1a650a93e9cead96cf116b9":{"kind":"REWORK","_number":2,"created":"2020-05-29 10:29:24.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/15/731715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/15/731715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/2"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:04:38.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:29:11.000000000","tz":540},"subject":"Support error handling based on ETSI NFV-SOL specification","message":"Support error handling based on ETSI NFV-SOL specification\n\nIntroduces a new interface to execute rollback operation for VNF\nlifecycle management operations.\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: I851df2373bd4df01aa45c40506db47fa572723c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/376c439cc292a3bdb1a650a93e9cead96cf116b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/376c439cc292a3bdb1a650a93e9cead96cf116b9"}]},"branch":"refs/heads/master"},"6ea851984648a2f3edc82b4bc7f671a2d7870d2a":{"kind":"REWORK","_number":3,"created":"2020-07-28 11:48:45.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/15/731715/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/15/731715/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/3"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:04:38.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-07-28 11:45:55.000000000","tz":540},"subject":"Support error handling based on ETSI NFV-SOL specification","message":"Support error handling based on ETSI NFV-SOL specification\n\nIntroduces a new interface to execute rollback operation for VNF\nlifecycle management operations.\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: I851df2373bd4df01aa45c40506db47fa572723c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6ea851984648a2f3edc82b4bc7f671a2d7870d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6ea851984648a2f3edc82b4bc7f671a2d7870d2a"}]},"branch":"refs/heads/master"},"87ca8d00ab8ea4b279c334d061a545b3f61f351a":{"kind":"REWORK","_number":4,"created":"2020-08-25 06:10:05.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/15/731715/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/15/731715/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/15/731715/4"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:04:38.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-08-25 06:10:01.000000000","tz":540},"subject":"Support error handling based on ETSI NFV-SOL specification","message":"Support error handling based on ETSI NFV-SOL specification\n\nIntroduces a new interface to execute rollback operation for VNF\nlifecycle management operations.\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: I851df2373bd4df01aa45c40506db47fa572723c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/87ca8d00ab8ea4b279c334d061a545b3f61f351a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/87ca8d00ab8ea4b279c334d061a545b3f61f351a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
