)]}'
{"id":"openstack%2Ftacker-specs~731718","triplet_id":"openstack%2Ftacker-specs~master~I9caf0a2a4f11e38280f7c1219ce98d659f528db6","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-etsi-nfv-specs","hashtags":[],"change_id":"I9caf0a2a4f11e38280f7c1219ce98d659f528db6","subject":"Support LCM notification based on ETSI NFV-SOL specification","status":"MERGED","created":"2020-05-29 10:21:23.000000000","updated":"2020-08-05 11:57:53.000000000","submitted":"2020-07-20 03:09:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731718-1595214554439-98bbf7d3","meta_rev_id":"016bc28040b4005438ecfac766d1d12070189da7","_number":731718,"virtual_id_number":731718,"owner":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 03:09:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2020-07-20 03:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-18 15:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":1,"date":"2020-07-20 03:01:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 18:53:25.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2020-07-18 15:24:42.000000000","updated_by":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"reviewer":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"state":"REVIEWER"},{"updated":"2020-07-20 03:01:07.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-07-20 03:09:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26f2fa72980880ad09b1cfc3350c57285c2e555e","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-05-29 10:21:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"292090b471abf313efb8c47d84b82f5b38b7c79a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 10:27:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb3760a3c399434ea86b984334aa8609 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd50dd897f04bae84784e30935a08ca : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e379ea16b06baa3ceff293de9b7be73acfb64f87","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-06-18 18:53:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"017b131fce41c7ad1478b0c49dbb3db2d492ed89","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-07-01 05:23:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32f066f1f4b150d9ee0d8458e8e96ed3144114a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 05:27:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/738ac7e4fe9042a885f9967ec24641cb : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b960a4d1318a42ed8d00ccb7e506bfd0 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a54ed4916e18b8ad1221d0d6980b3fd2b1a33766","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2020-07-05 15:07:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b60415cdc4d2aafe18506eb0d19083c925b6e6f1","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-07-14 06:36:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8909ce39ab869414822cc541671fa77a5373f80","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-07-16 04:18:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aefe42aa3aab3eabe94e5f993e70776e623b3350","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-07-16 04:19:29.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b693f12bb45a2228eb57f491d5e13889ed23fbb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 04:27:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37eb5f35450d40799c06a6e24e45b9ad : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f351f305549c4c08bca4276578b94493 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":3},{"id":"875035707c20042053d33bd9c5af8b1a0950d81b","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2020-07-18 15:24:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d76fbbc7069b4441585833c7bd89a81289b1358f","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-07-20 03:01:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"04fd84bbebdbed9ab00648fafbcd30141b30751f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 03:01:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"47fe89f7516a97b20c5233ed3b02e87bde8e2411","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 03:09:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f45c700b40a6fb954644e47cb772412439ab6b71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 03:09:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da780ac319644a98f6ea3778de56630 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae8b6c597a343bf98b75ef1968caeed : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"565891212716fc9fb68b41d3fc4b5f9ae3ac6bd0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 03:10:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e97bf0ecd4a4405e97ffb12825655be3 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3},{"id":"d130fd218aae6fc8567d25fbc1e1489e42775778","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2020-08-05 11:57:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f7b6c84b6dbf05cf6846c860cb13b307da35c015","revisions":{"b0efb128a5be9efaa64a63e570fedda2393abd90":{"kind":"REWORK","_number":1,"created":"2020-05-29 10:21:23.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/18/731718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/18/731718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/1"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:16:32.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:20:50.000000000","tz":540},"subject":"Support LCM notification based on ETSI NFV-SOL specification","message":"Support LCM notification based on ETSI NFV-SOL specification\n\nIntroduce new interfaces to notify LCM operation status of VNF\ninstances towards clients.\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: I9caf0a2a4f11e38280f7c1219ce98d659f528db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b0efb128a5be9efaa64a63e570fedda2393abd90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b0efb128a5be9efaa64a63e570fedda2393abd90"}]},"branch":"refs/heads/master"},"f5545160575498dc8ffcdf45b556b7d50c1cb298":{"kind":"REWORK","_number":2,"created":"2020-07-01 05:23:14.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/18/731718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/18/731718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/2"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:16:32.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-07-01 05:22:51.000000000","tz":540},"subject":"Support LCM notification based on ETSI NFV-SOL specification","message":"Support LCM notification based on ETSI NFV-SOL specification\n\nIntroduce new interfaces to notify LCM operation status of VNF\ninstances towards clients.\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: I9caf0a2a4f11e38280f7c1219ce98d659f528db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f5545160575498dc8ffcdf45b556b7d50c1cb298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f5545160575498dc8ffcdf45b556b7d50c1cb298"}]},"branch":"refs/heads/master"},"f7b6c84b6dbf05cf6846c860cb13b307da35c015":{"kind":"REWORK","_number":3,"created":"2020-07-16 04:18:52.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/18/731718/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/18/731718/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/18/731718/3"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-05-29 10:16:32.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-07-16 04:18:22.000000000","tz":540},"subject":"Support LCM notification based on ETSI NFV-SOL specification","message":"Support LCM notification based on ETSI NFV-SOL specification\n\nIntroduce new interfaces to notify LCM operation status of VNF\ninstances towards clients.\n\nBlueprint: support-etsi-nfv-specs\nChange-Id: I9caf0a2a4f11e38280f7c1219ce98d659f528db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f7b6c84b6dbf05cf6846c860cb13b307da35c015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f7b6c84b6dbf05cf6846c860cb13b307da35c015"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
