)]}'
{"id":"openstack%2Ftacker-specs~733454","triplet_id":"openstack%2Ftacker-specs~master~I51273e4991906f4a5f774ee40fe561eaa8dc994b","project":"openstack/tacker-specs","branch":"master","topic":"bp/cnf-support-with-etsi-nfv-specs","hashtags":[],"change_id":"I51273e4991906f4a5f774ee40fe561eaa8dc994b","subject":"DNM: CNF on Kubernetes cluster inside VM","status":"ABANDONED","created":"2020-06-04 07:10:53.000000000","updated":"2020-09-15 00:31:42.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4e3b2c091a0663ad2022f3f540055cad361ea13b","_number":733454,"virtual_id_number":733454,"owner":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"date":"2020-08-09 17:36:16.000000000","_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"date":"2020-08-19 09:04:51.000000000","_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-07 03:20:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 08:48:11.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-08-07 03:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-09 17:36:16.000000000","updated_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"}],"messages":[{"id":"0e8e5804c630a072787b56ba9b91a4c5cb48db4f","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-06-04 07:10:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"308fdee0a06df10938fb0e5171d4c73b732ab279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 07:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54fe13c3f5cb4ae5a33a8e06e6ac60e1 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c8c2b269d1a482face55c7776747211 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1bc166d04a2e7e7d2e2b6253a5fb7e6651226124","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-07-06 16:43:15.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6630ab285c76b4285a55a8c1cdb48e31e28c0051","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-07-21 23:59:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14f1146824acc780ddd412d618fd150c29428d92","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-07-22 00:00:10.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4b02a8e47d4982d343ac0ae5c85785d20d9a9566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 00:05:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9224a52dd3214dd1bd093f7ca001ec06 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eea7ab341e94b17b71ec226624ce610 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"deb95d26c21c3f0f3dbbff01fa43bf040f94cf93","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-07-27 02:14:34.000000000","message":"Patch Set 2: Workflow-1\n\nNeed to discuss the relationship between ActionDriver(https://review.opendev.org/#/c/740896/) and UserScript.","accounts_in_message":[],"_revision_number":2},{"id":"57bdfdbae879b44bd7c6562a4620e6647dee22a8","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-07-31 08:48:11.000000000","message":"Patch Set 2:\n\nCould you add prefix `[DNM]` to show it is not ready to be merged? Thanks.","accounts_in_message":[],"_revision_number":2},{"id":"14be8967a66ccc89c58a003b11668f2bb45e83ce","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-08-05 08:21:07.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9e76d1facdc1fee7a272b63767c3e94f90cfc52a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 08:29:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd310c9e5d947a192cb29f1ee1c3a9c : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffffbdc7244945dca9444d0dd072942b : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":3},{"id":"40d5ff95fdd131067ac7007eacf70caee86c7853","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-08-07 03:04:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad47169912a34512c57da51deb58e695cef7170e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 03:13:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ef22236e364e07a0944cb5ca10e4f9 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb0620d3c4b447c4bd8e2adeb667eba6 : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":4},{"id":"0b8bed51cf542d665f631bd1b5a9bfeef145c90a","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-08-07 03:16:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3d61c9708f43e523a1b1d295f1e4d66e424aca77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 03:20:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf73dfdfdb1e4ad9827a54fc0a899852 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cad34f981ad40bc90f6ecc6c8981649 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":5},{"id":"3b071f89d9bd18cb4565635cf7f5cc948fc76a67","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-08-07 04:11:08.000000000","message":"Patch Set 5:\n\nI updated the design of `instantiate_end` execution to utilize Management Driver to configure a Kubernetes cluster. This update doesn\u0027t change the purpose and the goal of this spec. The logic for the configuration will be implemented in Management Driver or a script called by Management Driver.","accounts_in_message":[],"_revision_number":5},{"id":"7de57d612312117946c425038e0138ecf898fd36","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-08-09 17:36:16.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThank you for updating.\nAs you know, this is related to our ActionDriver script.\nLet\u0027s continue to discuss.\n\nI have a question.","accounts_in_message":[],"_revision_number":5},{"id":"779836555ec1c01edb20c17ca25ae10c9d5a8fa1","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-08-19 09:04:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d5ca04b5afb27c1545599ffd6f1d5971601bf2b0","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-09-15 00:31:42.000000000","message":"Abandoned\n\nThis spec is postponed to W-release.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1323a389f6c71ca915009a951304b3c74af302d0","revisions":{"82ea2f2504721353d6b750aa8ec09362abe84208":{"kind":"REWORK","_number":1,"created":"2020-06-04 07:10:53.000000000","uploader":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"ref":"refs/changes/54/733454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/733454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/1"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-06-04 07:10:04.000000000","tz":540},"committer":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-06-04 07:10:04.000000000","tz":540},"subject":"CNF on Kubernetes cluster inside VM","message":"CNF on Kubernetes cluster inside VM\n\nThis specification describes enhancement of\nVNF Lifecycle Management for Container Network Function\non Kubernetes cluster in Virtual Machines.\n\nThe CNF instantiation/termination process will be splited into:\n* VNF-A: VMs and set up Kubernetes cluster\n* VNF-B: CNF on the Kubernetes cluster inside VNF-A\n\nChange-Id: I51273e4991906f4a5f774ee40fe561eaa8dc994b\nBlueprint: cnf-support-with-etsi-nfv-specs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/82ea2f2504721353d6b750aa8ec09362abe84208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/82ea2f2504721353d6b750aa8ec09362abe84208"}]},"branch":"refs/heads/master"},"1ffc409312a1bd96a33459286237fa98b750661f":{"kind":"REWORK","_number":2,"created":"2020-07-21 23:59:08.000000000","uploader":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"ref":"refs/changes/54/733454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/733454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/2"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-06-04 07:10:04.000000000","tz":540},"committer":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-07-21 23:57:21.000000000","tz":540},"subject":"CNF on Kubernetes cluster inside VM","message":"CNF on Kubernetes cluster inside VM\n\nThis specification describes enhancement of\nVNF Lifecycle Management for Container Network Function\non Kubernetes cluster in Virtual Machines.\n\nThe CNF instantiation/termination process will be splited into:\n* VNF-A: VMs and set up Kubernetes cluster\n* VNF-B: CNF on the Kubernetes cluster inside VNF-A\n\nChange-Id: I51273e4991906f4a5f774ee40fe561eaa8dc994b\nBlueprint: cnf-support-with-etsi-nfv-specs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ffc409312a1bd96a33459286237fa98b750661f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ffc409312a1bd96a33459286237fa98b750661f"}]},"branch":"refs/heads/master"},"6272a0e6e831cc6ba113997124ab101c7d12c2d1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-05 08:21:07.000000000","uploader":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"ref":"refs/changes/54/733454/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/733454/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/3"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-06-04 07:10:04.000000000","tz":540},"committer":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-08-05 08:20:45.000000000","tz":0},"subject":"DNM: CNF on Kubernetes cluster inside VM","message":"DNM: CNF on Kubernetes cluster inside VM\n\nThis specification describes enhancement of\nVNF Lifecycle Management for Container Network Function\non Kubernetes cluster in Virtual Machines.\n\nThe CNF instantiation/termination process will be splited into:\n* VNF-A: VMs and set up Kubernetes cluster\n* VNF-B: CNF on the Kubernetes cluster inside VNF-A\n\nChange-Id: I51273e4991906f4a5f774ee40fe561eaa8dc994b\nBlueprint: cnf-support-with-etsi-nfv-specs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6272a0e6e831cc6ba113997124ab101c7d12c2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6272a0e6e831cc6ba113997124ab101c7d12c2d1"}]},"branch":"refs/heads/master"},"1796d526af04b5561115839f3046008753404006":{"kind":"REWORK","_number":4,"created":"2020-08-07 03:04:47.000000000","uploader":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"ref":"refs/changes/54/733454/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/733454/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/4"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-06-04 07:10:04.000000000","tz":540},"committer":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-08-07 03:03:46.000000000","tz":540},"subject":"DNM: CNF on Kubernetes cluster inside VM","message":"DNM: CNF on Kubernetes cluster inside VM\n\nThis specification describes enhancement of\nVNF Lifecycle Management for Container Network Function\non Kubernetes cluster in Virtual Machines.\n\nThe CNF instantiation/termination process will be splited into:\n* VNF-A: VMs and set up Kubernetes cluster\n* VNF-B: CNF on the Kubernetes cluster inside VNF-A\n\nChange-Id: I51273e4991906f4a5f774ee40fe561eaa8dc994b\nBlueprint: cnf-support-with-etsi-nfv-specs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1796d526af04b5561115839f3046008753404006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1796d526af04b5561115839f3046008753404006"}]},"branch":"refs/heads/master"},"1323a389f6c71ca915009a951304b3c74af302d0":{"kind":"REWORK","_number":5,"created":"2020-08-07 03:16:03.000000000","uploader":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"ref":"refs/changes/54/733454/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/733454/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/733454/5"}}},"commit":{"parents":[{"commit":"96f0c2026979636ce4988345934cb9b1a7bf1491","subject":"Merge \"add Victoria specs directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/96f0c2026979636ce4988345934cb9b1a7bf1491"}]}],"author":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-06-04 07:10:04.000000000","tz":540},"committer":{"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","date":"2020-08-07 03:15:49.000000000","tz":540},"subject":"DNM: CNF on Kubernetes cluster inside VM","message":"DNM: CNF on Kubernetes cluster inside VM\n\nThis specification describes enhancement of\nVNF Lifecycle Management for Container Network Function\non Kubernetes cluster in Virtual Machines.\n\nThe CNF instantiation/termination process will be splited into:\n* VNF-A: VMs and set up Kubernetes cluster\n* VNF-B: CNF on the Kubernetes cluster inside VNF-A\n\nChange-Id: I51273e4991906f4a5f774ee40fe561eaa8dc994b\nBlueprint: cnf-support-with-etsi-nfv-specs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1323a389f6c71ca915009a951304b3c74af302d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1323a389f6c71ca915009a951304b3c74af302d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
