)]}'
{"id":"openstack%2Ftacker-specs~744611","triplet_id":"openstack%2Ftacker-specs~master~I9e25e9d550f6b12b164b01825419fa47d6b05a1b","project":"openstack/tacker-specs","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I9e25e9d550f6b12b164b01825419fa47d6b05a1b","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-04 03:32:26.000000000","updated":"2020-08-04 15:43:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ea91c549dab1f704de7298455e81ceea7f94f63b","_number":744611,"virtual_id_number":744611,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-04 04:55:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 04:55:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e75db086907d7414fc32f1bfc6d79b3d4a2ca546","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 03:32:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b9a67ad17f6553b7f0c892c2b61d4af99d19e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 04:55:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6084af422a03479c80b82acfbd9a04a3 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e821e68103764535b43f8a9fcb9b5f43 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"29a22e82e46cd9fffc753e75bfe93fbe4ae23d23","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 15:43:02.000000000","message":"Abandoned\n\nall working fine, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7a18e629519bdbb39ba34bffa62b26255291437","revisions":{"d7a18e629519bdbb39ba34bffa62b26255291437":{"kind":"REWORK","_number":1,"created":"2020-08-04 03:32:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/744611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/11/744611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/11/744611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/11/744611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/11/744611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/11/744611/1"}}},"commit":{"parents":[{"commit":"37784e2650296fa4022604dfc7f6ff2743c3373e","subject":"Merge \"Support LCM notification based on ETSI NFV-SOL specification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/37784e2650296fa4022604dfc7f6ff2743c3373e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 02:33:25.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 03:32:24.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40222\n\nDepends-On: https://review.opendev.org/#/c/734700/\nChange-Id: I9e25e9d550f6b12b164b01825419fa47d6b05a1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/d7a18e629519bdbb39ba34bffa62b26255291437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/d7a18e629519bdbb39ba34bffa62b26255291437"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
