)]}'
{"id":"openstack%2Ftacker-specs~750639","triplet_id":"openstack%2Ftacker-specs~master~I175b48de60cda6629db362d0819a980af363e8e8","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-etsi-nfv-specs","hashtags":[],"change_id":"I175b48de60cda6629db362d0819a980af363e8e8","subject":"Update notification spec to follow ETSI NFV SOL specs","status":"MERGED","created":"2020-09-09 12:25:43.000000000","updated":"2020-09-25 10:14:28.000000000","submitted":"2020-09-25 10:10:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750639-1601028649155-b51fe808","meta_rev_id":"3599f62cb6f5b68063d055fe83299d7f281f3681","_number":750639,"virtual_id_number":750639,"owner":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 10:10:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-17 05:08:15.000000000","_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2020-09-17 06:19:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2020-09-17 05:36:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2020-09-25 10:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":1,"date":"2020-09-25 10:04:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"date":"2020-09-17 06:06:38.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"date":"2020-09-17 06:19:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 05:36:54.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2020-09-17 06:06:38.000000000","updated_by":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"reviewer":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"state":"REVIEWER"},{"updated":"2020-09-17 06:19:23.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-09-17 06:19:31.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2020-09-25 10:04:06.000000000","updated_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2020-09-25 10:10:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a4d864df740bebb5b980d2d69c0653c75e6745c","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-09-09 12:25:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb6bc7c5c6f97c71733bde51672f6b0a4d0ccb85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 12:34:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03170e8b68cd44a6b35784387667f831 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b1ca56a4d641dd94390310917081ed : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fbf7e796c45f2852956598306c6cc44a63368b09","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-09-09 13:52:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"267e00808ef6cc8d7c9b0639d86b4f483ac4b436","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-10 18:29:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41aa429e6b4049ab81c16a5af38a9d62d7dbc164","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-10 18:34:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50a894e1aa1ee256d3d28e2775fcde534bad47da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:37:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/680a83c2c5e14eb8976495af484415aa : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc925d765e843cd91fee1148bf71a4b : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2d2a6982ca0570cd04d01991aab531bb28b824cb","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-09-11 04:05:02.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nThanks, Yasufumi Ogawa for running recheck.","accounts_in_message":[],"_revision_number":1},{"id":"8f0715032c54de407338dfe14e1b871270e83cd8","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-16 05:19:15.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"8259955b80238eb9ef6c9e2387d8b0aadb0e6c07","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-09-17 05:07:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"361fe8834e9eb02ad7efc0d66228678463d5c1af","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-09-17 05:08:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87ef995209b04be0d23cbfe0e77c4476dc29608c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 05:15:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a90b3e2bf24bd8b0d45e44ccc2ee54 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a18b0d0cdde44828ee2b6b2df241361 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"83f36f7b8c943a13426897501a3f6679249018d9","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-09-17 05:36:54.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"2939097be3f6ddfebc4a2fe010f1459f01ca825e","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-17 06:19:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dbfd1edaf2151082f446d7f75ae35f3e3c082e80","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-25 10:04:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf6ca474fea809d498e7a5a1c1b08c046a36d7ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 10:04:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8fe7eb71c3a5206816b392d38dc3d170ccfd1799","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 10:10:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c237e3d9e25f465c86969ffee321786e : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aabf924328542d3a11554aa6959ebb3 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"a6b3b58fcea2ceac12302241787dd2ebb40bc031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 10:10:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"973826860c82b5e3003879eb5c9b06d4efe86d93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 10:14:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/fd5de05c93334420b043c60df0e30f10 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d34239372660cc907247ddd1d92e7846ce94105a","revisions":{"4ec441cac430bc1905a6e7d866f9c261da199984":{"kind":"REWORK","_number":1,"created":"2020-09-09 12:25:43.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/39/750639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/39/750639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/39/750639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/39/750639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/39/750639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/39/750639/1"}}},"commit":{"parents":[{"commit":"1d672f6eb338fcea3ae50bad71fb43566990e6f1","subject":"Merge \"Support error handling based on ETSI NFV-SOL specification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1d672f6eb338fcea3ae50bad71fb43566990e6f1"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-09-09 11:14:38.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-09-09 11:14:38.000000000","tz":540},"subject":"Update notification spec to follow ETSI NFV SOL specs","message":"Update notification spec to follow ETSI NFV SOL specs\n\nTo follow resource type:VnfLcmOpOcc in ETSI NFV SOL 002/003,\nwe would like to update vnf_lcm_op_occ table and\nvnf_lcm_filters table.\n\nChange-Id: I175b48de60cda6629db362d0819a980af363e8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/4ec441cac430bc1905a6e7d866f9c261da199984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/4ec441cac430bc1905a6e7d866f9c261da199984"}]},"branch":"refs/heads/master"},"d34239372660cc907247ddd1d92e7846ce94105a":{"kind":"REWORK","_number":2,"created":"2020-09-17 05:07:41.000000000","uploader":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"ref":"refs/changes/39/750639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/39/750639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/39/750639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/39/750639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/39/750639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/39/750639/2"}}},"commit":{"parents":[{"commit":"25ce73641df248755bc96a291b2e631f9c350847","subject":"Merge \"VNF Workflows Customization by Action Driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/25ce73641df248755bc96a291b2e631f9c350847"}]}],"author":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-09-09 11:14:38.000000000","tz":540},"committer":{"name":"Keiko Kuriu","email":"keiko.kuriu.wa@hco.ntt.co.jp","date":"2020-09-17 05:08:58.000000000","tz":0},"subject":"Update notification spec to follow ETSI NFV SOL specs","message":"Update notification spec to follow ETSI NFV SOL specs\n\nTo follow resource type:VnfLcmOpOcc in ETSI NFV SOL 002/003,\nwe would like to update vnf_lcm_op_occ table and\nvnf_lcm_filters table.\n\nChange-Id: I175b48de60cda6629db362d0819a980af363e8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/d34239372660cc907247ddd1d92e7846ce94105a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/d34239372660cc907247ddd1d92e7846ce94105a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
