)]}'
{"id":"openstack%2Ftacker-specs~814102","triplet_id":"openstack%2Ftacker-specs~master~I540212b357241db8d2f86c8d0a7305a56464fffb","project":"openstack/tacker-specs","branch":"master","topic":"bp/container-update","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-01-06 04:55:10.000000000","reason":"Ayumu Ueha replied on the change"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2021-12-23 08:59:47.000000000","reason":"Yasufumi Ogawa replied on the change"},"32102":{"account":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"last_update":"2021-12-23 09:42:25.000000000","reason":"Manpreet Kaur replied on the change"},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-01-17 01:38:19.000000000","reason":"Change was submitted"},"33292":{"account":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"last_update":"2022-01-17 01:38:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I540212b357241db8d2f86c8d0a7305a56464fffb","subject":"Support VNF update operations using MgmtDriver","status":"MERGED","created":"2021-10-15 03:11:24.000000000","updated":"2022-01-17 01:40:41.000000000","submitted":"2022-01-17 01:38:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":44,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814102-bp/container-update","meta_rev_id":"36612316169f9c2fc6a7ba8a1f7372cc163e9d52","_number":814102,"virtual_id_number":814102,"owner":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-17 01:38:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2022-01-06 04:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2022-01-17 01:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2022-01-06 06:03:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":1,"date":"2022-01-17 01:29:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-15 03:23:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-15 08:09:40.000000000","updated_by":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2021-10-15 08:09:40.000000000","updated_by":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2021-10-15 08:09:40.000000000","updated_by":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2021-10-15 08:09:40.000000000","updated_by":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"}],"messages":[{"id":"8eb20eb15087bf0899ce535e316bf593c2d53cec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-10-15 03:11:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4faa5fb60d5758e993c99183ec1f22d6c5a0744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 03:23:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055cfca2f3064f79abfc1a421116ddc2 : FAILURE in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29bcfa15a2184999a8a3de628d072551 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0ed18fa8dc45634b0ede5820bad366c77786022e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-10-15 07:49:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b3096d2eb805e60fdb605b1e3166184ecc246c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 07:58:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673febd6e8ca45ba9dc57adc6b2f9cd6 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f488a464088a4be2946f68a588518826 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"40d89e97ffd488e1a4b2aa4382b85e7d9e1a6234","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-11-18 18:36:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7f95f31fd7033eec55b3a593cf288985d7edabf","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-12-15 01:33:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b1eee5c0934f4e0a4a0a6113b7a392f946ad597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 01:44:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5dfcf3053084791b8df92f0a1dea4e8 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4122aa11793943439c51be49f3323593 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"f6aa15ad557e54bd127ac68a863e741f6db392f3","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-12-15 05:58:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"574e88dc9cac171ba1184a0d4bbc3c2ed21ddbdb","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-12-15 07:58:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ac40d7b47f50bab5d618576e28c074143e34eb58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-12-22 04:10:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2d389b62ffba0ce169a1c160d611f8b6955a924d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 04:19:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794417ead95b4ea58f2e465865826680 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56d38cb785341fd9e4b076f6dd88370 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"6e7865027066426c335da8c362cf9fb8f3787d70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-12-23 07:31:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c32bb13f697c9d5cd42887c0c75ca52016178caa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 07:38:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2238a683c74d7188b80a99054bedb6 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2047b61526b241cab1f4e99d9b0eabc5 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"b77a35184381ecbe746e5697de17738e9db537b9","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-12-23 08:20:33.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"604f141cb11b5a8a5b5f5af570d6e050fb36f4bc","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-12-23 08:59:47.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0e079293b7fcc007c19a57c06dbae4e69539fd2c","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-12-23 09:42:25.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"188b755a97e48ce83ae6540fb082d43a8dec4c23","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-12-24 08:21:58.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1a8de4ca0415f85cd7209178cd6525c873542d11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-12-24 08:32:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9896a15a147dd7c4680d312f797f1ae17892a110","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-12-24 08:35:24.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4d6a5c4a54a9c3750ab28f789656c4d5a4b4111e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 08:37:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4053c210fba4afc8ba32fadd672c166 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/493f1c82a2a74266b8084fb9a3957e1f : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":5},{"id":"8a69d3e5522061479e0227c441ce83d5bb0987e5","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-12-24 08:54:30.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"43982014683c58e98b9a64b0f2c74bca98263a57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-12-24 09:05:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5f71d2d2fd2f2b385254b39b35f0d22ef573e9b2","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-12-24 09:08:18.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3a3cbe00bd173766f0eebcc86222233c47852de3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 09:11:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f6b40650dd4f4cb774de9f73786690 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf7d9319491f46028ffdfa250c75d976 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":6},{"id":"602d36bb0441fa71bea83502b8022ad789c8016a","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-12-24 09:12:24.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"61886a15a556d46cc751661b8715304720fd67a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-12-24 09:13:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0f8aaf71bc56994ba1c9d6fcfa7be55a55745f9e","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2021-12-24 09:14:49.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b4a572517fa1c4a9d4feca9056ce45941d8f68be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 09:23:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df4befcb2154cf0a068824c9824b7a8 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee5eecc673ce462fa0c2bf1db6880858 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":7},{"id":"8f35f84194d602dc2e695521d329d6c90a8221df","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-12-24 09:24:56.000000000","message":"Patch Set 7: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ea89c275d7e727980bdda37d9c6f635e48f75c01","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-12-24 09:39:52.000000000","message":"Patch Set 7: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b8c18f7101005dc70f53ae820482448e58e8b414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-01-05 05:29:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"32757a381e82922a3830c1be52d4525ea409e954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 05:33:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0dd562223a41f0ba71c9236bd5523e : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e12ebabf7741ac9062f6dc13be575f : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":8},{"id":"f2bfbf1ab056bbf14b9f497f22280b2fe5cdaa9d","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-01-05 05:36:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"68313fb9504b8b4ea457ccb98dcd5e79a20f03b3","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-01-06 04:55:10.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c130abc67b124c3e53e088a387aa3dd4e45b1cec","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-01-06 06:03:27.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"0b59d31e3931e810933da4dd17c72e17a8f45955","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-01-17 01:29:11.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"9afd40b9fb9b4eafa84637623f6e0cd8070956f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 01:29:36.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"e11bc64ad75e8dff178a6ff11aff69a486b197f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 01:38:19.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560b34513c404e449e52f7a72502ac19 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edf29dc80834f3e9ebb7e0cb46ea8e6 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":8},{"id":"c46e876b6db97b36ec4e4612d481ccfbd21ad0f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 01:38:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"36612316169f9c2fc6a7ba8a1f7372cc163e9d52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 01:40:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/d11e2c8009944b95a437cce90db348d5 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c83226f79b30880932da09ac30f2518bdcdff001","revisions":{"eb20b5a0fe187ab9f2fffab0d83dd260fd83a511":{"kind":"REWORK","_number":1,"created":"2021-10-15 03:11:24.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/02/814102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/02/814102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/1"}}},"commit":{"parents":[{"commit":"f78eb5ae411f2d1b95d56d4bad9ffc2307abbc4a","subject":"Merge \"Remove testr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f78eb5ae411f2d1b95d56d4bad9ffc2307abbc4a"}]}],"author":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 03:09:46.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 03:09:46.000000000","tz":540},"subject":"Support VNF update operations using MgmtDriver","message":"Support VNF update operations using MgmtDriver\n\nCurrently there is no preamble or postamble supported in the MgmtDriver\nfor the Modify VNF operation, so we will support configuration changes\nto VNF instances by adding preamble and postamble to MgmtDriver.\nThere are many use cases for Modify VNF operations, but this\nspecification focuses on changing the configuration of ConfigMap and\nSecret in Kubernetes and changing the image parameters in the Pod and\nDeployment manifests.\n\nImplements: blueprint container-update\nChange-Id: I540212b357241db8d2f86c8d0a7305a56464fffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/eb20b5a0fe187ab9f2fffab0d83dd260fd83a511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/eb20b5a0fe187ab9f2fffab0d83dd260fd83a511"}]},"branch":"refs/heads/master"},"6a2c494d784698a217cfbb7780633730292b171b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-15 07:49:49.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/02/814102/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/02/814102/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/2"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 03:09:46.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 07:49:26.000000000","tz":540},"subject":"Support VNF update operations using MgmtDriver","message":"Support VNF update operations using MgmtDriver\n\nCurrently there is no preamble or postamble supported in the MgmtDriver\nfor the Modify VNF operation, so we will support configuration changes\nto VNF instances by adding preamble and postamble to MgmtDriver.\nThere are many use cases for Modify VNF operations, but this\nspecification focuses on changing the configuration of ConfigMap and\nSecret in Kubernetes and changing the image parameters in the Pod and\nDeployment manifests.\n\nImplements: blueprint container-update\nChange-Id: I540212b357241db8d2f86c8d0a7305a56464fffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6a2c494d784698a217cfbb7780633730292b171b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/6a2c494d784698a217cfbb7780633730292b171b"}]},"branch":"refs/heads/master"},"b36cbe2e8193ea26c442dbb82d62249e63703edc":{"kind":"REWORK","_number":3,"created":"2021-12-22 04:10:57.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/02/814102/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/02/814102/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/3"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 03:09:46.000000000","tz":540},"committer":{"name":"Masaki Ueno","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-12-22 04:10:45.000000000","tz":540},"subject":"Support VNF update operations using MgmtDriver","message":"Support VNF update operations using MgmtDriver\n\nCurrently there is no preamble or postamble supported in the MgmtDriver\nfor the Modify VNF operation, so we will support configuration changes\nto VNF instances by adding preamble and postamble to MgmtDriver.\nThere are many use cases for Modify VNF operations, but this\nspecification focuses on changing the configuration of ConfigMap and\nSecret in Kubernetes and changing the image parameters in the Pod and\nDeployment manifests.\n\nImplements: blueprint container-update\nChange-Id: I540212b357241db8d2f86c8d0a7305a56464fffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b36cbe2e8193ea26c442dbb82d62249e63703edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b36cbe2e8193ea26c442dbb82d62249e63703edc"}]},"branch":"refs/heads/master"},"94000c63b44f21c54efaf146bd870a6012f2944f":{"kind":"REWORK","_number":4,"created":"2021-12-23 07:31:46.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/02/814102/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/02/814102/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/4"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 03:09:46.000000000","tz":540},"committer":{"name":"Masaki Ueno","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-12-23 07:30:46.000000000","tz":540},"subject":"Support VNF update operations using MgmtDriver","message":"Support VNF update operations using MgmtDriver\n\nCurrently there is no preamble or postamble supported in the MgmtDriver\nfor the Modify VNF operation, so we will support configuration changes\nto VNF instances by adding preamble and postamble to MgmtDriver.\nThere are many use cases for Modify VNF operations, but this\nspecification focuses on changing the configuration of ConfigMap and\nSecret in Kubernetes and changing the image parameters in the Pod and\nDeployment manifests.\n\nImplements: blueprint container-update\nChange-Id: I540212b357241db8d2f86c8d0a7305a56464fffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/94000c63b44f21c54efaf146bd870a6012f2944f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/94000c63b44f21c54efaf146bd870a6012f2944f"}]},"branch":"refs/heads/master"},"e31de518a0aadf4dd2319d4d63b39a6b6dfe6ad7":{"kind":"REWORK","_number":5,"created":"2021-12-24 08:32:35.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/02/814102/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/02/814102/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/5"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 03:09:46.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-12-24 08:32:26.000000000","tz":540},"subject":"Support VNF update operations using MgmtDriver","message":"Support VNF update operations using MgmtDriver\n\nCurrently there is no preamble or postamble supported in the MgmtDriver\nfor the Modify VNF operation, so we will support configuration changes\nto VNF instances by adding preamble and postamble to MgmtDriver.\nThere are many use cases for Modify VNF operations, but this\nspecification focuses on changing the configuration of ConfigMap and\nSecret in Kubernetes and changing the image parameters in the Pod and\nDeployment manifests.\n\nImplements: blueprint container-update\nChange-Id: I540212b357241db8d2f86c8d0a7305a56464fffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e31de518a0aadf4dd2319d4d63b39a6b6dfe6ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e31de518a0aadf4dd2319d4d63b39a6b6dfe6ad7"}]},"branch":"refs/heads/master"},"aea7a430f5ca2cef4c8016c87b7bed76e7b1a627":{"kind":"REWORK","_number":6,"created":"2021-12-24 09:05:28.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/02/814102/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/02/814102/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/6"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 03:09:46.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-12-24 09:05:20.000000000","tz":540},"subject":"Support VNF update operations using MgmtDriver","message":"Support VNF update operations using MgmtDriver\n\nCurrently there is no preamble or postamble supported in the MgmtDriver\nfor the Modify VNF operation, so we will support configuration changes\nto VNF instances by adding preamble and postamble to MgmtDriver.\nThere are many use cases for Modify VNF operations, but this\nspecification focuses on changing the configuration of ConfigMap and\nSecret in Kubernetes and changing the image parameters in the Pod and\nDeployment manifests.\n\nImplements: blueprint container-update\nChange-Id: I540212b357241db8d2f86c8d0a7305a56464fffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/aea7a430f5ca2cef4c8016c87b7bed76e7b1a627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/aea7a430f5ca2cef4c8016c87b7bed76e7b1a627"}]},"branch":"refs/heads/master"},"46fac614065118e9b71c065c2ae618d258b78b06":{"kind":"REWORK","_number":7,"created":"2021-12-24 09:13:24.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/02/814102/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/02/814102/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/7"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 03:09:46.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-12-24 09:13:16.000000000","tz":540},"subject":"Support VNF update operations using MgmtDriver","message":"Support VNF update operations using MgmtDriver\n\nCurrently there is no preamble or postamble supported in the MgmtDriver\nfor the Modify VNF operation, so we will support configuration changes\nto VNF instances by adding preamble and postamble to MgmtDriver.\nThere are many use cases for Modify VNF operations, but this\nspecification focuses on changing the configuration of ConfigMap and\nSecret in Kubernetes and changing the image parameters in the Pod and\nDeployment manifests.\n\nImplements: blueprint container-update\nChange-Id: I540212b357241db8d2f86c8d0a7305a56464fffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/46fac614065118e9b71c065c2ae618d258b78b06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/46fac614065118e9b71c065c2ae618d258b78b06"}]},"branch":"refs/heads/master"},"c83226f79b30880932da09ac30f2518bdcdff001":{"kind":"REWORK","_number":8,"created":"2022-01-05 05:29:04.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/02/814102/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/02/814102/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/02/814102/8"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2021-10-15 03:09:46.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-01-05 05:29:00.000000000","tz":540},"subject":"Support VNF update operations using MgmtDriver","message":"Support VNF update operations using MgmtDriver\n\nCurrently there is no preamble or postamble supported in the MgmtDriver\nfor the Modify VNF operation, so we will support configuration changes\nto VNF instances by adding preamble and postamble to MgmtDriver.\nThere are many use cases for Modify VNF operations, but this\nspecification focuses on changing the configuration of ConfigMap and\nSecret in Kubernetes and changing the image parameters in the Pod and\nDeployment manifests.\n\nImplements: blueprint container-update\nChange-Id: I540212b357241db8d2f86c8d0a7305a56464fffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c83226f79b30880932da09ac30f2518bdcdff001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c83226f79b30880932da09ac30f2518bdcdff001"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
