)]}'
{"id":"openstack%2Ftacker-specs~814105","triplet_id":"openstack%2Ftacker-specs~master~Ib8910e549076d0fe782909dfe702976948a5461f","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-nfv-solv3-scale-vnf","attention_set":{},"removed_from_attention_set":{"31072":{"account":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"last_update":"2021-12-24 06:30:17.000000000","reason":"Change was submitted"},"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2021-12-15 01:04:59.000000000","reason":"Ayumu Ueha replied on the change"},"31668":{"account":{"_account_id":31668,"deleted":true},"last_update":"2021-10-20 01:18:19.000000000","reason":"Ai Hamano replied on the change"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2021-11-17 01:34:35.000000000","reason":"Yasufumi Ogawa replied on the change"},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2021-12-24 06:30:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8910e549076d0fe782909dfe702976948a5461f","subject":"Enhance NFV SOL_v3 LCM operation","status":"MERGED","created":"2021-10-15 06:29:32.000000000","updated":"2021-12-24 06:32:05.000000000","submitted":"2021-12-24 06:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"814105-bp/support-nfv-solv3-scale-vnf","meta_rev_id":"6fe116abec5a4962f6b47e8d23bce23a404f06be","_number":814105,"virtual_id_number":814105,"owner":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-24 06:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2021-12-15 01:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2021-12-24 06:25:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":1,"date":"2021-12-24 06:25:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"CC":[{"_account_id":31668,"deleted":true}],"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-15 06:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-15 07:57:09.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2021-10-15 07:57:09.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2021-10-15 07:57:09.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2021-10-19 23:18:44.000000000","updated_by":{"_account_id":31668,"deleted":true},"reviewer":{"_account_id":31668,"deleted":true},"state":"CC"}],"messages":[{"id":"f8ffa49ed25a290151bf2411d1295bc2b7b54f29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2021-10-15 06:29:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98e7ab246e66ac4d02adacb08183f68063757d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 06:55:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920d0ece5a4544a08afce3480f1401bf : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a013265671420b9bad54cb1473713d : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"347ddb8a40eece5ab64969a210055b95580aab5d","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-10-15 08:04:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d94894a468ca8094a7a02a6e443471d0a9d0bd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2021-10-18 01:06:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eb94cf6d2f38b19711ce87ce6aa0a73baf9693ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 01:12:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f905a368f1042f79eb7dc2964c36526 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/242ea6d063f54dc39d849eeb16d8adfa : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"33d9ba4f99f5a282837c93f470aeac7ef17a7c23","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2021-10-18 01:13:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f2b803ca667e0600b4c9c6e175db2f601a92e00","author":{"_account_id":31668,"deleted":true},"date":"2021-10-19 23:18:44.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f1a5f2643629b860adc7b9880d41a2ee897f2e1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2021-10-20 00:32:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2641db972a47d843dd53d8df5ef1b892832bd90b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 00:40:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a256267f7ed4a65ad30168c63b2e0a6 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c29843e500949f29b5a5e61a14671ae : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"36eae962e6424c4c05e3306c2298147036c5bde8","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2021-10-20 00:53:14.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"97fe00cd8450c4c68a9de5599abdf2b891028e09","author":{"_account_id":31668,"deleted":true},"date":"2021-10-20 01:18:19.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"eaa1d644de5120b52fde1c9d7607fbe86b563d97","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-11-17 01:34:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d51ae8d8c6ffb8cff28b55d61542c755f61b4e51","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-12-14 06:46:35.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"646cc035fe718c7215f54b26c7c76b64f0063226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2021-12-15 00:51:30.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"da7a45ba37b727957726a95738f829f1216ab220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2021-12-15 00:52:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"129ec7a3b54e0210ce9732ae69012e3c74b30396","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2021-12-15 00:53:57.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"58e3763633d9a72b30d5b4af828437a4c3aeaeac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 00:59:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a130a90f756843fcb005551de280d70f : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58accaa435ba4d07b079a2a145f267a5 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":5},{"id":"704694870efb4a77f9ff3112cc11e91d11fcdb94","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-12-15 01:04:59.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1849fa5ca71f377da75564ea5c9936f4fc9fea2a","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-12-24 06:25:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3c8a8d76ca2613d8495fdd77b5dd714a72d6b9aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 06:26:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c55156d1518e8bf8a220a5eeb773c43f9c4bde58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 06:30:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa10fc6cc85e42be943962967f5ec12d : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c87d736f5040feb848ad7045c3b70c : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":5},{"id":"586a1d10c7e4bba5c13c7ef76684d8b7eca82a59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 06:30:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6fe116abec5a4962f6b47e8d23bce23a404f06be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 06:32:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/480e6350b5c74137a52e8b754b819b79 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"66de31f8359e708393de425af4f4da504bf2ad49","revisions":{"5ff4d839c3885771eb15a6f8aff711e80e6c52a6":{"kind":"REWORK","_number":1,"created":"2021-10-15 06:29:32.000000000","uploader":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"ref":"refs/changes/05/814105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/05/814105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/1"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-10-15 05:30:06.000000000","tz":0},"committer":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-10-15 06:25:40.000000000","tz":0},"subject":"Enhance NFV SOL_v3 LCM operation","message":"Enhance NFV SOL_v3 LCM operation\n\nThis specification supports a new version of VNF lifecycle management APIs\ncomplying with ETSI NFV SOL v3.\nIt adds a new version of APIs involved in operating instantiated VNF instances\n(scale, heal, modify, and change external vnf connectivity).\n\nCo-authored-By: Hirofumi Noguchi \u003chirofumi.noguchi.rs@hco.ntt.co.jp\u003e\nImplements: blueprint support-nfv-solv3-scale-vnf\nImplements: blueprint support-nfv-solv3-heal-vnf\nImplements: blueprint support-nfv-solv3-modify-vnf\nImplements: blueprint support-nfv-solv3-change-external-connectivity\nChange-Id: Ib8910e549076d0fe782909dfe702976948a5461f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/5ff4d839c3885771eb15a6f8aff711e80e6c52a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/5ff4d839c3885771eb15a6f8aff711e80e6c52a6"}]},"branch":"refs/heads/master"},"5c300feb79df1196d0eb6999def00c465b950851":{"kind":"REWORK","_number":2,"created":"2021-10-18 01:06:48.000000000","uploader":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"ref":"refs/changes/05/814105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/05/814105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/2"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-10-15 05:30:06.000000000","tz":0},"committer":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-10-18 00:59:49.000000000","tz":0},"subject":"Enhance NFV SOL_v3 LCM operation","message":"Enhance NFV SOL_v3 LCM operation\n\nThis specification supports a new version of VNF lifecycle management APIs\ncomplying with ETSI NFV SOL v3.\nIt adds a new version of APIs involved in operating instantiated VNF instances\n(scale, heal, modify, and change external vnf connectivity).\n\nCo-authored-By: Hirofumi Noguchi \u003chirofumi.noguchi.rs@hco.ntt.co.jp\u003e\nImplements: blueprint support-nfv-solv3-scale-vnf\nImplements: blueprint support-nfv-solv3-heal-vnf\nImplements: blueprint support-nfv-solv3-modify-vnf\nImplements: blueprint support-nfv-solv3-change-external-connectivity\nChange-Id: Ib8910e549076d0fe782909dfe702976948a5461f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/5c300feb79df1196d0eb6999def00c465b950851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/5c300feb79df1196d0eb6999def00c465b950851"}]},"branch":"refs/heads/master"},"d163af3ddf5cd5ddfcefbdbef641fc1e9016215d":{"kind":"REWORK","_number":3,"created":"2021-10-20 00:32:52.000000000","uploader":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"ref":"refs/changes/05/814105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/05/814105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/3"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-10-15 05:30:06.000000000","tz":0},"committer":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-10-20 00:32:14.000000000","tz":0},"subject":"Enhance NFV SOL_v3 LCM operation","message":"Enhance NFV SOL_v3 LCM operation\n\nThis specification supports a new version of VNF lifecycle management APIs\ncomplying with ETSI NFV SOL v3.\nIt adds a new version of APIs involved in operating instantiated VNF instances\n(scale, heal, modify, and change external vnf connectivity).\n\nCo-authored-By: Hirofumi Noguchi \u003chirofumi.noguchi.rs@hco.ntt.co.jp\u003e\nImplements: blueprint support-nfv-solv3-scale-vnf\nImplements: blueprint support-nfv-solv3-heal-vnf\nImplements: blueprint support-nfv-solv3-modify-vnf\nImplements: blueprint support-nfv-solv3-change-external-connectivity\nChange-Id: Ib8910e549076d0fe782909dfe702976948a5461f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/d163af3ddf5cd5ddfcefbdbef641fc1e9016215d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/d163af3ddf5cd5ddfcefbdbef641fc1e9016215d"}]},"branch":"refs/heads/master"},"8c656407f6674c413875e8568c8c4700cd812cba":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-12-15 00:51:30.000000000","uploader":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"ref":"refs/changes/05/814105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/05/814105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/4"}}},"commit":{"parents":[{"commit":"0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045","subject":"Delete placeholder in xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045"}]}],"author":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-10-15 05:30:06.000000000","tz":0},"committer":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-12-14 10:11:49.000000000","tz":0},"subject":"Enhance NFV SOL_v3 LCM operation","message":"Enhance NFV SOL_v3 LCM operation\n\nThis specification supports a new version of VNF lifecycle management APIs\ncomplying with ETSI NFV SOL v3.\nIt adds a new version of APIs involved in operating instantiated VNF instances\n(scale, heal, modify, and change external vnf connectivity).\n\nCo-authored-By: Hirofumi Noguchi \u003chirofumi.noguchi.rs@hco.ntt.co.jp\u003e\nImplements: blueprint support-nfv-solv3-scale-vnf\nImplements: blueprint support-nfv-solv3-heal-vnf\nImplements: blueprint support-nfv-solv3-modify-vnf\nImplements: blueprint support-nfv-solv3-change-external-connectivity\nChange-Id: Ib8910e549076d0fe782909dfe702976948a5461f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8c656407f6674c413875e8568c8c4700cd812cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8c656407f6674c413875e8568c8c4700cd812cba"}]},"branch":"refs/heads/master"},"66de31f8359e708393de425af4f4da504bf2ad49":{"kind":"REWORK","_number":5,"created":"2021-12-15 00:52:54.000000000","uploader":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"ref":"refs/changes/05/814105/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/05/814105/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/05/814105/5"}}},"commit":{"parents":[{"commit":"0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045","subject":"Delete placeholder in xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045"}]}],"author":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-10-15 05:30:06.000000000","tz":0},"committer":{"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","date":"2021-12-14 10:13:09.000000000","tz":0},"subject":"Enhance NFV SOL_v3 LCM operation","message":"Enhance NFV SOL_v3 LCM operation\n\nThis specification supports a new version of VNF lifecycle management APIs\ncomplying with ETSI NFV SOL v3.\nIt adds a new version of APIs involved in operating instantiated VNF instances\n(scale, heal, modify, and change external vnf connectivity).\n\nCo-authored-By: Hirofumi Noguchi \u003chirofumi.noguchi.rs@hco.ntt.co.jp\u003e\nImplements: blueprint support-nfv-solv3-scale-vnf\nImplements: blueprint support-nfv-solv3-heal-vnf\nImplements: blueprint support-nfv-solv3-modify-vnf\nImplements: blueprint support-nfv-solv3-change-external-connectivity\nChange-Id: Ib8910e549076d0fe782909dfe702976948a5461f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/66de31f8359e708393de425af4f4da504bf2ad49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/66de31f8359e708393de425af4f4da504bf2ad49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
