)]}'
{"id":"openstack%2Ftacker-specs~814379","triplet_id":"openstack%2Ftacker-specs~master~Ie614fc63861b7dabe527369add74bf5d65b7c994","project":"openstack/tacker-specs","branch":"master","topic":"bp/paging-query-result","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-01-11 14:33:54.000000000","reason":"Ayumu Ueha replied on the change"},"31813":{"account":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"last_update":"2022-01-17 01:38:17.000000000","reason":"Change was submitted"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-01-17 01:28:23.000000000","reason":"Yasufumi Ogawa replied on the change"},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-01-17 01:38:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie614fc63861b7dabe527369add74bf5d65b7c994","subject":"Support handling large query results by ETSI NFV","status":"MERGED","created":"2021-10-18 09:01:16.000000000","updated":"2022-01-17 01:39:37.000000000","submitted":"2022-01-17 01:38:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"814379-bp/paging-query-result","meta_rev_id":"4c493a4fcd38de9c0436d8cb7c3c5df969f79eb1","_number":814379,"virtual_id_number":814379,"owner":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":34107,"name":"Masaki Oyama","email":"ma-ooyama@kddi.com","username":"oyamamasaki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-17 01:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2022-01-13 04:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2022-01-17 01:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":34107,"name":"Masaki Oyama","email":"ma-ooyama@kddi.com","username":"oyamamasaki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":1,"date":"2022-01-17 01:28:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":34107,"name":"Masaki Oyama","email":"ma-ooyama@kddi.com","username":"oyamamasaki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":34107,"name":"Masaki Oyama","email":"ma-ooyama@kddi.com","username":"oyamamasaki"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-18 09:17:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-15 12:33:34.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2021-12-17 02:14:05.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2021-12-17 02:14:05.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2021-12-28 10:14:07.000000000","updated_by":{"_account_id":34107,"name":"Masaki Oyama","email":"ma-ooyama@kddi.com","username":"oyamamasaki"},"reviewer":{"_account_id":34107,"name":"Masaki Oyama","email":"ma-ooyama@kddi.com","username":"oyamamasaki"},"state":"REVIEWER"}],"messages":[{"id":"55f837a7b3c3f96fc65addb8627b78e72d44edfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-10-18 09:01:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"488f9600c2c8280968d076bf9c94035d9d7ef8b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:17:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9002d9bcefb44c682ef478e645518c8 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7923acfe9f941869f3205ea985e26e5 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ceaf77dc70f808cc5d5ec01dcdccddc4d9de1aca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-10-19 05:19:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8319abe332dd301d45a3f593b29d1fde961e0cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 05:28:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59fb08937c4f495c843c6bdf2be2e434 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a021e6a8dd5449f99cab33c2b5a1d73 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6cdf5efadc15509f61ca1a48738c33ea6e585f10","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-11-22 02:11:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26727024b15192e809d5507aeba205a5c85258fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 02:35:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5714feb7c4d148058a222f76afd837d1 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab51e9b97e654271b57dcad48762ba11 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"3037e70404cbed65799376a799094d553d56ad30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-11-30 06:13:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ada6b5e3eb815b121cbb0eb6f3ae7721a3b1e82a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 06:18:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ece19ee99d4c4692d16dbb083876f3 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b30f238a67f649aeabf095c9a266f3cf : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":3},{"id":"7feca22cc911a3379b85cb82e6d49ee853c3480d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-11-30 10:20:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7e0186ae5911d97bfa874b23cddcc5bbaf41d376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 10:30:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a5db077f8c41be85df3a0cb9126a1a : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df71d148bf754c23a0ad60db6da5ad40 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":4},{"id":"bc0b547fa04460122cca008ffe6c101581bd141b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-03 10:21:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ffa338aed9d445cdaa7dc28624ea1c934c92d443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 10:36:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d9e1e540724bd3885b708d46541a59 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4771a2653547c7b7ff924cafdc4058 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":5},{"id":"05657845bfc35ad6a41e8a445d8f176f50b8ccb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-06 09:12:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e579be91cc4185c66bdaae6ac24fb458b331b834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 09:21:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c30b3fdabe451b8a4f81219198b739 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5eab9e7ea0a4e0d93a2104e93fc2ead : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":6},{"id":"511d71989a854b115b394f0576bdf03c179b1918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-07 05:16:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"739e37bdd58c3c85a324a2f4ff3e0b18378bdac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 05:23:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83bf0dca2b544e980f10b0c8da899c1 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b0a27638364ad1a47789fb765f1b4a : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":7},{"id":"06046e13e794358e49280cc6a142305aec9701d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-10 01:37:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6e6e8548af132133628104807aeaf8fcacc9793d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 01:47:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99882603671b4984ba7c4f63c9bf1596 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a397302ed004212a40a481bd7558b38 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":8},{"id":"544528bf7a79a18b89aad9a7d90ee93b34f8d7c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-14 00:39:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"26ea522a2f2d575efe9ff6fa5ea45f0b84b94ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 00:47:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37538d275c5d4c3d8d33d37c03e6ed6c : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e30624c92d147cbb4e08a7f0041d89e : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":9},{"id":"76e31ab90d0024cc96749da9b456271814e9fceb","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-12-15 12:33:34.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"862113166f0fed527633093756ad7900c5535f7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-22 01:11:47.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"418e6eaf76d33764151c4f6064f31850a91554b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 01:20:51.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e1e57b2a6664f7cb2088ce95b671d82 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ebf6d6c942e4e469dd3b2ec9a83f5c1 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":10},{"id":"9d722a3fd50ac2b5b72089e002ea6131adcbb215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-22 06:01:23.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"76d7c47b6130dd4f65b176aed8171830f0e75834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 06:09:17.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c1da3a6b8b4b77896297217a91ada0 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/214f06e27ff44196b5b5a5a3a1e9ac8b : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":11},{"id":"8b0f14fa33f7ea8e8366112cf3828632c43a6889","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-22 08:19:44.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"af8ca242f5b82f6f60b0bf0462085c68430ee8b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-22 08:27:28.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2806ace2429eee7774d5074b184fd505fc352f8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 08:36:47.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f728d0ac2b4403899f5c6f377e310ec : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1828551b833548c89bbe4fa8f3556e4a : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":12},{"id":"9682d985d489899d91852f3cc16460967dad13bb","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-22 09:03:19.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"12833bbe003f2b2e2543d76dc2682c3d2913dc02","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-12-24 02:58:58.000000000","message":"Patch Set 12: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"f15801c85d3e73bef3c52c0bb20f3afbdb3470c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-28 00:22:45.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"77469231aa541eacace058265de7653eab36ab39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 00:29:14.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f37493710cc46bf91e800bba5717fb8 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/551818184a164e92a3478800f73b3986 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":13},{"id":"4092c05c79ee1e4968a90ddc9bb45dfb4e5656eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-28 08:06:32.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"422822f4a088e5ccef05f441823a83e8b25e90c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 08:14:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11fa9b9eedde45a3bda8a88116dcdaee : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5712240f97934d5e8c150e5001fea0ab : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":14},{"id":"76bd51a848a11f2ffd4ab75b189b6705a9a2b0a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-28 08:18:39.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"b00adf942380a386079102ece9d5ed3bb3da5f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 08:28:08.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06351b46965045139152a3a5ff7ec968 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d9533f504b0476b95813b6dd3b944a3 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":15},{"id":"22172a07e9cc52b79bcdef086853963db6cbae03","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2021-12-28 08:40:56.000000000","message":"Patch Set 15:\n\n(5 comments)","accounts_in_message":[],"_revision_number":15},{"id":"ad0ed6278af77abc951a641ecca820f327e2a769","author":{"_account_id":34107,"name":"Masaki Oyama","email":"ma-ooyama@kddi.com","username":"oyamamasaki"},"date":"2021-12-28 10:14:07.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c5e044f5a277813e42d9d73305ec3f42366f3dab","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-01-06 17:53:06.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"c0bd7f4ce91d69724433ff10836cbfd90315e8fc","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-01-11 14:33:54.000000000","message":"Patch Set 15: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":15},{"id":"ac197426eb6e8b910f2a00f6afb022e9555cfe09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2022-01-13 02:31:27.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"28469c1a5ddbfa162e389267e45f0d2cf3f2f1dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 02:47:43.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98526fb5370047faa0d9a44bdfebd848 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2411c6623f4a6b88d21b7129d15f2e : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":16},{"id":"af6819bb19e173a36d34afff44573df32dbc2ca5","author":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"date":"2022-01-13 03:04:14.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"db7526fe42585b6f0254eaf17ac6d9871aa9ee6a","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-01-13 04:14:22.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"735a1e04a14adfd6fce709c7938cdb7ec69e9530","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-01-17 01:28:23.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"69e43ee432513a441353da772b86c42954c620d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 01:28:44.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"e8ab1721f1aa6603fd2ced88e462ef81b51ebc44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 01:38:17.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9cd55f9ad9e4cb9aa65ed924211486e : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd0b9244c0364bb9ba40a96809bb80d5 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":16},{"id":"09d2b1b1cd9f8903eee4b1fae1063e1a50f38543","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 01:38:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"4c493a4fcd38de9c0436d8cb7c3c5df969f79eb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 01:39:37.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/2facaf8e6a1b4d09bd26df113072aa4f : SUCCESS in 56s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"30b153e0924dda5f744fb055aa55e0f0e5fd5a00","revisions":{"5cf03c2cbb79e8808ce63341fab43d29641eb158":{"kind":"REWORK","_number":1,"created":"2021-10-18 09:01:16.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/1"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"subject":"[WIP] Support handling large query results by ETSI NFV","message":"[WIP] Support handling large query results by ETSI NFV\n\nThis patch supports handling large query results with paging\naccording to ETSI NFV SOL013.\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/5cf03c2cbb79e8808ce63341fab43d29641eb158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/5cf03c2cbb79e8808ce63341fab43d29641eb158"}]},"branch":"refs/heads/master"},"26d1e70996694c1c68bdc29bd95f6fc3122eff29":{"kind":"REWORK","_number":2,"created":"2021-10-19 05:19:00.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/2"}}},"commit":{"parents":[{"commit":"1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0","subject":"Add Yoga specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1ef1646d2ad4f0fc2dbd48439018a4f8ffa34ff0"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-19 05:19:01.000000000","tz":540},"subject":"[WIP] Support handling large query results by ETSI NFV","message":"[WIP] Support handling large query results by ETSI NFV\n\nThis patch supports handling large query results with paging\naccording to ETSI NFV SOL013.\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/26d1e70996694c1c68bdc29bd95f6fc3122eff29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/26d1e70996694c1c68bdc29bd95f6fc3122eff29"}]},"branch":"refs/heads/master"},"69264c4b38e737e49672885060da985ad82af4bd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-30 06:13:36.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/3"}}},"commit":{"parents":[{"commit":"0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045","subject":"Delete placeholder in xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-11-30 06:13:36.000000000","tz":0},"subject":"[WIP] Support handling large query results by ETSI NFV","message":"[WIP] Support handling large query results by ETSI NFV\n\nThis patch supports handling large query results with paging\naccording to ETSI NFV SOL013.\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/69264c4b38e737e49672885060da985ad82af4bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/69264c4b38e737e49672885060da985ad82af4bd"}]},"branch":"refs/heads/master","description":"Rebase"},"c7b5ab6e84f252c984854b394949efc79a38e010":{"kind":"REWORK","_number":4,"created":"2021-11-30 10:20:59.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/4"}}},"commit":{"parents":[{"commit":"0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045","subject":"Delete placeholder in xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-11-30 10:20:29.000000000","tz":540},"subject":"[WIP] Support handling large query results by ETSI NFV","message":"[WIP] Support handling large query results by ETSI NFV\n\nThis patch supports handling large query results with paging\naccording to ETSI NFV SOL013.\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c7b5ab6e84f252c984854b394949efc79a38e010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c7b5ab6e84f252c984854b394949efc79a38e010"}]},"branch":"refs/heads/master"},"116644b261931d34fa758925012c280e31190547":{"kind":"REWORK","_number":5,"created":"2021-12-03 10:21:36.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/5"}}},"commit":{"parents":[{"commit":"0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045","subject":"Delete placeholder in xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-03 10:21:14.000000000","tz":540},"subject":"[WIP] Support handling large query results by ETSI NFV","message":"[WIP] Support handling large query results by ETSI NFV\n\nThis patch supports handling large query results with paging\naccording to ETSI NFV SOL013.\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/116644b261931d34fa758925012c280e31190547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/116644b261931d34fa758925012c280e31190547"}]},"branch":"refs/heads/master"},"9e89ee66baaba52e74819fd480b683bb239cc67a":{"kind":"REWORK","_number":6,"created":"2021-12-06 09:12:17.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/6"}}},"commit":{"parents":[{"commit":"0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045","subject":"Delete placeholder in xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-06 08:59:44.000000000","tz":540},"subject":"[WIP] Support handling large query results by ETSI NFV","message":"[WIP] Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Rejecting request in case of too large query results according\n   to ETSI NFV SOL013\n - Fetching entire query results forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/9e89ee66baaba52e74819fd480b683bb239cc67a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/9e89ee66baaba52e74819fd480b683bb239cc67a"}]},"branch":"refs/heads/master"},"319d20082ab74ebd47320ce0542cb9810b97b06b":{"kind":"REWORK","_number":7,"created":"2021-12-07 05:16:14.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/7"}}},"commit":{"parents":[{"commit":"0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045","subject":"Delete placeholder in xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-07 05:14:38.000000000","tz":540},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Rejecting request in case of too large query results according\n   to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/319d20082ab74ebd47320ce0542cb9810b97b06b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/319d20082ab74ebd47320ce0542cb9810b97b06b"}]},"branch":"refs/heads/master"},"285c399cd6ca97e84431e82cf7af8cc2d746c70e":{"kind":"REWORK","_number":8,"created":"2021-12-10 01:37:54.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/8"}}},"commit":{"parents":[{"commit":"0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045","subject":"Delete placeholder in xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-10 01:37:58.000000000","tz":540},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Rejecting request in case of too large query results according\n   to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/285c399cd6ca97e84431e82cf7af8cc2d746c70e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/285c399cd6ca97e84431e82cf7af8cc2d746c70e"}]},"branch":"refs/heads/master"},"0f81b59c1cc6a891a7b7809acf0da89f3568558a":{"kind":"REWORK","_number":9,"created":"2021-12-14 00:39:40.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/9"}}},"commit":{"parents":[{"commit":"0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045","subject":"Delete placeholder in xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0aedec4a6b7fdc4cd2eabed7f3dd577fd83f0045"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-14 00:40:00.000000000","tz":540},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Rejecting request in case of too large query results according\n   to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0f81b59c1cc6a891a7b7809acf0da89f3568558a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0f81b59c1cc6a891a7b7809acf0da89f3568558a"}]},"branch":"refs/heads/master"},"cfbf5f0057be5c6e791a8ec6a2c68deb5054cb8a":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-12-22 01:11:47.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/10"}}},"commit":{"parents":[{"commit":"2feecc925f1fa8924ee807355cab752a74b7d1a2","subject":"Merge \"Heal and scale method support of lcm_user_data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/2feecc925f1fa8924ee807355cab752a74b7d1a2"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-22 01:11:47.000000000","tz":0},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Rejecting request in case of too large query results according\n   to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/cfbf5f0057be5c6e791a8ec6a2c68deb5054cb8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/cfbf5f0057be5c6e791a8ec6a2c68deb5054cb8a"}]},"branch":"refs/heads/master","description":"Rebase"},"e465e12a2e8c89e12e3dce92051be132c227fd66":{"kind":"REWORK","_number":11,"created":"2021-12-22 06:01:23.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/11"}}},"commit":{"parents":[{"commit":"2feecc925f1fa8924ee807355cab752a74b7d1a2","subject":"Merge \"Heal and scale method support of lcm_user_data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/2feecc925f1fa8924ee807355cab752a74b7d1a2"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-22 06:01:24.000000000","tz":540},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e465e12a2e8c89e12e3dce92051be132c227fd66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e465e12a2e8c89e12e3dce92051be132c227fd66"}]},"branch":"refs/heads/master"},"f77cf36761b1c384f69dc6d960f9d1abb5481f44":{"kind":"REWORK","_number":12,"created":"2021-12-22 08:27:28.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/12"}}},"commit":{"parents":[{"commit":"2feecc925f1fa8924ee807355cab752a74b7d1a2","subject":"Merge \"Heal and scale method support of lcm_user_data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/2feecc925f1fa8924ee807355cab752a74b7d1a2"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-22 08:28:23.000000000","tz":540},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f77cf36761b1c384f69dc6d960f9d1abb5481f44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f77cf36761b1c384f69dc6d960f9d1abb5481f44"}]},"branch":"refs/heads/master"},"0b91678b66a5924a082fef99829106a458642c57":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-12-28 00:22:45.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/13"}}},"commit":{"parents":[{"commit":"0536044e92a256459d861486b0a8383420cafbc1","subject":"Merge \"Use Robot Framework and TST test code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0536044e92a256459d861486b0a8383420cafbc1"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-28 00:22:45.000000000","tz":0},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0b91678b66a5924a082fef99829106a458642c57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0b91678b66a5924a082fef99829106a458642c57"}]},"branch":"refs/heads/master","description":"Rebase"},"a15984d5f727b6f6d7914fa55e5be7a9bddafd71":{"kind":"REWORK","_number":14,"created":"2021-12-28 08:06:32.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/14"}}},"commit":{"parents":[{"commit":"0536044e92a256459d861486b0a8383420cafbc1","subject":"Merge \"Use Robot Framework and TST test code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0536044e92a256459d861486b0a8383420cafbc1"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-28 08:07:37.000000000","tz":540},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/a15984d5f727b6f6d7914fa55e5be7a9bddafd71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/a15984d5f727b6f6d7914fa55e5be7a9bddafd71"}]},"branch":"refs/heads/master"},"3151b081c40baa869aa60c8eb5ceb0496f8b3940":{"kind":"REWORK","_number":15,"created":"2021-12-28 08:18:39.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/15"}}},"commit":{"parents":[{"commit":"0536044e92a256459d861486b0a8383420cafbc1","subject":"Merge \"Use Robot Framework and TST test code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0536044e92a256459d861486b0a8383420cafbc1"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-12-28 08:19:44.000000000","tz":540},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/3151b081c40baa869aa60c8eb5ceb0496f8b3940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/3151b081c40baa869aa60c8eb5ceb0496f8b3940"}]},"branch":"refs/heads/master"},"30b153e0924dda5f744fb055aa55e0f0e5fd5a00":{"kind":"REWORK","_number":16,"created":"2022-01-13 02:31:27.000000000","uploader":{"_account_id":31813,"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","username":"k-edagawa"},"ref":"refs/changes/79/814379/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/79/814379/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/79/814379/16"}}},"commit":{"parents":[{"commit":"0536044e92a256459d861486b0a8383420cafbc1","subject":"Merge \"Use Robot Framework and TST test code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/0536044e92a256459d861486b0a8383420cafbc1"}]}],"author":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2021-10-18 08:55:57.000000000","tz":540},"committer":{"name":"Koichi Edagawa","email":"edagawa.kc@nec.com","date":"2022-01-13 02:31:29.000000000","tz":540},"subject":"Support handling large query results by ETSI NFV","message":"Support handling large query results by ETSI NFV\n\nThis specification describes about supporting handling large query\nresults in a response of target APIs. The features to be added are\nlike the following.\n\n - Paging query results according to ETSI NFV SOL013\n - Fetching entire records forcibly\n\nImplements: blueprint paging-query-result\nChange-Id: Ie614fc63861b7dabe527369add74bf5d65b7c994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/30b153e0924dda5f744fb055aa55e0f0e5fd5a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/30b153e0924dda5f744fb055aa55e0f0e5fd5a00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
