)]}'
{"id":"openstack%2Ftacker-specs~834156","triplet_id":"openstack%2Ftacker-specs~master~Idf398ea9bf083afdecb99a0cea3c539a34ec17d3","project":"openstack/tacker-specs","branch":"master","topic":"bp/enhance-multi-tenant-policy","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-08-10 01:12:32.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"32102":{"account":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"last_update":"2022-08-10 04:06:16.000000000","reason":"Change was submitted"},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-08-10 04:06:16.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-08-10 04:06:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idf398ea9bf083afdecb99a0cea3c539a34ec17d3","subject":"Enhance multi tenant policy in VNF LCM","status":"MERGED","created":"2022-03-17 11:40:29.000000000","updated":"2022-08-10 04:07:31.000000000","submitted":"2022-08-10 04:06:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"834156-bp/enhance-multi-tenant-policy","meta_rev_id":"60a8c6f6cc7d32c8bb67e474d09ddd5383cb28ab","_number":834156,"virtual_id_number":834156,"owner":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 04:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2022-08-10 01:12:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2022-08-10 04:00:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":1,"date":"2022-08-10 04:00:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"reviewers":{"CC":[{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 11:50:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 02:04:41.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-04-06 07:11:38.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"state":"CC"},{"updated":"2022-07-29 08:53:20.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2022-08-04 12:59:27.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2022-08-04 12:59:27.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2022-08-04 12:59:27.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"85931ffc17ccb5f0b57cfd83b5f908a2163270bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-03-17 11:40:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f422175a6d794179ab4e8587b658734d933af48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 11:50:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94ed9c57ccf4cf283db96732824eece : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/127420bc04d5457da68f90dee43dbb66 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"35daa7fc28d4d9d83da7844152f7846906c59b28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-04-04 08:42:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a8fbc8b6f4d04419aa2025c739bb5d9273e7a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 08:47:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d92d8bc16a4ff3a78432e29dc3f0c4 : FAILURE in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af828c1fef224c64a1449476433d29ca : FAILURE in 3m 23s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/zed/enhance-multi-tenant-policy.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/zed/enhance-multi-tenant-policy.rst","accounts_in_message":[],"_revision_number":2},{"id":"a8bbaa21d078bf6eabc0190645dcd1b1e1d3ed13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-04-04 08:48:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc6025aed69f18a08702c8d251761944cc582399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 08:55:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70b254818c545eebd2059e167c98195 : FAILURE in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a9b880addc4b4f900a94fe1a75a09f : FAILURE in 2m 56s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/zed/enhance-multi-tenant-policy.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/zed/enhance-multi-tenant-policy.rst","accounts_in_message":[],"_revision_number":3},{"id":"6b3f6af9055d6039e82542b6a526f2b650d754ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-04-04 11:07:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d9c9cde97f4a21722dbf8ddc569843825562ba73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 11:18:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8fe67e075b40f9b29d238b0c42ba63 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39286fdac78b4d508270d77946d17ec7 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":4},{"id":"471ca81ad0033beccbea1605875744ef00482034","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-04-04 11:44:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1e549dd1d54f743a26fc4c400c77406ef6deb5c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 12:05:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d74df4e73e644eba0cb1fa1b9982247 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe35eec95674b69acd69ee867e4fe45 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":5},{"id":"ec006a9029fc421350dc0186bb1dc63ab9b7e98a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-04-04 14:04:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5d5ec8e1a0bc9a4dc07ba081578c390e2954190f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 14:16:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4546d4664e44b74938904a4f3b89178 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4197b9a2dd4b8e81ccb8ad30887073 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":6},{"id":"6021d2a414c7d61e2cca2471399a8796b4052227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-04-04 14:56:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5d6c2d2fee9f5fd48661f9c1b1f487c9e88dcaf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 15:07:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aafe4c91c6114e129326d7abbd8de2b8 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8209a3364f34b35b44fe22838358a41 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":7},{"id":"17a0de45c54d75d4507acfe8208ca771f77f1b74","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-06 02:04:41.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1a7231e9af681e5d9b216ea5d90c64ca4e101559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-04-06 03:02:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"922bfdad892438d1621cf923dba49aaf224fc859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 03:11:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/195171583e884c16b3ce9721c525dbe7 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518191522a804e52938bfd88907731fa : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":8},{"id":"ed42f2a9214a3bd1b50892acabcd7f878981f8c8","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-04-06 04:28:45.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"57f7f54e84ef3dec1e0cfbd22fd4ae9c47b8d624","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-07-29 08:53:20.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0ed7f5d257f3c33ee105102e380ceb622153679e","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-08-02 08:36:25.000000000","message":"Patch Set 8: -Code-Review","accounts_in_message":[],"_revision_number":8},{"id":"80cea7fe389a8a8c6e67b31eaa3510701a22a3ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-08-04 10:18:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d5cbc37d262b119650dedb91eec2a0e44d049fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:26:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/181f4d28f7ad4eafa8817b80f17ceb8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c7ea37cca644726ad4a282adea85554 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bee362168df4bcf96d8b1deab487df5 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":9},{"id":"8595131779f4704ce812bf5d388cb6f9e2ed4765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-08-04 11:02:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d6025d077cef2e7d20030be8509ad8ee1f983572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 11:09:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18e3cd63ed4432d921f3d540ca5928b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9206557a64354cc6a40245bef290a2dd : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3c3c2c143f04455983de03fbe1e11c6 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":10},{"id":"b3de4946e17b49dd8eef3233b8da8a6b90974dcd","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-08-05 04:21:21.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9a2b1f5dc0494fed458c584022e167cd4c0b139c","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-08-08 14:35:42.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2d0ea75d9fa7f43270b9ee19870d0d14f6ea1e89","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-08-09 06:01:52.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"896d81eef0624d46e42f0905ad087ce8479366e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-08-09 06:54:01.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"88c2e8ad8cabd2ff6e476fc5e48da35e4553446d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 06:59:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b93702468df64f5e890c351c09c59c0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b98298a908a48d1934d5f36d772aac0 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5272268f003c4a8cb6e6a5d0f0238452 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":11},{"id":"0df0b95e1e3523abb97929f2d9d847f088bcca8a","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-08-09 07:05:30.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"11f807aa3900f76575207b6d0bb2dcdc5e487e78","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-08-10 01:12:32.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7ec4f36fdcb4e32639fefd23758e2d87c17296eb","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-08-10 04:00:41.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"1c32aef7df77cbc084e27958293716e3d992e8d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 04:01:03.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1ad80c37d017a311a2b4d302428770eff1a8b557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 04:06:15.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df8f13188e994107bbee0b62c7f5e2c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7535d790e044a4b8524d6a752b0a82 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ec74c480f34f5e96273e98bac86f8a : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":11},{"id":"fc8af1d834918008509cd2ee6d13e7f7c6f4a870","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 04:06:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"60a8c6f6cc7d32c8bb67e474d09ddd5383cb28ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 04:07:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/559e277f909d48f2b48892d27ecfccb0\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/f1206f83fd9d41a9adfa2b2fd079dbca : SUCCESS in 51s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"aed90244a9a355e0de93e8e33d2a4b0e70c41049","revisions":{"cce520b3f11d373af7ba3c5a688f4431476e970b":{"kind":"REWORK","_number":1,"created":"2022-03-17 11:40:29.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/1"}}},"commit":{"parents":[{"commit":"fc25d850ff65169bbb4b914a937ca6ab5a816d4a","subject":"Add Zed specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/fc25d850ff65169bbb4b914a937ca6ab5a816d4a"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:40:05.000000000","tz":330},"subject":"[WIP]Enhance multi tenant policy in VNF LCM","message":"[WIP]Enhance multi tenant policy in VNF LCM\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/cce520b3f11d373af7ba3c5a688f4431476e970b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/cce520b3f11d373af7ba3c5a688f4431476e970b"}]},"branch":"refs/heads/master"},"681ce6cd5c7b0f3e927223c94e05d56f0703ee23":{"kind":"REWORK","_number":2,"created":"2022-04-04 08:42:39.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/2"}}},"commit":{"parents":[{"commit":"fc25d850ff65169bbb4b914a937ca6ab5a816d4a","subject":"Add Zed specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/fc25d850ff65169bbb4b914a937ca6ab5a816d4a"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-04-04 08:42:09.000000000","tz":330},"subject":"[WIP]Enhance multi tenant policy in VNF LCM","message":"[WIP]Enhance multi tenant policy in VNF LCM\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/681ce6cd5c7b0f3e927223c94e05d56f0703ee23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/681ce6cd5c7b0f3e927223c94e05d56f0703ee23"}]},"branch":"refs/heads/master"},"c5965362547b879fa7649a45ac6b72d9174d5c9a":{"kind":"REWORK","_number":3,"created":"2022-04-04 08:48:00.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/3"}}},"commit":{"parents":[{"commit":"fc25d850ff65169bbb4b914a937ca6ab5a816d4a","subject":"Add Zed specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/fc25d850ff65169bbb4b914a937ca6ab5a816d4a"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-04-04 08:47:39.000000000","tz":330},"subject":"[WIP]Enhance multi tenant policy in VNF LCM","message":"[WIP]Enhance multi tenant policy in VNF LCM\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c5965362547b879fa7649a45ac6b72d9174d5c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c5965362547b879fa7649a45ac6b72d9174d5c9a"}]},"branch":"refs/heads/master"},"540aa508963d25102e30ac5167df9738dceab3fa":{"kind":"REWORK","_number":4,"created":"2022-04-04 11:07:37.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/4"}}},"commit":{"parents":[{"commit":"fc25d850ff65169bbb4b914a937ca6ab5a816d4a","subject":"Add Zed specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/fc25d850ff65169bbb4b914a937ca6ab5a816d4a"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-04-04 11:07:09.000000000","tz":330},"subject":"Enhance multi tenant policy in VNF LCM","message":"Enhance multi tenant policy in VNF LCM\n\nAn enhancement of the multi-tenant policy in the VNF lifecycle\nmanagement.\n* It enables a non-admin role user to instantiate VNF.\n* Add design for negative functional test cases for VNF LCM\n  operations and validate notification is received by the tenant\n  specified in the subscription sequence.\n\nImplements: bp enhance-multi-tenant-policy\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/540aa508963d25102e30ac5167df9738dceab3fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/540aa508963d25102e30ac5167df9738dceab3fa"}]},"branch":"refs/heads/master"},"45dc19a3e003706bcbd6ef90c788dbd90457b9e8":{"kind":"REWORK","_number":5,"created":"2022-04-04 11:44:31.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/5"}}},"commit":{"parents":[{"commit":"fc25d850ff65169bbb4b914a937ca6ab5a816d4a","subject":"Add Zed specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/fc25d850ff65169bbb4b914a937ca6ab5a816d4a"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-04-04 11:44:11.000000000","tz":330},"subject":"Enhance multi tenant policy in VNF LCM","message":"Enhance multi tenant policy in VNF LCM\n\nAn enhancement of the multi-tenant policy in the VNF lifecycle\nmanagement.\n* It enables a non-admin role user to instantiate VNF.\n* Add design for negative functional test cases for VNF LCM\n  operations and validate notification is received by the tenant\n  specified in the subscription sequence.\n\nImplements: bp enhance-multi-tenant-policy\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/45dc19a3e003706bcbd6ef90c788dbd90457b9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/45dc19a3e003706bcbd6ef90c788dbd90457b9e8"}]},"branch":"refs/heads/master"},"b839d00813ab2cc4536ae9a4548900a60439375a":{"kind":"REWORK","_number":6,"created":"2022-04-04 14:04:47.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/6"}}},"commit":{"parents":[{"commit":"fc25d850ff65169bbb4b914a937ca6ab5a816d4a","subject":"Add Zed specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/fc25d850ff65169bbb4b914a937ca6ab5a816d4a"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-04-04 14:04:24.000000000","tz":330},"subject":"Enhance multi tenant policy in VNF LCM","message":"Enhance multi tenant policy in VNF LCM\n\nAn enhancement of the multi-tenant policy in the VNF lifecycle\nmanagement.\n* It enables a non-admin role user to instantiate VNF.\n* Add design for negative functional test cases for VNF LCM\n  operations and validate notification is received by the tenant\n  specified in the subscription sequence.\n\nImplements: bp enhance-multi-tenant-policy\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b839d00813ab2cc4536ae9a4548900a60439375a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b839d00813ab2cc4536ae9a4548900a60439375a"}]},"branch":"refs/heads/master"},"13a8454a0107554b424d18ea196596803f78d729":{"kind":"REWORK","_number":7,"created":"2022-04-04 14:56:19.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/7"}}},"commit":{"parents":[{"commit":"fc25d850ff65169bbb4b914a937ca6ab5a816d4a","subject":"Add Zed specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/fc25d850ff65169bbb4b914a937ca6ab5a816d4a"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-04-04 14:55:55.000000000","tz":330},"subject":"Enhance multi tenant policy in VNF LCM","message":"Enhance multi tenant policy in VNF LCM\n\nAn enhancement of the multi-tenant policy in the VNF lifecycle\nmanagement.\n* It enables a non-admin role user to instantiate VNF.\n* Add design for negative functional test cases for VNF LCM\n  operations and validate notification is received by the tenant\n  specified in the subscription sequence.\n\nImplements: bp enhance-multi-tenant-policy\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/13a8454a0107554b424d18ea196596803f78d729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/13a8454a0107554b424d18ea196596803f78d729"}]},"branch":"refs/heads/master"},"ca70a63cbbb5905e022d2eed48e9a52e24ca2a0e":{"kind":"REWORK","_number":8,"created":"2022-04-06 03:02:47.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/8"}}},"commit":{"parents":[{"commit":"fc25d850ff65169bbb4b914a937ca6ab5a816d4a","subject":"Add Zed specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/fc25d850ff65169bbb4b914a937ca6ab5a816d4a"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-04-06 03:02:06.000000000","tz":330},"subject":"Enhance multi tenant policy in VNF LCM","message":"Enhance multi tenant policy in VNF LCM\n\nAn enhancement of the multi-tenant policy in the VNF lifecycle\nmanagement.\n* It enables a non-admin role user to instantiate VNF.\n* Add design for negative functional test cases for VNF LCM\n  operations and validate notification is received by the tenant\n  specified in the subscription sequence.\n\nImplements: bp enhance-multi-tenant-policy\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/ca70a63cbbb5905e022d2eed48e9a52e24ca2a0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/ca70a63cbbb5905e022d2eed48e9a52e24ca2a0e"}]},"branch":"refs/heads/master"},"f25ed61113e40d505fe4db37bc65953af0494aaa":{"kind":"REWORK","_number":9,"created":"2022-08-04 10:18:00.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/9"}}},"commit":{"parents":[{"commit":"ab50b4706959f3f1adb6296bbe0e2ad523486867","subject":"Merge \"A DB migration tool for API version update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/ab50b4706959f3f1adb6296bbe0e2ad523486867"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-08-04 10:17:26.000000000","tz":330},"subject":"Enhance multi tenant policy in VNF LCM","message":"Enhance multi tenant policy in VNF LCM\n\nAn enhancement of the multi-tenant policy in the VNF lifecycle\nmanagement.\n* Discussion to enable a non-admin role user to instantiate VNF.\n* Add design for negative functional test cases for VNF LCM\n  operations and validate notification is received by the tenant\n  specified in the subscription sequence.\n\nImplements: bp enhance-multi-tenant-policy\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f25ed61113e40d505fe4db37bc65953af0494aaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f25ed61113e40d505fe4db37bc65953af0494aaa"}]},"branch":"refs/heads/master"},"caf4594f76fa903f10fafbfce10735fb8a7d93f9":{"kind":"REWORK","_number":10,"created":"2022-08-04 11:02:28.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/10"}}},"commit":{"parents":[{"commit":"ab50b4706959f3f1adb6296bbe0e2ad523486867","subject":"Merge \"A DB migration tool for API version update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/ab50b4706959f3f1adb6296bbe0e2ad523486867"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-08-04 11:02:04.000000000","tz":330},"subject":"Enhance multi tenant policy in VNF LCM","message":"Enhance multi tenant policy in VNF LCM\n\nAn enhancement of the multi-tenant policy in the VNF lifecycle\nmanagement.\n* Discussion to enable a non-admin role user to instantiate VNF.\n* Add design for negative functional test cases for VNF LCM\n  operations and validate notification is received by the tenant\n  specified in the subscription sequence.\n\nImplements: bp enhance-multi-tenant-policy\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/caf4594f76fa903f10fafbfce10735fb8a7d93f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/caf4594f76fa903f10fafbfce10735fb8a7d93f9"}]},"branch":"refs/heads/master"},"aed90244a9a355e0de93e8e33d2a4b0e70c41049":{"kind":"REWORK","_number":11,"created":"2022-08-09 06:54:01.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/56/834156/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/56/834156/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/56/834156/11"}}},"commit":{"parents":[{"commit":"ab50b4706959f3f1adb6296bbe0e2ad523486867","subject":"Merge \"A DB migration tool for API version update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/ab50b4706959f3f1adb6296bbe0e2ad523486867"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-03-17 11:37:08.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2022-08-09 06:53:21.000000000","tz":330},"subject":"Enhance multi tenant policy in VNF LCM","message":"Enhance multi tenant policy in VNF LCM\n\nAn enhancement of the multi-tenant policy in the VNF lifecycle\nmanagement.\n* Discussion to enable a non-admin role user to instantiate VNF.\n* Add design for negative functional test cases for VNF LCM\n  operations and validate notification is received by the tenant\n  specified in the subscription sequence.\n\nImplements: bp enhance-multi-tenant-policy\n\nChange-Id: Idf398ea9bf083afdecb99a0cea3c539a34ec17d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/aed90244a9a355e0de93e8e33d2a4b0e70c41049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/aed90244a9a355e0de93e8e33d2a4b0e70c41049"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[]}
