)]}'
{"id":"openstack%2Ftacker-specs~842077","triplet_id":"openstack%2Ftacker-specs~master~I1a810e9d3c926a18709f175fba6a907b09746a4d","project":"openstack/tacker-specs","branch":"master","topic":"spelling","attention_set":{},"removed_from_attention_set":{"34864":{"account":{"_account_id":34864,"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","username":"hicham-el-gharbi"},"last_update":"2022-05-25 02:23:25.000000000","reason":"Change was submitted"},"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-05-19 08:40:31.000000000","reason":"Ayumu Ueha replied on the change"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-05-25 02:05:42.000000000","reason":"Yasufumi Ogawa replied on the change"},"32102":{"account":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"last_update":"2022-05-25 02:23:25.000000000","reason":"Change was submitted"},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-05-25 02:23:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a810e9d3c926a18709f175fba6a907b09746a4d","subject":"Fix spelling mistakes","status":"MERGED","created":"2022-05-17 10:22:49.000000000","updated":"2022-05-25 02:25:42.000000000","submitted":"2022-05-25 02:23:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842077-spelling","meta_rev_id":"6e7c6f4bfe245eb1fb185dd2af9e37839b095ecf","_number":842077,"virtual_id_number":842077,"owner":{"_account_id":34864,"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","username":"hicham-el-gharbi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-25 02:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2022-05-19 08:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2022-05-25 02:05:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2022-05-18 07:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":1,"date":"2022-05-25 02:05:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 10:28:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 05:21:16.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2022-05-18 07:50:54.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2022-05-18 07:50:54.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2022-05-18 07:50:54.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"ecdb0de80b7258b94c8e150572f69aed7aee266a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34864,"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","username":"hicham-el-gharbi"},"date":"2022-05-17 10:22:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e07570b0ad7f992a8d96b1a61a75e1b8fadc15a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 10:28:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b42d88318fb43d585b83bb3a5ff01c4 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5f18a81fb246bebea0ab5c18c2363b : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1c734e5066f4e7d4f9ee497ee2363df0955f2b0e","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-05-18 05:21:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"00add8b84c8959a75de8cfde581624b80d6d09a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34864,"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","username":"hicham-el-gharbi"},"date":"2022-05-18 07:40:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8d40e4c9846a047b255e3ae862ef453e3e893461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 07:47:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f29b8e171f4595994e8cdaedf17e18 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a8fc80040ed498d8164e4c55063d14f : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"7139dd7734240451f600a48b64e0a2078a2b33ee","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-05-18 07:50:54.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ccf030dd8413c95b3005bc666054cbe0ffb684e5","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-05-19 08:40:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b99eaf5bf3fd3189441e77a193e4d69dab30f79","author":{"_account_id":34864,"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","username":"hicham-el-gharbi"},"date":"2022-05-24 09:05:53.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ea138efb080ac12e2df68becbba03a3e130399e4","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-05-25 02:05:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5c3229224524858cd08e50cbb85377298a8de9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 02:16:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"39ce4fa8e80622988d4d6ebe8c279ada85295642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 02:23:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd4480039444f8eab2f5ce47d3cac0c : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c37e6661bfb4590a997b8aa0f18ff3e : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9d3017a0ec1a06b3352998ffe607b22d42ea6317","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 02:23:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6e7c6f4bfe245eb1fb185dd2af9e37839b095ecf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 02:25:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/05282604b1bb45f79d900a95ff1eb4fe : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"923fd016802b9faa52827d969a94778fd7f2d210","revisions":{"992cca8dcb919c36eb52561545320943830af42a":{"kind":"REWORK","_number":1,"created":"2022-05-17 10:22:49.000000000","uploader":{"_account_id":34864,"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","username":"hicham-el-gharbi"},"ref":"refs/changes/77/842077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/77/842077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/77/842077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/77/842077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/77/842077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/77/842077/1"}}},"commit":{"parents":[{"commit":"1b6fc93b41fe6045a8a874f6a91a5a96d8c6c5ef","subject":"Merge \"Fix typo in multi tenant for LCM specification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1b6fc93b41fe6045a8a874f6a91a5a96d8c6c5ef"}]}],"author":{"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","date":"2022-05-17 10:21:10.000000000","tz":120},"committer":{"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","date":"2022-05-17 10:21:10.000000000","tz":120},"subject":"Fix spelling mistake","message":"Fix spelling mistake\n\nChange-Id: I1a810e9d3c926a18709f175fba6a907b09746a4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/992cca8dcb919c36eb52561545320943830af42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/992cca8dcb919c36eb52561545320943830af42a"}]},"branch":"refs/heads/master"},"923fd016802b9faa52827d969a94778fd7f2d210":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-18 07:40:38.000000000","uploader":{"_account_id":34864,"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","username":"hicham-el-gharbi"},"ref":"refs/changes/77/842077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/77/842077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/77/842077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/77/842077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/77/842077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/77/842077/2"}}},"commit":{"parents":[{"commit":"1b6fc93b41fe6045a8a874f6a91a5a96d8c6c5ef","subject":"Merge \"Fix typo in multi tenant for LCM specification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1b6fc93b41fe6045a8a874f6a91a5a96d8c6c5ef"}]}],"author":{"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","date":"2022-05-17 10:21:10.000000000","tz":120},"committer":{"name":"Hicham El Gharbi","email":"hicham.elgharbi@canonical.com","date":"2022-05-18 07:27:57.000000000","tz":120},"subject":"Fix spelling mistakes","message":"Fix spelling mistakes\n\nThe current version of the documentation contains two spelling mistakes\n\nThis update fix two spelling errors on the doc,\nthe first one is a misspelled word lifcycle instead of lifecycle,\nthe second is a missing space \"actionsuch\" instead of\n\"action such\"\n\nChange-Id: I1a810e9d3c926a18709f175fba6a907b09746a4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
