)]}'
{"id":"openstack%2Ftacker-specs~848806","triplet_id":"openstack%2Ftacker-specs~master~Ia3d4a9e99323b4039bda172e03b5d3a4f303e0a4","project":"openstack/tacker-specs","branch":"master","topic":"bp/centos-stream-testing","attention_set":{},"removed_from_attention_set":{"32736":{"account":{"_account_id":32736,"name":"Wataru Juso","email":"w-juso@nec.com","username":"w-juso"},"last_update":"2022-07-29 08:58:51.000000000","reason":"Change was submitted"},"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-07-21 00:06:33.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-07-29 08:58:51.000000000","reason":"Change was submitted"},"34346":{"account":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"last_update":"2022-07-29 08:58:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3d4a9e99323b4039bda172e03b5d3a4f303e0a4","subject":"Add CentOS Stream 9 Testing in Zuul","status":"MERGED","created":"2022-07-06 05:07:36.000000000","updated":"2022-07-29 08:59:59.000000000","submitted":"2022-07-29 08:58:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848806-bp/centos-stream-testing","meta_rev_id":"b3f521ea251206b1a530c820b6d3085d3116af63","_number":848806,"virtual_id_number":848806,"owner":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-29 08:58:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2022-07-21 00:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2022-07-29 08:50:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2022-07-20 14:00:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":1,"date":"2022-07-29 08:50:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32736,"name":"Wataru Juso","email":"w-juso@nec.com","username":"w-juso"},{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":32736,"name":"Wataru Juso","email":"w-juso@nec.com","username":"w-juso"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-06 05:10:02.000000000","updated_by":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2022-07-06 05:10:02.000000000","updated_by":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"reviewer":{"_account_id":32736,"name":"Wataru Juso","email":"w-juso@nec.com","username":"w-juso"},"state":"REVIEWER"},{"updated":"2022-07-06 05:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 09:28:41.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2022-07-20 14:00:55.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2022-07-29 08:50:25.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"cf00b60236c72b087d2842fe6e1bf1f5c30f6664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-06 05:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58496d0ac8a2ee973a45dd16e67c53eea7fa036d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-06 05:09:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1940dd964a5aac4b0e8be332beab13e769144e19","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-06 05:10:10.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"3f1c7c869d38f48df5e2a24414383b6a00443888","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-06 05:10:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6d6fb317075c8911a46a9e2395f892bca326540a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 05:16:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fe65c0341ad44e3be90c63d2fa654d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c818a7c032c7401898bc7d34cd487168 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8562f99196a7433b8dc7c511e6ce1c85 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":3},{"id":"fe186e89c0eb6bf37d14cdaddb89e12bf4767a89","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-11 07:01:31.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"8c353ef4214a178b6dfbe967a662f70cca36d77e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-11 07:01:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"e1851b4eb671cd66572f46e2247fa24c6ba30798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 07:07:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a74edc8290864159a03224ec7ae58ac6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f9a7acc55540a493eb6e57af6585a6 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ffea30f7a67491fb74e4465887ee27c : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":4},{"id":"c71d0404a1eff3c6eaf4d1fe7f626e99cac1135f","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-11 09:28:41.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1b2d75f9d81f585a72e2a709c01bd5b6030baf00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-12 15:35:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3620b8501189c457b31bd5fbfc065c83d45e9bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 15:44:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/303ff1b8a5fb41caa639fe91b8d1c765\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ce1c6e3cd74ef4bef0f6ad22f5f15d : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d605fba3394b48c89fc079714983c382 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":5},{"id":"450d0f1daf9596bc52696046a1c185aceace7105","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-12 15:46:59.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3dea4ed3ce20daf2a0538b5620e0e02bd972d9d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-19 08:17:40.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"f95d5a7d255a746591879a71de7038e1fae3af25","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"date":"2022-07-19 08:18:19.000000000","message":"Topic set to bp/centos-stream-testing","accounts_in_message":[],"_revision_number":6},{"id":"5f29455f85bcdf022b494752fc2cc997e911813c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 08:26:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfb31f83de4c48899372ec1d1e6389a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa55ef2ade2429fae6f6a4f3fa3b239 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f34229836ff481483b7966f5f352ead : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":6},{"id":"524a1d28ab6f88f30473a8390492a11b8ba3a041","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-07-20 14:00:55.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c745c99ec7e9fa5473f201f19f66e7c4ba546159","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-21 00:06:33.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e0e7fd051f57bd3e1600b9ea44e2de8513be1b37","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-07-29 08:50:25.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"da30a8c4dd68254787a9bcb4c5f6b2cef6ad883e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 08:50:45.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c4ab4fe5edc5593f3ce564217af608b8be3c9f60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 08:58:51.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19a0af0e829f433e828be6bebead05e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0509b4b2b3746769011abc2c10f6ed2 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d26b6280d3d4ec9a5ebd115fd8ed34a : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":6},{"id":"5b011c2a4a7eda9a33d1fa0c055affba85294879","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 08:58:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b3f521ea251206b1a530c820b6d3085d3116af63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 08:59:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1e8f6f39cb4f49a94b0f37ea90996f\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/f9427c3031354302ac82dadb6acf5836 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"77cfb6e1c0dc377c9e39524ba84a527a1bfb7b32","revisions":{"ed3791e512c72c599c3e328f9c8881f1c5bfa4bf":{"kind":"REWORK","_number":1,"created":"2022-07-06 05:07:36.000000000","uploader":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"ref":"refs/changes/06/848806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/06/848806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/1"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-06 05:05:53.000000000","tz":0},"committer":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-06 05:07:34.000000000","tz":0},"subject":"Add CentOS Stream 9 Testing in Zuul","message":"Add CentOS Stream 9 Testing in Zuul\n\nCurrently OpenStack is tested only on Ubuntu OS. However, some\ncommercial systems use Fedora family OS.\n\nTo check that Tacker works properly on Fedora family OS, we plan\nto add testsets on CentOS Stream 9 nodeset to the testsets\nin Zuul.\n\nChange-Id: Ia3d4a9e99323b4039bda172e03b5d3a4f303e0a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/ed3791e512c72c599c3e328f9c8881f1c5bfa4bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/ed3791e512c72c599c3e328f9c8881f1c5bfa4bf"}]},"branch":"refs/heads/master"},"e776bd2c59475087b834047ce1fa9be3750a6e71":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-06 05:09:18.000000000","uploader":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"ref":"refs/changes/06/848806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/06/848806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/2"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-06 05:05:53.000000000","tz":0},"committer":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-06 05:09:18.000000000","tz":0},"subject":"Add CentOS Stream 9 Testing in Zuul","message":"Add CentOS Stream 9 Testing in Zuul\n\nCurrently OpenStack is tested only on Ubuntu OS. However, some\ncommercial systems use Fedora family OS.\n\nTo check that Tacker works properly on Fedora family OS, we plan\nto add testsets on CentOS Stream 9 nodeset to the testsets\nin Zuul.\n\nChange-Id: Ia3d4a9e99323b4039bda172e03b5d3a4f303e0a4\nImplements: blueprint centos-stream-testing\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e776bd2c59475087b834047ce1fa9be3750a6e71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e776bd2c59475087b834047ce1fa9be3750a6e71"}]},"branch":"refs/heads/master","description":"Edit commit message"},"893a3f3deea9604eafcc68010216f01645560198":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-06 05:10:20.000000000","uploader":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"ref":"refs/changes/06/848806/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/06/848806/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/3"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-06 05:05:53.000000000","tz":0},"committer":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-06 05:10:20.000000000","tz":0},"subject":"[WIP] Add CentOS Stream 9 Testing in Zuul","message":"[WIP] Add CentOS Stream 9 Testing in Zuul\n\nCurrently OpenStack is tested only on Ubuntu OS. However, some\ncommercial systems use Fedora family OS.\n\nTo check that Tacker works properly on Fedora family OS, we plan\nto add testsets on CentOS Stream 9 nodeset to the testsets\nin Zuul.\n\nChange-Id: Ia3d4a9e99323b4039bda172e03b5d3a4f303e0a4\nImplements: blueprint centos-stream-testing\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/893a3f3deea9604eafcc68010216f01645560198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/893a3f3deea9604eafcc68010216f01645560198"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1a84ff0dc17a41b7228fbd60e6b098c993320738":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-07-11 07:01:31.000000000","uploader":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"ref":"refs/changes/06/848806/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/06/848806/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/4"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-06 05:05:53.000000000","tz":0},"committer":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-11 07:01:31.000000000","tz":0},"subject":"Add CentOS Stream 9 Testing in Zuul","message":"Add CentOS Stream 9 Testing in Zuul\n\nCurrently OpenStack is tested only on Ubuntu OS. However, some\ncommercial systems use Fedora family OS.\n\nTo check that Tacker works properly on Fedora family OS, we plan\nto add testsets on CentOS Stream 9 nodeset to the testsets\nin Zuul.\n\nChange-Id: Ia3d4a9e99323b4039bda172e03b5d3a4f303e0a4\nImplements: blueprint centos-stream-testing\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1a84ff0dc17a41b7228fbd60e6b098c993320738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1a84ff0dc17a41b7228fbd60e6b098c993320738"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e4b163352c7c3250c8adc1a36b97e2967617f8ef":{"kind":"REWORK","_number":5,"created":"2022-07-12 15:35:29.000000000","uploader":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"ref":"refs/changes/06/848806/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/06/848806/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/5"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-06 05:05:53.000000000","tz":0},"committer":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-12 15:35:14.000000000","tz":0},"subject":"Add CentOS Stream 9 Testing in Zuul","message":"Add CentOS Stream 9 Testing in Zuul\n\nCurrently OpenStack is tested only on Ubuntu OS. However, some\ncommercial systems use Fedora family OS.\n\nTo check that Tacker works properly on Fedora family OS, we plan\nto add testsets on CentOS Stream 9 nodeset to the testsets\nin Zuul.\n\nChange-Id: Ia3d4a9e99323b4039bda172e03b5d3a4f303e0a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e4b163352c7c3250c8adc1a36b97e2967617f8ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/e4b163352c7c3250c8adc1a36b97e2967617f8ef"}]},"branch":"refs/heads/master"},"77cfb6e1c0dc377c9e39524ba84a527a1bfb7b32":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-07-19 08:17:40.000000000","uploader":{"_account_id":34346,"name":"Pooja Singla","email":"pooja.singla@india.nec.com","username":"psingla"},"ref":"refs/changes/06/848806/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/06/848806/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/06/848806/6"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-06 05:05:53.000000000","tz":0},"committer":{"name":"Pooja Singla","email":"pooja.singla@india.nec.com","date":"2022-07-19 08:17:40.000000000","tz":0},"subject":"Add CentOS Stream 9 Testing in Zuul","message":"Add CentOS Stream 9 Testing in Zuul\n\nCurrently OpenStack is tested only on Ubuntu OS. However, some\ncommercial systems use Fedora family OS.\n\nTo check that Tacker works properly on Fedora family OS, we plan\nto add testsets on CentOS Stream 9 nodeset to the testsets\nin Zuul.\n\nImplement: blueprint centos-stream-testing\nChange-Id: Ia3d4a9e99323b4039bda172e03b5d3a4f303e0a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/77cfb6e1c0dc377c9e39524ba84a527a1bfb7b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/77cfb6e1c0dc377c9e39524ba84a527a1bfb7b32"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[]}
