)]}'
{"id":"openstack%2Ftacker-specs~849833","triplet_id":"openstack%2Ftacker-specs~master~I0a856508a22754eaa34ac2bb66137382209d416a","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-nfv-solv3-scale-vnf","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-07-25 00:27:22.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"32581":{"account":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"last_update":"2022-07-26 00:08:28.000000000","reason":"Change was submitted"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-07-26 00:03:05.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-07-26 00:08:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a856508a22754eaa34ac2bb66137382209d416a","subject":"Support CNF Scale in v2 LCM API","status":"MERGED","created":"2022-07-15 01:43:01.000000000","updated":"2022-07-26 00:09:30.000000000","submitted":"2022-07-26 00:08:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849833-bp/support-nfv-solv3-scale-vnf","meta_rev_id":"53c44b9806fcfd640c05dcf004840d56062b1b3a","_number":849833,"virtual_id_number":849833,"owner":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 00:08:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2022-07-25 00:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2022-07-26 00:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":1,"date":"2022-07-26 00:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 01:52:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 07:57:52.000000000","updated_by":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2022-07-15 07:57:52.000000000","updated_by":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2022-07-15 07:57:52.000000000","updated_by":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"d26e6c68490654bc4745f4ecf1113a5f6ebacac4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"date":"2022-07-15 01:43:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f614965877527bfd25e54b58d668951bc383dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 01:52:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9c9a810e643455c805974f84a1fbf40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5ab4f996844af9ab354efd3dccfdbd : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06eba5d882f547e088c520bb4f8d8590 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1e94c02df487e940e0a4b06ca275a0e799ff1209","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-20 06:58:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2a1a442331829d8b9c61bf5a99ef9dcc5231e5f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"date":"2022-07-22 08:07:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c9199b35c40c7692243761c8a1fb27737fd21c9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 08:14:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d83a18a429af406f82fd4045ad2c7e05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524ad49d92e241d697e6fcd32bbb6761 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9db1974f9c7e41ab9c4e7afb95e4a6a9 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"5c1046e8f64148d7463af52a570a729bcaf76709","author":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"date":"2022-07-22 08:21:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aa2820acd0d9ec18ce9f4d96dd6f00c9eba394b5","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-25 00:27:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a59d3b1accc8ce131383f30190f91a0cb212091e","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-07-26 00:03:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bbd52744dc897d8178a0b40b5beef9220d29e4aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 00:03:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef12d4fbfb8ae3b10e8e9ee4140287cffad9de98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 00:08:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d4b326e5e8431ea923b3058875fa7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c51884135e94a008d28eb129d7af253 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9f69dd1b6c41e08ec0ab77668965c1 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0fe4f0333ddecf65d91e32c28103d05299f7d81c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 00:08:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"53c44b9806fcfd640c05dcf004840d56062b1b3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 00:09:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be0c7a1bd2714e7cb9f22b4313daa55e\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/d6068d55f1c84b0cb44ebc6af1fe3f38 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"080dfc5fe432b3910756d2dce113637311c12d7b","revisions":{"46b0f275878e6121b5a1d8c1ea0dc27a79ab58fd":{"kind":"REWORK","_number":1,"created":"2022-07-15 01:43:01.000000000","uploader":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"ref":"refs/changes/33/849833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/33/849833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/33/849833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/33/849833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/33/849833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/33/849833/1"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-15 01:41:42.000000000","tz":540},"committer":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-15 01:41:42.000000000","tz":540},"subject":"Support CNF Scale in v2 LCM API","message":"Support CNF Scale in v2 LCM API\n\nThis specification enhances version 2 (v2) of Scale API\nfor supporting CNF.\nPrevious release supported VNF Lifecycle Management (LCM) operations\ndefined in ETSI NFV SOL002 v3.3.1 and SOL003 v3.3.1.\nHowever, v2 Scale API has not supported CNF yet.\nThis specification enables v2 Scale API to handle CNF.\n\nChange-Id: I0a856508a22754eaa34ac2bb66137382209d416a\nImplements: blueprint support-nfv-solv3-scale-vnf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/46b0f275878e6121b5a1d8c1ea0dc27a79ab58fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/46b0f275878e6121b5a1d8c1ea0dc27a79ab58fd"}]},"branch":"refs/heads/master"},"080dfc5fe432b3910756d2dce113637311c12d7b":{"kind":"REWORK","_number":2,"created":"2022-07-22 08:07:33.000000000","uploader":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"ref":"refs/changes/33/849833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/33/849833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/33/849833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/33/849833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/33/849833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/33/849833/2"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-15 01:41:42.000000000","tz":540},"committer":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-22 08:10:37.000000000","tz":540},"subject":"Support CNF Scale in v2 LCM API","message":"Support CNF Scale in v2 LCM API\n\nThis specification enhances version 2 (v2) of Scale API\nfor supporting CNF.\nPrevious release supported VNF Lifecycle Management (LCM) operations\ndefined in ETSI NFV SOL002 v3.3.1 and SOL003 v3.3.1.\nHowever, v2 Scale API has not supported CNF yet.\nThis specification enables v2 Scale API to handle CNF.\n\nChange-Id: I0a856508a22754eaa34ac2bb66137382209d416a\nImplements: blueprint support-nfv-solv3-scale-vnf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/080dfc5fe432b3910756d2dce113637311c12d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/080dfc5fe432b3910756d2dce113637311c12d7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[]}
