)]}'
{"id":"openstack%2Ftacker-specs~849834","triplet_id":"openstack%2Ftacker-specs~master~I136a996cae49e8d13e1bfc7703c80137afebf320","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-nfv-solv3-error-handling","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-07-26 02:15:53.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"32581":{"account":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"last_update":"2022-07-27 08:59:07.000000000","reason":"Change was submitted"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-07-27 08:52:44.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-07-27 08:59:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I136a996cae49e8d13e1bfc7703c80137afebf320","subject":"Support CNF Rollback in v2 LCM API","status":"MERGED","created":"2022-07-15 01:49:13.000000000","updated":"2022-07-27 09:18:30.000000000","submitted":"2022-07-27 08:59:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"849834-bp/support-nfv-solv3-error-handling","meta_rev_id":"19f4b45df2629c571f3bf249f9fd43b6ec4915cf","_number":849834,"virtual_id_number":849834,"owner":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-27 08:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"recommended":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"all":[{"value":1,"date":"2022-07-26 08:05:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},{"value":2,"date":"2022-07-26 02:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2022-07-27 08:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":1,"date":"2022-07-27 08:52:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"CC":[{"_account_id":31668,"deleted":true}],"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 01:59:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 07:58:19.000000000","updated_by":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2022-07-15 07:58:19.000000000","updated_by":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2022-07-15 07:58:19.000000000","updated_by":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2022-07-26 08:05:19.000000000","updated_by":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"reviewer":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"state":"REVIEWER"},{"updated":"2022-07-27 09:15:20.000000000","updated_by":{"_account_id":31668,"deleted":true},"reviewer":{"_account_id":31668,"deleted":true},"state":"CC"}],"messages":[{"id":"7feeb99bd3a68a457d31a1de103824667c97db5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"date":"2022-07-15 01:49:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9bb27cf4a9a2491e40dfcfd6ef45f62b41b6a79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 01:59:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6c191708b7a4ea29c15a2112270008d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7640c25c31394c7ab15b0fe09600350e : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c6239f241ac449c9703026d41055722 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"67dbf4909b7cb8fd70c3c596907d7f5a873dd637","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-21 02:44:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"da608a3a33f3705f39de7ed622370bf09cca9dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"date":"2022-07-25 06:38:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90474fd6ed2dec3acdcb2812698cd89bedd45a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 06:44:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aff9769f7b2c4d619c8da97f30ad8d16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49c64eed201a4a27a5fe312e6cdbe3d6 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c664bce3b8704d04949f3a1b65d1dc19 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e39f7d8804f832f1f48232a233d59dd86088e4ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"date":"2022-07-25 07:05:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"30dd2808de6986a0906a1ce8668b0a9ea96f65a8","author":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"date":"2022-07-25 07:13:48.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"32fb926ad3176e32b7cd2c6ef80f491aef49772b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 07:14:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e47d3598c6b473c85d3d171e96dac74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d4c75a1a2de4c13b24034f074c77504 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a598c216c51c44eabac7a2b1baad7d92 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":3},{"id":"b3c92f6fece5e12712fb9f9e334bf3cb49b122f3","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-25 15:14:44.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fa1582916f9b79985ea2e2c53bee69274176d5f5","author":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"date":"2022-07-26 02:05:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3a935ecd51a5574536b05c6c06c5ee9c669ff122","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-26 02:15:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"23b4c0c42fa309c1208787154219cf53f9abf135","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2022-07-26 08:05:19.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d8f700e914ba8f815c5716329643b1ff5905a7c","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-07-27 08:52:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8cc7b66566c0089860b71fd982022c5357ff63d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:53:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"783a9ec05e1929414566d69ee646d861ef189740","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:59:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/385cd104a1f04851ae0a6f53701fbf30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde1d4cb1e2d46309de5c509d8d76157 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32af27ded36042d6b018a25b9651373d : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":3},{"id":"af997bd9686bd163cd51ae0dfb82df3c30ab59cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:59:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"58e2f8f528483c076fb7eb40f20d64ab042d5d81","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 09:00:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0437add5eb964b70bc190511d981ce25\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/b955e1e961b44cee99533107efe70090 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3},{"id":"80f6612bbeec7e7f5cb69f055bd5d85bc1068a0d","author":{"_account_id":31668,"deleted":true},"date":"2022-07-27 09:15:20.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"19f4b45df2629c571f3bf249f9fd43b6ec4915cf","author":{"_account_id":31668,"deleted":true},"date":"2022-07-27 09:18:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bf04bd9a76a61718e1f33497fd7eada0d990b412","revisions":{"f0e70328b2febb9b6ab4ff84c433791aee690391":{"kind":"REWORK","_number":1,"created":"2022-07-15 01:49:13.000000000","uploader":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"ref":"refs/changes/34/849834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/34/849834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/1"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-15 01:51:28.000000000","tz":540},"committer":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-15 01:51:28.000000000","tz":540},"subject":"Support CNF Rollback in v2 LCM API","message":"Support CNF Rollback in v2 LCM API\n\nThis specification enhances version 2 (v2) of Rollback API\nfor supporting CNF.\nPrevious release supported VNF Lifecycle Management (LCM) operations\ndefined in ETSI NFV SOL002 v3.3.1 and SOL003 v3.3.1.\nHowever, v2 Rollback API for CNF LCM was limited.\nThis specification enables v2 Rollback API\nto handle Instantiation and Scale-out of CNF.\n\nChange-Id: I136a996cae49e8d13e1bfc7703c80137afebf320\nImplements: blueprint support-nfv-solv3-error-handling\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f0e70328b2febb9b6ab4ff84c433791aee690391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f0e70328b2febb9b6ab4ff84c433791aee690391"}]},"branch":"refs/heads/master"},"1bd7038150cd9becef081eb1db19f74dea647535":{"kind":"REWORK","_number":2,"created":"2022-07-25 06:38:06.000000000","uploader":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"ref":"refs/changes/34/849834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/34/849834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/2"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-15 01:51:28.000000000","tz":540},"committer":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-25 06:41:18.000000000","tz":540},"subject":"Support CNF Rollback in v2 LCM API","message":"Support CNF Rollback in v2 LCM API\n\nThis specification enhances version 2 (v2) of Rollback API\nfor supporting CNF.\nPrevious release supported VNF Lifecycle Management (LCM) operations\ndefined in ETSI NFV SOL002 v3.3.1 and SOL003 v3.3.1.\nHowever, v2 Rollback API for CNF LCM was limited.\nThis specification enables v2 Rollback API\nto handle Instantiation and Scale-out of CNF.\n\nChange-Id: I136a996cae49e8d13e1bfc7703c80137afebf320\nImplements: blueprint support-nfv-solv3-error-handling\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1bd7038150cd9becef081eb1db19f74dea647535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1bd7038150cd9becef081eb1db19f74dea647535"}]},"branch":"refs/heads/master"},"bf04bd9a76a61718e1f33497fd7eada0d990b412":{"kind":"REWORK","_number":3,"created":"2022-07-25 07:05:23.000000000","uploader":{"_account_id":32581,"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","username":"hirofumi-noguchi"},"ref":"refs/changes/34/849834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/34/849834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/34/849834/3"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-15 01:51:28.000000000","tz":540},"committer":{"name":"Hirofumi Noguchi","email":"hirofumi.noguchi.rs@hco.ntt.co.jp","date":"2022-07-25 07:08:37.000000000","tz":540},"subject":"Support CNF Rollback in v2 LCM API","message":"Support CNF Rollback in v2 LCM API\n\nThis specification enhances version 2 (v2) of Rollback API\nfor supporting CNF.\nPrevious release supported VNF Lifecycle Management (LCM) operations\ndefined in ETSI NFV SOL002 v3.3.1 and SOL003 v3.3.1.\nHowever, v2 Rollback API for CNF LCM was limited.\nThis specification enables v2 Rollback API\nto handle Instantiation and Scale-out of CNF.\n\nChange-Id: I136a996cae49e8d13e1bfc7703c80137afebf320\nImplements: blueprint support-nfv-solv3-error-handling\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/bf04bd9a76a61718e1f33497fd7eada0d990b412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/bf04bd9a76a61718e1f33497fd7eada0d990b412"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[]}
