)]}'
{"id":"openstack%2Ftacker-specs~850457","triplet_id":"openstack%2Ftacker-specs~master~Ia5633c2b6571bc5364f0ccd344d81f1a10236096","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-autoheal-queue","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-07-29 06:52:50.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"31668":{"account":{"_account_id":31668,"deleted":true},"last_update":"2022-07-29 01:48:35.000000000","reason":"\u003cGERRIT_ACCOUNT_31668\u003e replied on the change","reason_account":{"_account_id":31668,"deleted":true}},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-07-29 08:00:14.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"32102":{"account":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"last_update":"2022-07-29 08:11:30.000000000","reason":"Change was submitted"},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-07-29 08:11:30.000000000","reason":"Change was submitted"},"33292":{"account":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"last_update":"2022-07-29 08:11:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia5633c2b6571bc5364f0ccd344d81f1a10236096","subject":"Support VNF AutoHeal triggered by FaultNotification","status":"MERGED","created":"2022-07-20 07:46:19.000000000","updated":"2022-07-29 08:12:48.000000000","submitted":"2022-07-29 08:11:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":59,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"850457-bp/support-autoheal-queue","meta_rev_id":"473536582a726717fdc0945e7910442c4d2edf31","_number":850457,"virtual_id_number":850457,"owner":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":31668},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-29 08:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2022-07-29 06:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":1,"date":"2022-07-29 07:49:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31668},{"value":2,"date":"2022-07-29 08:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":31668},{"value":1,"date":"2022-07-29 08:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31668,"deleted":true},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-20 07:52:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-20 08:43:27.000000000","updated_by":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2022-07-20 08:43:27.000000000","updated_by":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2022-07-20 08:43:27.000000000","updated_by":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2022-07-25 07:53:33.000000000","updated_by":{"_account_id":31668,"deleted":true},"reviewer":{"_account_id":31668,"deleted":true},"state":"CC"},{"updated":"2022-07-26 01:45:36.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"CC"},{"updated":"2022-07-26 01:46:26.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2022-07-29 07:49:38.000000000","updated_by":{"_account_id":31668,"deleted":true},"reviewer":{"_account_id":31668,"deleted":true},"state":"REVIEWER"}],"messages":[{"id":"d029ff2fe25032b0440533574ce81c57b6f2325c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-07-20 07:46:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"538d5b6af407ada829f42b7a186a286ad7aaa288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 07:52:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c111d71519c4eb9a224ac67961d7470\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/647de40edbcf4f178a06797424e1cab7 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e984798e8cf44fa58a24f60c16cc55a8 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f0a014f223b0b705049e77ac00364ac7da6520b5","author":{"_account_id":31668,"deleted":true},"date":"2022-07-25 07:53:33.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"04dbbc452ee01e42fca857a7babc71d51dbf3404","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-07-26 01:45:36.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"60d210959f29b19a23577cc7f40bac0bf9d9b2a3","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2022-07-26 01:46:26.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"2739aba8d0f9df8d084f75e4c4247fe8283d4b35","author":{"_account_id":31668,"deleted":true},"date":"2022-07-26 02:19:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7636a3297466a2a4d52145f90a04275274d7d0f","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-27 09:32:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7bbef268619a2928484117c7b8114d27115f2c8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-07-28 08:11:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c6075490e7f812f6a12bac75f0e74f03c001b33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 08:20:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63cd3515fc5e464abb39fe61969dea94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f0d91b8d064cba94f874cab8556ec9 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aabbfc020e05493d8cfc6d391ad9f778 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ca72884f98a27c7af736db757c2f8a55a994c594","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-07-28 08:42:14.000000000","message":"Patch Set 2:\n\n(17 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b80091510c5973ceb01785f17b0d53057ffa1985","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-28 09:08:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"68adb5d0f31ee7e3f044f7109f58ee1fa171b9f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-07-28 09:31:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4e33a51fd1872b94c3e04d15f7cb77447491ec01","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-07-28 09:36:28.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"651abd42fba303189f5fc003d738c409ea969092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 09:36:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/588335dc43ee484e820f10548d6a276e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b16c5d832a47f0a96fed541dc4dacb : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48f34c749e71402e8402a2dd3f763913 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"9b37c0a219b220c03c0dcb9c6553355281e609e1","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-28 12:04:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d563ef32d9e1b213a4318cbf5da51426a9949ddd","author":{"_account_id":31668,"deleted":true},"date":"2022-07-29 01:48:35.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ed4ae5947bf79dd8e5d440591a99edf94d42f114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-07-29 05:27:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bb6fc3e0c37b342de96c45aecd04b63d8677a957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 05:33:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98bc6960a358474bbac5c449eb340081\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c3cafd7c00e4bf9bccefcc3a5874d1a : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6390ca97aaf4436bec52690f56745e1 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"96e18df6803d4a7f948c821991926ffb29a27904","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-29 05:48:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"063a3ed103d613e10af248573ba3b0e0e4de47ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-07-29 06:08:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"13752186b98d0fed8db4f8e1aef79da3e3ad3b23","author":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"date":"2022-07-29 06:12:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1ac6c380cec9e24a4440b037701e8ae97edd2868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 06:13:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19acd2e3071e4b4188fa77021164841f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2103daf113a4a88b0640d0ecd4f3f41 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f82c4f4f614471923f6cdbaba567df : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":5},{"id":"99edd6141caf9f15915680740855d154e80a7de0","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-07-29 06:52:50.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3f1f16b205ff436d4bf32af76febab444eed257e","author":{"_account_id":31668,"deleted":true},"date":"2022-07-29 07:49:38.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0e8b4a40a73b62e4813741b6f85ff89c8388650a","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-07-29 08:00:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c4d7776c9b640e68614a457f73125998a68e6adc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 08:00:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b1ac54c4311abc4eb92e25b2d6c0d7fa15b3db5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 08:11:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ec8fab9252e4a4d82075f4369733d61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651bfbe4b8f44457be13cd9178e41758 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6af03fbfe74751b7c45553ae4caf8c : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":5},{"id":"9d8843cec346e050a4d3cc73235de19ae850a319","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 08:11:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"473536582a726717fdc0945e7910442c4d2edf31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 08:12:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0a8ffa656a94ee5893ee4e92fd57b98\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/d2c8621856f24b17b31c66e85d799ff0 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4c4778c37212e317c10064f691321ea5d7560a11","revisions":{"54cd18d0d1edca5695470246a70c1c663febfbcc":{"kind":"REWORK","_number":1,"created":"2022-07-20 07:46:19.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/57/850457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/57/850457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/1"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Masaki Ueno","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-20 07:45:56.000000000","tz":540},"committer":{"name":"Masaki Ueno","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-20 07:45:56.000000000","tz":540},"subject":"Support VNF AutoHeal triggered by FaultNotification","message":"Support VNF AutoHeal triggered by FaultNotification\n\nThis spec provides an implementation for supporting AutoHealing using\nFaultNotification interface between Tacker and VIM.\n\nWhen fault events occur in VIM, VIM notifies fault event to Tacker\nvia the interface.\n\nTacker takes initiative of AutoHealing. According to configuration,\nTacker checks faultID attribute in the fault event and determines\nwhether AutoHealing should be performed. In case of performing\nAutoHealing, VMs are deleted or created via Heat.\n\nThe FaultNotification interface can be alerted by multiple events.\nTo prevent invoking multiple heal operations to single VNF,\nthe FaultNotification requests to Tacker are packed for a configured\nperiod of time.\n\nImplements: blueprint support-autoheal-queue\nSigned-off-by: Masaki Ueno \u003cmasaki.ueno.up@hco.ntt.co.jp\u003e\nChange-Id: Ia5633c2b6571bc5364f0ccd344d81f1a10236096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/54cd18d0d1edca5695470246a70c1c663febfbcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/54cd18d0d1edca5695470246a70c1c663febfbcc"}]},"branch":"refs/heads/master"},"47b5bb35ea3bd994f6157a913205eaa4161d6cb4":{"kind":"REWORK","_number":2,"created":"2022-07-28 08:11:50.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/57/850457/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/57/850457/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/2"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Masaki Ueno","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-20 07:45:56.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-28 08:11:37.000000000","tz":540},"subject":"Support VNF AutoHeal triggered by FaultNotification","message":"Support VNF AutoHeal triggered by FaultNotification\n\nThis spec provides an implementation for supporting AutoHealing using\nFaultNotification interface between Tacker and VIM.\n\nWhen fault events occur in VIM, VIM notifies fault event to Tacker\nvia the interface.\n\nTacker takes initiative of AutoHealing. According to configuration,\nTacker checks faultID attribute in the fault event and determines\nwhether AutoHealing should be performed. In case of performing\nAutoHealing, VMs are deleted or created via Heat.\n\nThe FaultNotification interface can be alerted by multiple events.\nTo prevent invoking multiple heal operations to single VNF,\nthe FaultNotification requests to Tacker are packed for a configured\nperiod of time.\n\nImplements: blueprint support-autoheal-queue\nSigned-off-by: Masaki Ueno \u003cmasaki.ueno.up@hco.ntt.co.jp\u003e\nChange-Id: Ia5633c2b6571bc5364f0ccd344d81f1a10236096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/47b5bb35ea3bd994f6157a913205eaa4161d6cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/47b5bb35ea3bd994f6157a913205eaa4161d6cb4"}]},"branch":"refs/heads/master"},"cfcd096e34997dea08540c6126add675b91c2f22":{"kind":"REWORK","_number":3,"created":"2022-07-28 09:31:32.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/57/850457/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/57/850457/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/3"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Masaki Ueno","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-20 07:45:56.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-28 09:31:23.000000000","tz":540},"subject":"Support VNF AutoHeal triggered by FaultNotification","message":"Support VNF AutoHeal triggered by FaultNotification\n\nThis spec provides an implementation for supporting AutoHealing using\nFaultNotification interface between Tacker and VIM.\n\nWhen fault events occur in VIM, VIM notifies fault event to Tacker\nvia the interface.\n\nTacker takes initiative of AutoHealing. According to configuration,\nTacker checks faultID attribute in the fault event and determines\nwhether AutoHealing should be performed. In case of performing\nAutoHealing, VMs are deleted or created via Heat.\n\nThe FaultNotification interface can be alerted by multiple events.\nTo prevent invoking multiple heal operations to single VNF,\nthe FaultNotification requests to Tacker are packed for a configured\nperiod of time.\n\nImplements: blueprint support-autoheal-queue\nSigned-off-by: Masaki Ueno \u003cmasaki.ueno.up@hco.ntt.co.jp\u003e\nChange-Id: Ia5633c2b6571bc5364f0ccd344d81f1a10236096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/cfcd096e34997dea08540c6126add675b91c2f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/cfcd096e34997dea08540c6126add675b91c2f22"}]},"branch":"refs/heads/master"},"cb865c4306db558eb8bc7e43a8359d69355f95aa":{"kind":"REWORK","_number":4,"created":"2022-07-29 05:27:31.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/57/850457/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/57/850457/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/4"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Masaki Ueno","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-20 07:45:56.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-29 05:27:21.000000000","tz":540},"subject":"Support VNF AutoHeal triggered by FaultNotification","message":"Support VNF AutoHeal triggered by FaultNotification\n\nThis spec provides an implementation for supporting AutoHealing using\nFaultNotification interface between Tacker and VIM.\n\nWhen fault events occur in VIM, VIM notifies fault event to Tacker\nvia the interface.\n\nTacker takes initiative of AutoHealing. According to configuration,\nTacker checks faultID attribute in the fault event and determines\nwhether AutoHealing should be performed. In case of performing\nAutoHealing, VMs are deleted or created via Heat.\n\nThe FaultNotification interface can be alerted by multiple events.\nTo prevent invoking multiple heal operations to single VNF,\nthe FaultNotification requests to Tacker are packed for a configured\nperiod of time.\n\nImplements: blueprint support-autoheal-queue\nSigned-off-by: Masaki Ueno \u003cmasaki.ueno.up@hco.ntt.co.jp\u003e\nChange-Id: Ia5633c2b6571bc5364f0ccd344d81f1a10236096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/cb865c4306db558eb8bc7e43a8359d69355f95aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/cb865c4306db558eb8bc7e43a8359d69355f95aa"}]},"branch":"refs/heads/master"},"4c4778c37212e317c10064f691321ea5d7560a11":{"kind":"REWORK","_number":5,"created":"2022-07-29 06:08:09.000000000","uploader":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"ref":"refs/changes/57/850457/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/57/850457/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/57/850457/5"}}},"commit":{"parents":[{"commit":"923fd016802b9faa52827d969a94778fd7f2d210","subject":"Fix spelling mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/923fd016802b9faa52827d969a94778fd7f2d210"}]}],"author":{"name":"Masaki Ueno","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-20 07:45:56.000000000","tz":540},"committer":{"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","date":"2022-07-29 06:08:02.000000000","tz":540},"subject":"Support VNF AutoHeal triggered by FaultNotification","message":"Support VNF AutoHeal triggered by FaultNotification\n\nThis spec provides an implementation for supporting AutoHealing using\nFaultNotification interface between Tacker and VIM.\n\nWhen fault events occur in VIM, VIM notifies fault event to Tacker\nvia the interface.\n\nTacker takes initiative of AutoHealing. According to configuration,\nTacker checks faultID attribute in the fault event and determines\nwhether AutoHealing should be performed. In case of performing\nAutoHealing, VMs are deleted or created via Heat.\n\nThe FaultNotification interface can be alerted by multiple events.\nTo prevent invoking multiple heal operations to single VNF,\nthe FaultNotification requests to Tacker are packed for a configured\nperiod of time.\n\nImplements: blueprint support-autoheal-queue\nSigned-off-by: Masaki Ueno \u003cmasaki.ueno.up@hco.ntt.co.jp\u003e\nChange-Id: Ia5633c2b6571bc5364f0ccd344d81f1a10236096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/4c4778c37212e317c10064f691321ea5d7560a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/4c4778c37212e317c10064f691321ea5d7560a11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[]}
