)]}'
{"id":"openstack%2Ftacker-specs~868254","triplet_id":"openstack%2Ftacker-specs~master~I7bc8af96e3e2a3290891f42821597184d5f77a25","project":"openstack/tacker-specs","branch":"master","topic":"bp/support-auto-lcm","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-12-28 05:13:59.000000000","reason":"Change was submitted"},"32707":{"account":{"_account_id":32707,"name":"Kenta Fukaya","email":"kenta.fukaya.xv@hco.ntt.co.jp","username":"fukayan"},"last_update":"2022-12-28 05:13:59.000000000","reason":"Change was submitted"},"31668":{"account":{"_account_id":31668,"deleted":true},"last_update":"2022-12-27 06:39:44.000000000","reason":"\u003cGERRIT_ACCOUNT_31668\u003e replied on the change","reason_account":{"_account_id":31668,"deleted":true}},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-12-28 05:04:27.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-12-27 10:03:18.000000000","reason":"\u003cGERRIT_ACCOUNT_17255\u003e replied on the change","reason_account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}},"34712":{"account":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"last_update":"2022-12-28 05:13:59.000000000","reason":"Change was submitted"},"33455":{"account":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"last_update":"2022-12-28 05:13:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7bc8af96e3e2a3290891f42821597184d5f77a25","subject":"Support Thresholds resource in PM Interface","status":"MERGED","created":"2022-12-21 03:26:28.000000000","updated":"2022-12-28 05:14:58.000000000","submitted":"2022-12-28 05:13:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":206,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868254-bp/support-auto-lcm","meta_rev_id":"9074134602770878c85511feb5bf2ee90c6a3155","_number":868254,"virtual_id_number":868254,"owner":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31668},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-28 05:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"recommended":{"_account_id":31668,"deleted":true},"all":[{"value":1,"date":"2022-12-27 06:39:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31668},{"value":2,"date":"2022-12-28 05:04:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2022-12-27 10:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31668},{"value":1,"date":"2022-12-28 05:04:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":32707,"name":"Kenta Fukaya","email":"kenta.fukaya.xv@hco.ntt.co.jp","username":"fukayan"},{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31668,"deleted":true},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":32707,"name":"Kenta Fukaya","email":"kenta.fukaya.xv@hco.ntt.co.jp","username":"fukayan"},{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-21 03:35:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-21 04:54:50.000000000","updated_by":{"_account_id":31668,"deleted":true},"reviewer":{"_account_id":31668,"deleted":true},"state":"REVIEWER"},{"updated":"2022-12-22 07:20:51.000000000","updated_by":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2022-12-22 07:20:51.000000000","updated_by":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2022-12-22 07:20:51.000000000","updated_by":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2022-12-22 07:20:51.000000000","updated_by":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"reviewer":{"_account_id":32707,"name":"Kenta Fukaya","email":"kenta.fukaya.xv@hco.ntt.co.jp","username":"fukayan"},"state":"REVIEWER"},{"updated":"2022-12-22 07:20:51.000000000","updated_by":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"reviewer":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"state":"REVIEWER"}],"messages":[{"id":"87671bafbf27488297c174b6fe80c0835861e2e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-21 03:26:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"701dddf010125a4b5ab7f6b3de39664cf2e5fff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 03:35:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b9b562657ca429e995e6846e029b2a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3b27d6dbb543cc8429af08397ec6c7 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f70449a0085a431a88bfc3c3f067c6d4 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1},{"id":"119ff4b5a4b145c7c139dc54259b6e1c4377fcee","author":{"_account_id":31668,"deleted":true},"date":"2022-12-21 04:54:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(21 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8addde6b5d8082035cb55ae4bc5190f0e60c250b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-21 07:45:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7444c862a8b05fa1f1af23eb01fa3f25e65168f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 07:50:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4b9366abb804ca8a1c60f078e02b3c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ba56903cbf4b47aafc96b035db1462 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf67fe18b1e04f24bd6871ecb41c0f6b : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b074efcafbe1caf5331c38433909969e15f16f4d","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-21 07:56:33.000000000","message":"Patch Set 2:\n\n(21 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0073ae52f66a5a46baec4fd91e8801ee32324460","author":{"_account_id":31668,"deleted":true},"date":"2022-12-21 08:19:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(50 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a23735cc90802288674d759d8e38c57aaa4f3f4d","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-22 01:43:03.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9b713a2763d2399b78bf17ffb75303ccd16e9bc2","author":{"_account_id":31668,"deleted":true},"date":"2022-12-22 02:58:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b4dc30e7efc4463e07e17d6d2653c25f7409d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-22 05:08:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"42db7b1c2f43f268fac47fe1eeabfbcff4a731df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 05:18:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aad97fdc64f54986be0c7bd4c84627ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d6de6ff894041f8841df9497f3fb1aa : POST_FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6579f63dded4d7887f82046692c7e4e : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"bd4a85520ac83089ce14693a92ddc487af55220d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-22 06:44:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a9916966b658b2a85437668f2ba0156e604e5825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 06:50:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66da93c4c283471ab293e10c057ca693\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c89f469bb5c4e7eb1b696259056ed03 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/753faf658b6047f598773cbb64195c1e : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4},{"id":"bb89198afc87cfe8c54db1664bc7ac7e9b9a0f23","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-22 07:17:59.000000000","message":"Patch Set 4:\n\n(32 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5184b25993c09004d57ffa0f06669e27d1bb3b5d","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-22 07:20:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"63872002dcfb3adbaf74dbf9b4853a13d9d1e5ac","author":{"_account_id":31668,"deleted":true},"date":"2022-12-22 10:26:54.000000000","message":"Patch Set 4: Code-Review-1\n\n(31 comments)","accounts_in_message":[],"_revision_number":4},{"id":"44f58cc50b5d275587d23045efdad25c480b255c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-22 12:32:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4d02dc7419edc5e2bb76e76120f0beca6bd0c894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 12:42:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52f3af349aa24a69915260087e076474\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d895424f84462580b5e0bf6a3ea018 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d55a4e537d0249958f4d7e49bdaa7342 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":5},{"id":"a1a1901bac44519c83b6d94c8effe953a95ebf4a","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-22 12:44:50.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fe5f6e5fdf0a1c21f7a4c352fff703308826eacb","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-22 12:46:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e5d7ceea0231942cebfe43e0a76524c75c0da92c","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-22 12:48:02.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c13093bf09b2aad851aceb19864d177521ca768e","author":{"_account_id":31668,"deleted":true},"date":"2022-12-23 01:23:37.000000000","message":"Patch Set 5: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8898eae0bf382c20997da68724ab57ed9ab407a4","author":{"_account_id":31668,"deleted":true},"date":"2022-12-23 03:00:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cf6c714422316d9f8aaec00d2cc946b7554828a7","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-23 06:08:53.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"93f43af85a8fc79c466e657b2951c4329c12be77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-23 12:07:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"509235799a1f9681fccad31c8cac30622dc7d35e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 12:14:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86cac716f06e43bfaf4eae68ec6a2160\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd25d745046049d096e9128c3a24f21f : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f809e449cc4143ada24d2a6c415a960a : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":6},{"id":"5a83549dac4c8d241b148c2fc7db6d163b7907aa","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-23 12:29:18.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a861b4a9d5721018c8635cee3ff1d0ae3603f19c","author":{"_account_id":31668,"deleted":true},"date":"2022-12-26 04:31:51.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"149d95877a0d8f198ef84b3cb636e8f26f524510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-26 11:11:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a29de7058004d4c67d9d32371597924f0fd327d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 11:20:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5bdff0ca31549ff991db039db9dd710\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72243a477dd4ef0a266f0b235477add : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8b91ab1f9649f68009213ec7230f28 : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":7},{"id":"4d5dd162f0ef0b6e99f888d70b7ec4fa2f71519c","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-26 11:26:33.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"41e4701f9eefa71cbfd1bd7fcc09e1259e3f25aa","author":{"_account_id":31668,"deleted":true},"date":"2022-12-27 02:17:49.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0bbf9b7245e5425dd5994136639666399647f493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-27 02:32:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fff25d3aa0a9a39200eb8a088aecdc7e1f174cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 02:40:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/855502596bfd4a54acc1ea47e7cc724b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b395b7b07a4644f3a000d303d0f54ff9 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c2a8ecccc945228e130b448055e753 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":8},{"id":"095feb7d1bbefb29999af66b71ff80fc377e6283","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2022-12-27 02:47:19.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"60251a6586860cc018511c2edc33e917586e7c41","author":{"_account_id":31668,"deleted":true},"date":"2022-12-27 06:39:44.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b36477854c1b9950088bb12345a55fb0303daf00","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2022-12-27 10:03:18.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"cfb7db437fc3900fe6b630d1c30f97c23b7ec4e9","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-12-28 05:04:27.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"af546acbeb6703c4829608e02a85f94d2c987b98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 05:04:46.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"073135f528d84f3ef1c7e85743064ef534c3a607","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 05:13:59.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c9ddbaad1d4a13ab7ddbf6c1533ba6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49481788658b46e4af704945e0c2aed2 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c46929d0eb54f3c9af81f4f75adec66 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":8},{"id":"e83ca3900174792cd2fbf2c638430c4ff50aab22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 05:13:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"9074134602770878c85511feb5bf2ee90c6a3155","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 05:14:58.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4582e63a06b8475f8184ba36ebc352a7\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/72f70b6323464f1b8fe32417fb1b0fe3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"278ccb8bd096da8e7ebe1a847b1dfe13b8c5ee19","revisions":{"b2949d8ece5a98f86580465e89c2bc73c99e6b76":{"kind":"REWORK","_number":1,"created":"2022-12-21 03:26:28.000000000","uploader":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"ref":"refs/changes/54/868254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/868254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/1"}}},"commit":{"parents":[{"commit":"949c4ce423abda4b1c40dfb24699140f36f9dec9","subject":"Add 2023.1 specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/949c4ce423abda4b1c40dfb24699140f36f9dec9"}]}],"author":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 03:25:32.000000000","tz":540},"committer":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 03:25:32.000000000","tz":540},"subject":"Support Threshold resource in PM Interface","message":"Support Threshold resource in PM Interface\n\nThis specification provides implementations of \"Thresholds\"\nresource for supporting Performance Management interface\nwith external monitoring tools such as Prometheus.\n\nTacker Zed release has supported ETSI NFV-SOL\nbased Fault Management and Performance Management interfaces,\nbut Threshold resource is not implemented.\n\nThe following items will be supported.\n1. Add support Threshold resource for PM interface.\n2. Enhance notification to Client.\n3. Add support of RESTful API for communications between\n   Tacker and External Monitoring Tool.\n4. Add new unit and functional tests.\n\nImplements: blueprint support-auto-lcm\nChange-Id: I7bc8af96e3e2a3290891f42821597184d5f77a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b2949d8ece5a98f86580465e89c2bc73c99e6b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/b2949d8ece5a98f86580465e89c2bc73c99e6b76"}]},"branch":"refs/heads/master"},"f17dc7564ca6ddcfe75b063582920d7d1b552339":{"kind":"REWORK","_number":2,"created":"2022-12-21 07:45:47.000000000","uploader":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"ref":"refs/changes/54/868254/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/868254/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/2"}}},"commit":{"parents":[{"commit":"949c4ce423abda4b1c40dfb24699140f36f9dec9","subject":"Add 2023.1 specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/949c4ce423abda4b1c40dfb24699140f36f9dec9"}]}],"author":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 03:25:32.000000000","tz":540},"committer":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 07:45:07.000000000","tz":540},"subject":"Support Threshold resource in PM Interface","message":"Support Threshold resource in PM Interface\n\nThis specification provides implementations of \"Thresholds\"\nresource for supporting Performance Management interface\nwith external monitoring tools such as Prometheus.\n\nTacker Zed release has supported ETSI NFV-SOL\nbased Fault Management and Performance Management interfaces,\nbut Threshold resource is not implemented.\n\nThe following items will be supported.\n1. Add support Threshold resource for PM interface.\n2. Enhance notification to Client.\n3. Add support of RESTful API for communications between\n   Tacker and External Monitoring Tool.\n4. Add new unit and functional tests.\n\nImplements: blueprint support-auto-lcm\nChange-Id: I7bc8af96e3e2a3290891f42821597184d5f77a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f17dc7564ca6ddcfe75b063582920d7d1b552339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f17dc7564ca6ddcfe75b063582920d7d1b552339"}]},"branch":"refs/heads/master"},"8a3d66f9549ee7b501d3e767bf1324afb924079d":{"kind":"REWORK","_number":3,"created":"2022-12-22 05:08:19.000000000","uploader":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"ref":"refs/changes/54/868254/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/868254/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/3"}}},"commit":{"parents":[{"commit":"949c4ce423abda4b1c40dfb24699140f36f9dec9","subject":"Add 2023.1 specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/949c4ce423abda4b1c40dfb24699140f36f9dec9"}]}],"author":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 03:25:32.000000000","tz":540},"committer":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-22 05:07:41.000000000","tz":540},"subject":"Support Thresholds resource in PM Interface","message":"Support Thresholds resource in PM Interface\n\nThis specification provides implementations of \"Thresholds\"\nresource for supporting Performance Management interface\nwith external monitoring tools such as Prometheus.\n\nTacker Zed release has supported ETSI NFV-SOL\nbased Fault Management and Performance Management interfaces,\nbut Thresholds resource is not implemented.\n\nThe following items will be supported.\n1. Add support Thresholds resource for PM interface.\n2. Enhance notification to Client.\n3. Add support of RESTful API for communications between\n   Tacker and External Monitoring Tool.\n4. Add new unit and functional tests.\n\nImplements: blueprint support-auto-lcm\nChange-Id: I7bc8af96e3e2a3290891f42821597184d5f77a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8a3d66f9549ee7b501d3e767bf1324afb924079d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/8a3d66f9549ee7b501d3e767bf1324afb924079d"}]},"branch":"refs/heads/master"},"894b2088f3ca4230aa5a90ddd69873c38a9c584f":{"kind":"REWORK","_number":4,"created":"2022-12-22 06:44:49.000000000","uploader":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"ref":"refs/changes/54/868254/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/868254/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/4"}}},"commit":{"parents":[{"commit":"949c4ce423abda4b1c40dfb24699140f36f9dec9","subject":"Add 2023.1 specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/949c4ce423abda4b1c40dfb24699140f36f9dec9"}]}],"author":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 03:25:32.000000000","tz":540},"committer":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-22 06:41:36.000000000","tz":540},"subject":"Support Thresholds resource in PM Interface","message":"Support Thresholds resource in PM Interface\n\nThis specification provides implementations of \"Thresholds\"\nresource for supporting Performance Management interface\nwith external monitoring tools such as Prometheus.\n\nTacker Zed release has supported ETSI NFV-SOL\nbased Fault Management and Performance Management interfaces,\nbut Thresholds resource is not implemented.\n\nThe following items will be supported.\n1. Add support Thresholds resource for PM interface.\n2. Enhance notification to Client.\n3. Add support of RESTful API for communications between\n   Tacker and External Monitoring Tool.\n4. Add new unit and functional tests.\n\nImplements: blueprint support-auto-lcm\nChange-Id: I7bc8af96e3e2a3290891f42821597184d5f77a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/894b2088f3ca4230aa5a90ddd69873c38a9c584f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/894b2088f3ca4230aa5a90ddd69873c38a9c584f"}]},"branch":"refs/heads/master"},"f890f4dc4eef83a869523d28bb979faf709f280f":{"kind":"REWORK","_number":5,"created":"2022-12-22 12:32:34.000000000","uploader":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"ref":"refs/changes/54/868254/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/868254/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/5"}}},"commit":{"parents":[{"commit":"949c4ce423abda4b1c40dfb24699140f36f9dec9","subject":"Add 2023.1 specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/949c4ce423abda4b1c40dfb24699140f36f9dec9"}]}],"author":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 03:25:32.000000000","tz":540},"committer":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-22 12:32:08.000000000","tz":540},"subject":"Support Thresholds resource in PM Interface","message":"Support Thresholds resource in PM Interface\n\nThis specification provides implementations of \"Thresholds\"\nresource for supporting Performance Management interface\nwith external monitoring tools such as Prometheus.\n\nTacker Zed release has supported ETSI NFV-SOL\nbased Fault Management and Performance Management interfaces,\nbut Thresholds resource is not implemented.\n\nThe following items will be supported.\n1. Add support Thresholds resource for PM interface.\n2. Enhance notification to Client.\n3. Add support of RESTful API for communications between\n   Tacker and External Monitoring Tool.\n4. Add new unit and functional tests.\n\nImplements: blueprint support-auto-lcm\nChange-Id: I7bc8af96e3e2a3290891f42821597184d5f77a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f890f4dc4eef83a869523d28bb979faf709f280f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f890f4dc4eef83a869523d28bb979faf709f280f"}]},"branch":"refs/heads/master"},"069fb8b328908c97b91267fcca4b19e02e720304":{"kind":"REWORK","_number":6,"created":"2022-12-23 12:07:24.000000000","uploader":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"ref":"refs/changes/54/868254/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/868254/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/6"}}},"commit":{"parents":[{"commit":"949c4ce423abda4b1c40dfb24699140f36f9dec9","subject":"Add 2023.1 specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/949c4ce423abda4b1c40dfb24699140f36f9dec9"}]}],"author":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 03:25:32.000000000","tz":540},"committer":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-23 12:06:57.000000000","tz":540},"subject":"Support Thresholds resource in PM Interface","message":"Support Thresholds resource in PM Interface\n\nThis specification provides implementations of \"Thresholds\"\nresource for supporting Performance Management interface\nwith external monitoring tools such as Prometheus.\n\nTacker Zed release has supported ETSI NFV-SOL\nbased Fault Management and Performance Management interfaces,\nbut Thresholds resource is not implemented.\n\nThe following items will be supported.\n1. Add support Thresholds resource for PM interface.\n2. Enhance notification to Client.\n3. Add support of RESTful API for communications between\n   Tacker and External Monitoring Tool.\n4. Add new unit and functional tests.\n\nImplements: blueprint support-auto-lcm\nChange-Id: I7bc8af96e3e2a3290891f42821597184d5f77a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/069fb8b328908c97b91267fcca4b19e02e720304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/069fb8b328908c97b91267fcca4b19e02e720304"}]},"branch":"refs/heads/master"},"d2cd60c73061cbd8246759eb1f2a265a66b2bf78":{"kind":"REWORK","_number":7,"created":"2022-12-26 11:11:00.000000000","uploader":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"ref":"refs/changes/54/868254/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/868254/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/7"}}},"commit":{"parents":[{"commit":"949c4ce423abda4b1c40dfb24699140f36f9dec9","subject":"Add 2023.1 specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/949c4ce423abda4b1c40dfb24699140f36f9dec9"}]}],"author":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 03:25:32.000000000","tz":540},"committer":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-26 11:10:38.000000000","tz":540},"subject":"Support Thresholds resource in PM Interface","message":"Support Thresholds resource in PM Interface\n\nThis specification provides implementations of \"Thresholds\"\nresource for supporting Performance Management interface\nwith external monitoring tools such as Prometheus.\n\nTacker Zed release has supported ETSI NFV-SOL\nbased Fault Management and Performance Management interfaces,\nbut Thresholds resource is not implemented.\n\nThe following items will be supported.\n1. Add support Thresholds resource for PM interface.\n2. Enhance notification to Client.\n3. Add support of RESTful API for communications between\n   Tacker and External Monitoring Tool.\n4. Add new unit and functional tests.\n\nImplements: blueprint support-auto-lcm\nChange-Id: I7bc8af96e3e2a3290891f42821597184d5f77a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/d2cd60c73061cbd8246759eb1f2a265a66b2bf78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/d2cd60c73061cbd8246759eb1f2a265a66b2bf78"}]},"branch":"refs/heads/master"},"278ccb8bd096da8e7ebe1a847b1dfe13b8c5ee19":{"kind":"REWORK","_number":8,"created":"2022-12-27 02:32:01.000000000","uploader":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"ref":"refs/changes/54/868254/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/54/868254/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/54/868254/8"}}},"commit":{"parents":[{"commit":"949c4ce423abda4b1c40dfb24699140f36f9dec9","subject":"Add 2023.1 specs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/949c4ce423abda4b1c40dfb24699140f36f9dec9"}]}],"author":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-21 03:25:32.000000000","tz":540},"committer":{"name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","date":"2022-12-27 02:31:38.000000000","tz":540},"subject":"Support Thresholds resource in PM Interface","message":"Support Thresholds resource in PM Interface\n\nThis specification provides implementations of \"Thresholds\"\nresource for supporting Performance Management interface\nwith external monitoring tools such as Prometheus.\n\nTacker Zed release has supported ETSI NFV-SOL\nbased Fault Management and Performance Management interfaces,\nbut Thresholds resource is not implemented.\n\nThe following items will be supported.\n1. Add support Thresholds resource for PM interface.\n2. Enhance notification to Client.\n3. Add support of RESTful API for communications between\n   Tacker and External Monitoring Tool.\n4. Add new unit and functional tests.\n\nImplements: blueprint support-auto-lcm\nChange-Id: I7bc8af96e3e2a3290891f42821597184d5f77a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/278ccb8bd096da8e7ebe1a847b1dfe13b8c5ee19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/278ccb8bd096da8e7ebe1a847b1dfe13b8c5ee19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[]}
