)]}'
{"id":"openstack%2Ftacker-specs~887914","triplet_id":"openstack%2Ftacker-specs~master~I90f8e19da155b6bc7aad400e1d5d3a235034c607","project":"openstack/tacker-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2023-07-20 14:53:49.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2023-07-18 04:46:27.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"35978":{"account":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"last_update":"2023-07-20 15:08:50.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-07-20 15:08:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90f8e19da155b6bc7aad400e1d5d3a235034c607","subject":"Update metadata in setup.cfg","status":"MERGED","created":"2023-07-07 07:26:30.000000000","updated":"2023-07-20 15:08:50.000000000","submitted":"2023-07-20 15:08:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887914","meta_rev_id":"4ba1177136a369f3ce823d84f22754d7f2e7c84c","_number":887914,"virtual_id_number":887914,"owner":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-20 15:08:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2023-07-20 14:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2023-07-18 04:46:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":1,"date":"2023-07-20 14:53:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 07:29:16.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2023-07-07 07:29:16.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2023-07-07 07:29:16.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-07-07 07:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"759ee7f6db1384cad3d98aa5a435b8a723bab5fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-07-07 07:26:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf8d41060e3a1a3bf186481b384ad31438c7fb96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-07-07 07:27:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a854bbfccf8f729b2087b8f9507ed7091d938020","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-07-07 07:29:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34552e9d7e4d447883ba684656d8c2214ebc126d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 07:30:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d7cb04c315c45f5b53746e1a73ab89c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23363ca4c46c4c9394d389290b7a1881 : FAILURE in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d347aa53321045bf9417356b7a3f5f7a : FAILURE in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3e353f755c755c12ad78dacb57ccfaf6a3c9d665","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-07-12 05:42:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b255554f87a4b1a8a066c61f91bf15457a05f2a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 05:52:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc3ea817088f4483be8f3344525d1fd5\n\n- openstack-tacker-tox-docs https://zuul.opendev.org/t/openstack/build/3dc5d120636843cd888d5585e3e6b944 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f1af86b65504189a3a150c3fb7a1102 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"23e39a50ab3b403edaa83ea31024117f9af9b1fe","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-07-18 04:46:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"88687ce7f8649a8eb27ad41f05ff2438a88767f3","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-07-20 14:53:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0929a0c0c2c872e61550aabdfdb5c12466ea164a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:54:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6c1f2aa7dac02745f73cdb7279cc0d9031763a66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 15:08:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc3ee2b5ca3a4d06ae970eb09c9d2654\n\n- openstack-tacker-tox-docs https://zuul.opendev.org/t/openstack/build/ef0b003829e84d00bddbcac984d52ad3 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64fc968f1a4e41ce969e64785794cb63 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"4ba1177136a369f3ce823d84f22754d7f2e7c84c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 15:08:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f6220c87e42ece3d072ee6e247bb56d27b7b0869","revisions":{"1bf6f34112a45c2d2acd32a46fd4e99d196b289a":{"kind":"REWORK","_number":1,"created":"2023-07-07 07:26:30.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/14/887914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/14/887914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/1"}}},"commit":{"parents":[{"commit":"c3db3d4d1af21375bfb5d8f30f4b4a0e6612c5ed","subject":"Fix Spec of \"Enhance Placement Process\" for Bobcat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c3db3d4d1af21375bfb5d8f30f4b4a0e6612c5ed"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-07-07 07:26:00.000000000","tz":0},"committer":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-07-07 07:26:27.000000000","tz":0},"subject":"Update metadata in setup.cfg","message":"Update metadata in setup.cfg\n\nChange-Id: I90f8e19da155b6bc7aad400e1d5d3a235034c607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1bf6f34112a45c2d2acd32a46fd4e99d196b289a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/1bf6f34112a45c2d2acd32a46fd4e99d196b289a"}]},"branch":"refs/heads/master"},"7a5a24d0747977d7caf0d2ee8b2b0f5150fc36d3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-07-07 07:27:35.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/14/887914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/14/887914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/2"}}},"commit":{"parents":[{"commit":"c3db3d4d1af21375bfb5d8f30f4b4a0e6612c5ed","subject":"Fix Spec of \"Enhance Placement Process\" for Bobcat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c3db3d4d1af21375bfb5d8f30f4b4a0e6612c5ed"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-07-07 07:26:00.000000000","tz":0},"committer":{"name":"sairam vengala","email":"sairamvengala007@gmail.com","date":"2023-07-07 07:27:35.000000000","tz":0},"subject":"Update metadata in setup.cfg","message":"Update metadata in setup.cfg\n\nwe are using some \"aliases\" that the setuptools docs say\n\"are supported for compatibility reasons\" but their use is not advised[1].\n\n[1] https://setuptools.pypa.io/en/latest/userguide/declarative_config.html#metadata\n\nplease check this below reference link\n\n[2] https://review.opendev.org/c/openstack/charm-cinder/+/854987\n\nChange-Id: I90f8e19da155b6bc7aad400e1d5d3a235034c607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/7a5a24d0747977d7caf0d2ee8b2b0f5150fc36d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/7a5a24d0747977d7caf0d2ee8b2b0f5150fc36d3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f6220c87e42ece3d072ee6e247bb56d27b7b0869":{"kind":"REWORK","_number":3,"created":"2023-07-12 05:42:41.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/14/887914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/14/887914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/14/887914/3"}}},"commit":{"parents":[{"commit":"c3db3d4d1af21375bfb5d8f30f4b4a0e6612c5ed","subject":"Fix Spec of \"Enhance Placement Process\" for Bobcat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/c3db3d4d1af21375bfb5d8f30f4b4a0e6612c5ed"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-07-07 07:26:00.000000000","tz":0},"committer":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-07-12 05:41:07.000000000","tz":0},"subject":"Update metadata in setup.cfg","message":"Update metadata in setup.cfg\n\nwe are using some \"aliases\" that the setuptools docs say\n\"are supported for compatibility reasons\" but their use is not advised[1].\n\n[1] https://setuptools.pypa.io/en/latest/userguide/declarative_config.html#metadata\n\nplease check this below reference link\n\n[2] https://review.opendev.org/c/openstack/charm-cinder/+/854987\n\nChange-Id: I90f8e19da155b6bc7aad400e1d5d3a235034c607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f6220c87e42ece3d072ee6e247bb56d27b7b0869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f6220c87e42ece3d072ee6e247bb56d27b7b0869"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
