)]}'
{"id":"openstack%2Ftacker-specs~924763","triplet_id":"openstack%2Ftacker-specs~master~I164a55f349d71fe732127675c0851f8f48911095","project":"openstack/tacker-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2024-08-21 17:32:21.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2024-07-30 07:42:43.000000000","reason":"\u003cGERRIT_ACCOUNT_17255\u003e replied on the change","reason_account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}}},"hashtags":[],"change_id":"I164a55f349d71fe732127675c0851f8f48911095","subject":"Performance improvement of list APIs","status":"MERGED","created":"2024-07-23 16:58:54.000000000","updated":"2024-08-21 17:40:44.000000000","submitted":"2024-08-21 17:39:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"924763","meta_rev_id":"3cfb53d1e661b80e394b59d3533053b79457281f","_number":924763,"virtual_id_number":924763,"owner":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-21 17:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2024-08-21 17:32:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2024-08-20 08:14:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":1,"date":"2024-08-21 17:32:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 17:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 17:04:11.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"}],"messages":[{"id":"fd562ca296275b765e8449d0677c7ee38b88ba13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2024-07-23 16:58:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cca837aef0bfb216266d2abdf646f75b50a701ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:03:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de92633511e04de7913a56a753dd0f54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf98d23579042e5bd6450b9c5493942 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783ff2d6fac04c8bafd0e63eca89b03d : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f7718ed3b61335dc12c76358b90e4cb0e6aba7d9","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2024-07-30 07:42:43.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e3cf748e2f01e09448763cd9a304156fd9b8c5ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2024-08-08 05:33:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"92c1f11c84072f95c58873e776bfcd6f8c8c0529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 05:37:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb7bf27c2a5a4224b7265eeadeca43cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46d754b8c49438da81983a12668e87f : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e32d94a213443781fd82aba2e06aea : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":2},{"id":"575fc9735c1b6384b047d51ac749cc123dd499c5","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2024-08-20 08:14:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a326e17781428f94a4b0af05d293f775b675e331","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2024-08-21 17:32:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b72d59fa58f9fbb6b2fb89cb9bc1d12394ec9c2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 17:32:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"51bcb23e25b45b1c9859c5bbadbed339e0a189da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 17:39:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/715810979d4748bb95a894836f20b9a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c15a32ad4a404fb1495698efa00e04 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d5cf1db53345329e3785d346ba3a3f : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"7064a31fe0d12f9c013e5543d62ef9e6d04d86fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 17:39:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3cfb53d1e661b80e394b59d3533053b79457281f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 17:40:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/534679c55d014fc4b52680a6f0dcc6cf\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/a2bebcb9da4f477484f5a87a9b4fd735 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f43cc9ea9e57834268d05b5d260d994aef86c2ca","revisions":{"88ad337ad1bb42427962c5e2b564c1d80f7394b5":{"kind":"REWORK","_number":1,"created":"2024-07-23 16:58:54.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/63/924763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/63/924763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/63/924763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/63/924763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/63/924763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/63/924763/1"}}},"commit":{"parents":[{"commit":"f9af3dcfa0bea3c6c9bbcf0a52fcefb07389178d","subject":"OpenTelemetry for monitoring VNF and CNF","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f9af3dcfa0bea3c6c9bbcf0a52fcefb07389178d"}]}],"author":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2024-07-23 16:37:34.000000000","tz":0},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2024-07-23 16:58:32.000000000","tz":0},"subject":"Performance improvement of list APIs","message":"Performance improvement of list APIs\n\nThis proposal is for a performance improvement of list APIs of Tacker\nfor considering actual usecases in which a lot of entries, for\nexample thousands or more. From our simple evaluation, it takes 10\nseconds or more for the usecases and it must be too long for users.\nSo, we\u0027ll reduce the time of querying, especially for typical\nusecases.\n\nImplements: blueprint perf-improve-list-apis\nChange-Id: I164a55f349d71fe732127675c0851f8f48911095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/88ad337ad1bb42427962c5e2b564c1d80f7394b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/88ad337ad1bb42427962c5e2b564c1d80f7394b5"}]},"branch":"refs/heads/master"},"f43cc9ea9e57834268d05b5d260d994aef86c2ca":{"kind":"REWORK","_number":2,"created":"2024-08-08 05:33:09.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/63/924763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-specs","ref":"refs/changes/63/924763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/63/924763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/63/924763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-specs refs/changes/63/924763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-specs refs/changes/63/924763/2"}}},"commit":{"parents":[{"commit":"f9af3dcfa0bea3c6c9bbcf0a52fcefb07389178d","subject":"OpenTelemetry for monitoring VNF and CNF","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f9af3dcfa0bea3c6c9bbcf0a52fcefb07389178d"}]}],"author":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2024-07-23 16:37:34.000000000","tz":0},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2024-08-08 05:32:56.000000000","tz":0},"subject":"Performance improvement of list APIs","message":"Performance improvement of list APIs\n\nThis proposal is for a performance improvement of list APIs of Tacker\nfor considering actual usecases in which a lot of entries, for\nexample thousands or more. From our simple evaluation, it takes 10\nseconds or more for the usecases and it must be too long for users.\nSo, we\u0027ll reduce the time of querying, especially for typical\nusecases.\n\nImplements: blueprint perf-improve-list-apis\nChange-Id: I164a55f349d71fe732127675c0851f8f48911095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f43cc9ea9e57834268d05b5d260d994aef86c2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-specs/commit/f43cc9ea9e57834268d05b5d260d994aef86c2ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
