)]}'
{"id":"openstack%2Ftacker~1004165","triplet_id":"openstack%2Ftacker~master~Id399ba4df7f6acaf8ba4318fef997f3e503a41d1","project":"openstack/tacker","branch":"master","topic":"support-tca-lcm-operator","attention_set":{"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2026-09-10 12:02:14.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"13192":{"account":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"last_update":"2026-09-10 17:48:59.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"36672":{"account":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},"last_update":"2026-09-23 17:01:04.000000000","reason":"\u003cGERRIT_ACCOUNT_36672\u003e replied on the change","reason_account":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2026-09-18 07:45:37.000000000","reason":"\u003cGERRIT_ACCOUNT_17255\u003e replied on the change","reason_account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}}},"hashtags":[],"change_id":"Id399ba4df7f6acaf8ba4318fef997f3e503a41d1","subject":"Support Kubernetes TCA LCM Operator","status":"NEW","created":"2026-09-05 04:42:53.000000000","updated":"2026-09-23 17:01:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":13,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"ffd6feb845d76c7aa39092f7484d14c69fc7eaef","_number":1004165,"virtual_id_number":1004165,"owner":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-14 09:17:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},"all":[{"value":1,"date":"2026-09-23 17:01:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 04:43:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-06 03:11:28.000000000","updated_by":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2026-09-06 03:11:28.000000000","updated_by":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2026-09-18 07:45:37.000000000","updated_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"reviewer":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},"state":"REVIEWER"}],"messages":[{"id":"e45dcfbeb8a03fe15981af808f8034842a978349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-05 04:42:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d53f5caf092e595399a02e3cd86bd1524ec05d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 04:43:31.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/tacker on branch master.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/tacker/zuul.d/jobs/kubernetes.yaml@master\", line 1, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/tacker/zuul.d/jobs/kubernetes.yaml@master\", line 79, column 2","accounts_in_message":[],"_revision_number":1},{"id":"33f90180fdb754d441ec596855a18a652bd6667d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-05 04:48:32.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e8e098d58fdf7631088fba409c2d5bfd2da2277f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 06:52:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/159e9eabcaea47d4b0887b9bbe7cd1b6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/921dfe1d9a4744ac8c51e753a1733c68 : SUCCESS in 4m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/acc04410bbd642d582602375549747e8 : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5faae53d826949b3a4ce61cfbb0e6bb7 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb824911e574a87ae04dddf93ec9641 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2b76cfdb76541e293f053584ab5f1d3 : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7b48764ecdd4802920982272b227395 : SUCCESS in 8m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/41006b1a378d49cb8b1dafa74d1bc95f : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99253bd1f87e43c7babf7197c5b74965 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb131e7d4abd47e9b580fa932e63e329 : SUCCESS in 3m 05s\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/a85e914eafcd404aa29dfb798b1e5b77 : RETRY_LIMIT in 51m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bb845871edd82ac35aaae361eb782ffaa5ca5d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-05 07:53:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"eec4e700397a45484d49bea0a9d283801f6ee6d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 10:41:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72bcc41f9925427ea39eb626c3a6247c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e1249e2b42744e7ebb2574a0f54fbe35 : SUCCESS in 2m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bd51db48b4ff4c8f9e3b985c0277655d : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da03ce67f38243a9b0f197917c522997 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f88eb49a625431d9846567bcf741162 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f48d2c03ab14a0e91190982f3c3d041 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb4575f07337427c90264887acb10050 : SUCCESS in 8m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bdaef160d7234abcb89438eef7e5413d : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9d984114484b788889f75626b16881 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b56673c51ccf4857a047f29eff9ea383 : SUCCESS in 5m 49s\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/48e2ed2252e74a52a0d374916afc69ab : RETRY_LIMIT in 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fcb4a835cc8863febd63ed975467dd052217a84d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-05 11:11:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ae4485dbd7890b5570e7326ecc0c5c0909e4731f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 11:45:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e78a6d24180a4581a333b33c03d5a8c3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c8f5d256ffe74ffbbd08790b0efb3dd6 : SUCCESS in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0a37a53a08904be896ac0fc0f50377c0 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daf4dc3c7a6d45899d5cc4d2b5a6ce22 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/884c11af16fd457a9f08c457ec668ce0 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa7bc11208854acea3a0b97b3d9f0b9a : SUCCESS in 9m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2aface82bd074aafbdb13cd7e94ab7f4 : SUCCESS in 9m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/53f8ffe141d449008c00cd7877152820 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f56c7ca6195e440d86db7aebc82aa0d9 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99d4a6671a234fe4a3be96c14b78a4a5 : SUCCESS in 3m 25s\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/a1d7baead6ef4ec9b845ec0cda232c47 : RETRY_LIMIT in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e0138d33fa21a907ca6bac4f69f3ec5a11795225","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-05 11:56:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"3e04f33e4da66743bf8e9ac6c1ef56b67b635e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 12:46:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b53d42ee924345f4a05da132fe865610\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e52d6223d68149acb10ab1e27f3b4953 : SUCCESS in 4m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d575870aa22946818d5b712461c287d3 : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/921bb69ea11e4228960dd7dd8f071502 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153a234c97db4f139ce281c925f75844 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b1dd0b0d14d4ae682d905897231eecc : SUCCESS in 8m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c147e7ed65564603be867cbc98fbf0a5 : SUCCESS in 8m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f72d7ef31da4414ea6614e0cdde1f60b : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ab6eb2078841d58224ee1ef450b610 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5ba54b43ad4482dbfe39a4f707848aa : SUCCESS in 6m 14s\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/553bce63542b463aa9fc71e76af5f09e : RETRY_LIMIT in 42m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5e0ba0266b9fd3cb7d8253e847dcde63eea38d01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-05 13:08:22.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"53ff15aa79ccf6e2187ca5d39bf7d009474ecda6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 14:31:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/235b05ab0f314deda9fe3cdbaf11b9f9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ea451315658c4a8e8bd3eb78207109dd : SUCCESS in 5m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/437466961c044446a7e332162afef642 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/807d80cbe4564879970a8bd9e43fd319 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc6f0f2401a4d4dbeceff2a399e1b8d : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fd9aa5d8b1c46769d11adeebff12826 : SUCCESS in 7m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38fc3716f0164a70bf9b1bb2420dea84 : SUCCESS in 8m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4b3ad58b17ac405eb5db2e22383fc492 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86727cfe49a04ff3a442eca89e02fb3c : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6fd3dbfad414a2780ea34eacdfa8afa : SUCCESS in 6m 37s\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/99520e95fa8c43bba1f9b2e1b5bfe8d4 : SUCCESS in 1h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a80ecc8f2e775e501119bbb8710982366ef8dc62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-05 16:44:36.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"53c98ce243020d1e7edfd7ff739e526b3836fffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 16:45:09.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/tacker on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027fetch_subunit_output_additional_dirs\u0027]\n\nThe problem appears in the \"tacker-ft-v2-k8s-tcalcm-operator\" job stanza:\n\n  job:\n      name: tacker-ft-v2-k8s-tcalcm-operator\n      parent: tacker-ft-v2-k8s\n      description: |\n        Experimental V2 Kubernetes functional tests with TCA LCM Operator\n      vars:\n        setup_tcalcm_operator: true\n      fetch_subunit_output_additional_dirs:\n        - \"{{ ansible_user_dir }}/{{ zuul_work_dir }}/operator/tca.lcm/common\"\n  ...\n\n  in \"openstack/tacker/zuul.d/jobs/kubernetes.yaml@master\", line 79","accounts_in_message":[],"_revision_number":7},{"id":"36716ac5e548408d7561bcc4f7b04086147f6092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-05 16:55:08.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"9732c9c121c43a0e964679862d1f87b874c8f261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 18:08:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05a83e0ffe524f7bb8436a83644d1e79\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1513cbbe0c494f3389b46465d03786f3 : SUCCESS in 4m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b67581997254fbab1ec2ff6d3cf59df : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/394c7939ea7c42f2962f13795003adf7 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f651e97254204bf0b910eb046c74503f : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc08c413350b4a0595ac85da9b50caa8 : SUCCESS in 9m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9380337f5d0d4e93b40088b0c23acbb0 : SUCCESS in 5m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/23dc92edd118491e9f95f58c22c5b801 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8bf89f52c6c44639c6de3d6322ab605 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10f31fa15ffe436087d0a5874d72bc49 : SUCCESS in 5m 38s\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/6f7b1a4c0de14d62a9595798fd91ac3c : POST_FAILURE in 59m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3d3388089d20baf7a19e870f33c504198d8c3da7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-05 19:07:19.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"b69ac1ff1302d273b5891718ea76aa85c9d6b8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 20:02:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a718bf909e14159b2c7241c388ed9a8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c693febccb5b469aa8254c068b7a465c : SUCCESS in 3m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d5a0e1f562974035bbffc46236bbe681 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c8b38222bcc4ebf9d1f3811db524980 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f252ebe2a81f4495a3bc253bbe22686b : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4891356214174a1387704f8ff23ff121 : SUCCESS in 6m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/22703259387a4a119476724ff852a7d7 : SUCCESS in 8m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c2614bb9b0e143189cdaf746cd87dc25 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b072b0538a44acb49e0d9b963eab39 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eee27fd3e77247a99dcdc3a07c2be80a : SUCCESS in 3m 39s\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/cf8fe30ec7f245b09c35e7b4ceeacaaf : SUCCESS in 53m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2999503e3fbb62bddbc422030597db0156968fb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-06 03:10:43.000000000","message":"Patch Set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"047fb81c1866554594b08226155c74a3c196bcd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 04:08:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c8d1e8ecd44f8a835fe4d8db141cea\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/75a3e17c82c340b4aeabecf5735dc43f : SUCCESS in 7m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6f6ed81a44044b98ca9728068ac089e : SUCCESS in 2m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c15e4892584e4d599da7a6a2651dd7ab : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ffc50dc9dc74757b1757abb9ac72d67 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3efce206bd8240c9a56e24232b32dfb6 : SUCCESS in 7m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bc6d3192dc34ff5b51a2ccf058b1867 : SUCCESS in 5m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6694c08d3fc40f98fa9b43a0be67f79 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae5f68e80f548f2bfcc5c2eb9e45d6b : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbd10bad0f1f469995575724e36c523b : SUCCESS in 4m 43s\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/36037dd612ed49079a341ded73ef8049 : SUCCESS in 55m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cba3177a0efbe1a9372f7c9c348e3d264081bdb1","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-09-10 04:45:28.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1567ef9ad34480584c88ce975cf76543b406167d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-10 12:02:14.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"09510a51fba6869caf24c6141d4555f687a02f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:26:23.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23e2c66e219b498eb25f2a27e367f30d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2bd04d50b2fd4b2cbc55e6f3c56ee036 : SUCCESS in 4m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/480aac8a827649068a9299d3d441402f : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2161f34a629e4ad1bf988872da2335e2 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd67cc0d88d49f49103f0802f5f66a3 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e94c323cbc14c488c678113afdc517b : SUCCESS in 9m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2a70a9f970f42c38fda12e7bd7a5d7b : SUCCESS in 8m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee27defcf1eb489ea5484e52be12e9eb : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e126f7785b74cfb83e3e3467a02c45c : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cae903f8b094aad81d24d6dcea92ca4 : SUCCESS in 6m 02s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/667dedc752564f79ab343e14ed0fa642 : RETRY_LIMIT in 54m 55s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/c88b4d25f4af4c5a94e0e46040b2b034 : SUCCESS in 56m 54s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/32bd9cf42f5d4c538d34e7024475c473 : SUCCESS in 55m 56s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/4b98092951d6405994d289d485fa8be1 : SUCCESS in 51m 12s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/856b46247dab46dd9a84127279bd07b7 : SUCCESS in 54m 30s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/c28d5f59051b41659e60e5f467790efa : SUCCESS in 1h 02m 21s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/47fe1332158b476e981f39fbce4aa229 : SUCCESS in 36m 41s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/30b2f279abed415b9c4113faafb52953 : SUCCESS in 49m 54s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/39238c07119249dbbcd8a8170f939ba1 : SUCCESS in 34m 51s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/4a1a2ba6a6114d2eb999d6ed6e0e757d : SUCCESS in 1h 06m 06s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/1494df6a37ff4e8d83e3b5e6df49c4b4 : SUCCESS in 52m 17s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/5081e67330ea4ce89cacc446372c44c8 : RETRY_LIMIT in 8m 39s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/6a8472b7eadd415c82e8eb252abf23ca : SUCCESS in 1h 01m 21s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/c446ba50e86e4df29da435720a96719a : SUCCESS in 1h 10m 38s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/2ca61657217e419c81cbb04c569b231b : FAILURE in 2h 12m 59s (non-voting)\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/d6afd45a98684b379c0591fd84dbde01 : RETRY_LIMIT in 53m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6de4ed5a728b33570275ddadae62daa4c175724e","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-10 14:53:28.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"359e729148e21ce01e508bcec0c36473a69cf341","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-10 14:55:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"919770b83b83cfc3fbc45894f2247a8dd362d031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 17:48:59.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e024df672a2a4d508eca75a51881a612\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/080d12db56864ab6a3567926a9a88677 : SUCCESS in 5m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4edee70f172a4569b9d5aeff0847cbbf : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e23fa31e32d469087b1b3ece28186a7 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebcf46a798294c47acef70850e591432 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/162bc5522c774c099b116232ad4005c5 : SUCCESS in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ffea35eb12b14c3789725b1bbbba13ae : SUCCESS in 8m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f19ef14a6e914ff18bfca77577a48dfe : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b0bf5787db408bba8da67d7fdef285 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b94cbfb55e6b4cc9bf00d622e4781e3e : SUCCESS in 6m 21s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/958d540c23de4cc6b4a9040db5f21008 : RETRY_LIMIT in 50m 28s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/af259018458c427586adc93258db7391 : SUCCESS in 57m 18s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/8a536b157bac424e81d09b66fa05e6ce : SUCCESS in 34m 40s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/9a82e450a0e4449ab04d5513b2e71a8b : SUCCESS in 1h 08m 59s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/bc95056e4e7646c68a9a27f990cd36da : SUCCESS in 59m 46s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/8e01867553644f829c52311c2eb23362 : SUCCESS in 44m 22s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/4acff03238504235b458e1028286caaf : SUCCESS in 1h 06m 03s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/4c028808635b45b3a9f6ee0ef0cd17fe : SUCCESS in 55m 09s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/3c306e6b7baf4ed3adfcc2dfdd877b2f : SUCCESS in 1h 01m 28s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/2d4399c067834948ae3bc5dad1d189c3 : SUCCESS in 33m 03s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/ccc23981dd5c4235b395ef140dd8f015 : SUCCESS in 1h 08m 00s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/b5bae3cdc6ee4259b0faa302dd30cc1e : RETRY_LIMIT in 9m 38s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/afdff44cf9844c6c8f89e6d08f3e0b33 : SUCCESS in 32m 09s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/cf026aae8781420f9c4b49b9a45d10ef : SUCCESS in 1h 13m 30s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/653b093dc923425ea1dc04631ed14bcb : FAILURE in 2h 15m 22s (non-voting)\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/e1046a72e0de40ee85d41e0cb94542e9 : RETRY_LIMIT in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"04b559fbbf02ca4345e5fc5b65472eed97336092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-14 06:47:01.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"195181df75346e9e6545072a8f02887d5d0f1708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 06:47:32.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/tacker on branch master.  The\nproblem was:\n\n  Job openstack-tox-base not defined\n\nThe problem appears in the \"tacker-tox-operator-common\" job stanza:\n\n  job:\n      name: tacker-tox-operator-common\n      parent: openstack-tox-base\n      description: |\n        Run the TCA LCM Common Operator\u0027s own unit test suite via its\n        isolated tox environment (operator-common), which has its own\n        requirements.txt/test-requirements.txt independent of tacker\u0027s.\n      vars:\n        tox_envlist: operator-common\n  ...\n\n  in \"openstack/tacker/zuul.d/jobs/kubernetes.yaml@master\", line 88","accounts_in_message":[],"_revision_number":12},{"id":"365246cd013ec55016047f462994e482188ac414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2026-09-14 06:54:12.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"cf593b7be94387288fe5ca4061af31542a0f148e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 09:17:14.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db354a0fb06e4609bced55de2bf76b89\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1684707090d947688033f14fa7a8a4fc : SUCCESS in 4m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/83001b4ce53742c681c3f349ab0549f6 : SUCCESS in 2m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e6dd357178f418a80c959d4294f5298 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db14316a46d413c9bcb4e00552a3d1a : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c992eb6af0a462c9da69181c48309c2 : SUCCESS in 8m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4cebd0a16e647a7a5dad24b9a78219d : SUCCESS in 5m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/271331d470114d219646210d402b12b5 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbbc64700a29416eaf0b2258dc4f1851 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a338da450d3d49f0ab55ea1a409a725b : SUCCESS in 3m 05s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/0205a0a7b7c64e40a81b2424400a8a13 : SUCCESS in 1h 17m 50s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/bc63eb303bd84f8d96c9e1148b662996 : SUCCESS in 53m 23s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/b63a8941fd3f4ec5a5431a5f75f2e8b7 : SUCCESS in 43m 33s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/43d311c3650c446cb7c56b951ad887c0 : SUCCESS in 1h 03m 57s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/792f8920600d4ac3a9bc9e6f9a7e96c9 : SUCCESS in 31m 21s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/0764bfecaed5480894d45ded171e6db5 : SUCCESS in 39m 16s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/5954461afc0a491b827feb874c32ae74 : SUCCESS in 1h 05m 05s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/62f0b4cbfb8c438f9849b8a782e74d8e : SUCCESS in 28m 37s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/47b6dff2716e49729036931bcb156755 : SUCCESS in 59m 50s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/ee3e434f3d3746c38e8681c5cb587fb9 : SUCCESS in 26m 39s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/902809f11c1a48179e4151443d5b8778 : SUCCESS in 37m 32s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/30fc579e13d541888fa81a098b31f1fc : RETRY_LIMIT in 7m 41s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/4548290669724455aead18535478313e : SUCCESS in 56m 44s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/834eff901f4a4705b086ba1237b19f15 : SUCCESS in 49m 06s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/f349c28094294b6eb7e37b64579bdb8c : FAILURE in 2h 16m 38s (non-voting)\n- tacker-ft-v2-k8s-tcalcm-operator https://zuul.opendev.org/t/openstack/build/0ce101399c904c1eb040c54c66b217a5 : SUCCESS in 1h 02m 47s (non-voting)\n- tacker-tox-operator-common https://zuul.opendev.org/t/openstack/build/ad28298c6c024e4b86df8322c6758963 : SUCCESS in 2m 41s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ffd6feb845d76c7aa39092f7484d14c69fc7eaef","author":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},"date":"2026-09-23 17:01:04.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"7311bf6e378012dd98c530f395e3c9fc4195168a","revisions":{"db8f26dcfa3a28fb4e8ec2d0696609ce7769477f":{"kind":"REWORK","_number":1,"created":"2026-09-05 04:42:53.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/1"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"subject":"[WIP] Support Kubernetes TCA LCM Operator","message":"[WIP] Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/db8f26dcfa3a28fb4e8ec2d0696609ce7769477f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/db8f26dcfa3a28fb4e8ec2d0696609ce7769477f"}]},"branch":"refs/heads/master"},"a0d3e4d6a1ba2b2bfa8db9ab2c115dcd4e984e8e":{"kind":"REWORK","_number":2,"created":"2026-09-05 04:48:32.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/2"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:48:32.000000000","tz":0},"subject":"[WIP] Support Kubernetes TCA LCM Operator","message":"[WIP] Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/a0d3e4d6a1ba2b2bfa8db9ab2c115dcd4e984e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/a0d3e4d6a1ba2b2bfa8db9ab2c115dcd4e984e8e"}]},"branch":"refs/heads/master"},"1a21743d830833b086d8d4fc353b88222706447f":{"kind":"REWORK","_number":3,"created":"2026-09-05 07:53:04.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/3"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 07:52:29.000000000","tz":120},"subject":"[WIP] Support Kubernetes TCA LCM Operator","message":"[WIP] Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1a21743d830833b086d8d4fc353b88222706447f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1a21743d830833b086d8d4fc353b88222706447f"}]},"branch":"refs/heads/master"},"34b8d6c8836f6fa82d478fc37b89195c6c081f1c":{"kind":"REWORK","_number":4,"created":"2026-09-05 11:11:26.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/4"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 11:10:42.000000000","tz":120},"subject":"[WIP] Support Kubernetes TCA LCM Operator","message":"[WIP] Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/34b8d6c8836f6fa82d478fc37b89195c6c081f1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/34b8d6c8836f6fa82d478fc37b89195c6c081f1c"}]},"branch":"refs/heads/master"},"e1f2fd467ad50fcc7549328d21d7a9f702ae0652":{"kind":"REWORK","_number":5,"created":"2026-09-05 11:56:34.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/5"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 11:56:32.000000000","tz":120},"subject":"[WIP] Support Kubernetes TCA LCM Operator","message":"[WIP] Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e1f2fd467ad50fcc7549328d21d7a9f702ae0652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e1f2fd467ad50fcc7549328d21d7a9f702ae0652"}]},"branch":"refs/heads/master"},"31684c549bad8c40a63ec38651a22061ecabfcf2":{"kind":"REWORK","_number":6,"created":"2026-09-05 13:08:22.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/6"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 13:08:20.000000000","tz":120},"subject":"[WIP] Support Kubernetes TCA LCM Operator","message":"[WIP] Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/31684c549bad8c40a63ec38651a22061ecabfcf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/31684c549bad8c40a63ec38651a22061ecabfcf2"}]},"branch":"refs/heads/master"},"9b4119c59da4dcc9e4fd03a7fb070aa3e9aa501f":{"kind":"REWORK","_number":7,"created":"2026-09-05 16:44:36.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/7"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 16:44:35.000000000","tz":120},"subject":"[WIP] Support Kubernetes TCA LCM Operator","message":"[WIP] Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/9b4119c59da4dcc9e4fd03a7fb070aa3e9aa501f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/9b4119c59da4dcc9e4fd03a7fb070aa3e9aa501f"}]},"branch":"refs/heads/master"},"4c53d1beb5142da8bc13448b7b1225ee6a5b789d":{"kind":"REWORK","_number":8,"created":"2026-09-05 16:55:08.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/8"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 16:55:07.000000000","tz":0},"subject":"[WIP] Support Kubernetes TCA LCM Operator","message":"[WIP] Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4c53d1beb5142da8bc13448b7b1225ee6a5b789d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4c53d1beb5142da8bc13448b7b1225ee6a5b789d"}]},"branch":"refs/heads/master"},"e3c9e4c5e9890a01cda677222fee383267c4ab29":{"kind":"REWORK","_number":9,"created":"2026-09-05 19:07:19.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/9"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 19:07:19.000000000","tz":0},"subject":"[WIP] Support Kubernetes TCA LCM Operator","message":"[WIP] Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e3c9e4c5e9890a01cda677222fee383267c4ab29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e3c9e4c5e9890a01cda677222fee383267c4ab29"}]},"branch":"refs/heads/master"},"313df434228a5a8dd149ba74dd8e6e308bb85a75":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-09-06 03:10:43.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/10"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Souville","email":"souville@docomolab-euro.com","date":"2026-09-06 03:10:43.000000000","tz":0},"subject":"Support Kubernetes TCA LCM Operator","message":"Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/313df434228a5a8dd149ba74dd8e6e308bb85a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/313df434228a5a8dd149ba74dd8e6e308bb85a75"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d47d7d97f228a80a68533444d38700b38c08d7f9":{"kind":"REWORK","_number":11,"created":"2026-09-10 12:02:14.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/11"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-10 12:02:10.000000000","tz":120},"subject":"Support Kubernetes TCA LCM Operator","message":"Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/d47d7d97f228a80a68533444d38700b38c08d7f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/d47d7d97f228a80a68533444d38700b38c08d7f9"}]},"branch":"refs/heads/master"},"5c017888a8540ff94f6c65d10b6ddffaac72cb9c":{"kind":"REWORK","_number":12,"created":"2026-09-14 06:47:01.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/12"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-14 06:45:47.000000000","tz":120},"subject":"Support Kubernetes TCA LCM Operator","message":"Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/5c017888a8540ff94f6c65d10b6ddffaac72cb9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/5c017888a8540ff94f6c65d10b6ddffaac72cb9c"}]},"branch":"refs/heads/master"},"7311bf6e378012dd98c530f395e3c9fc4195168a":{"kind":"REWORK","_number":13,"created":"2026-09-14 06:54:12.000000000","uploader":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"ref":"refs/changes/65/1004165/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/1004165/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/1004165/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/1004165/13"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-05 04:15:37.000000000","tz":120},"committer":{"name":"Bertrand Souville","email":"souville@docomolab-euro.com","date":"2026-09-14 06:52:03.000000000","tz":120},"subject":"Support Kubernetes TCA LCM Operator","message":"Support Kubernetes TCA LCM Operator\n\nThis patch deploys experimental version of TCA LCM Operator\nwith several unit and functional tests.\n\nThe TCA LCM Operator is a generic, reusable reconciliation\nengine that orchestrate dependency-ordered Helm rollouts natively\non Kubernetes.\n\nA release note describing the new operator and its CRD is included.\n\nAssisted-By: Claude Sonnet 5\n\nChange-Id: Id399ba4df7f6acaf8ba4318fef997f3e503a41d1\nSigned-off-by: Bertrand Souville \u003csouville@docomolab-euro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/7311bf6e378012dd98c530f395e3c9fc4195168a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/7311bf6e378012dd98c530f395e3c9fc4195168a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
