)]}'
{"id":"openstack%2Ftacker~1004229","triplet_id":"openstack%2Ftacker~master~I0d0b403c7c7dbee79fb46623a47b89d5758f4ac2","project":"openstack/tacker","branch":"master","hashtags":[],"change_id":"I0d0b403c7c7dbee79fb46623a47b89d5758f4ac2","subject":"tox: Drop redundant passenv","status":"NEW","created":"2026-09-05 18:18:56.000000000","updated":"2026-09-05 20:12:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"63d80010df2eb8f33aa8b0e73d495aff92af4b94","_number":1004229,"virtual_id_number":1004229,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-05 20:12:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 20:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d63449cf5ff3828cef23d030680969ce43a8405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:18:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63d80010df2eb8f33aa8b0e73d495aff92af4b94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 20:12:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52bce3394c3f4bdd8467c01b0428a2b0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/259b9c25018f4519bfb7e9a16cf293db : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e7a1f3614bc4035bdf759f50f8642ac : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cae1cc07b43463d9aa876ed916684ef : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9491f7583024afeb495c3b08a338f11 : SUCCESS in 7m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05c7ffe0abbf4c6988601593c9267926 : SUCCESS in 7m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7216c048065b49aaa71fc899b1e5e1c9 : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5384ad542f6d4d4bb0e947e544c64012 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b21c7eada5841c38c284e076b8aa269 : SUCCESS in 2m 59s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/59c707c5888f4b9fb2c73625c3422a54 : SUCCESS in 1h 03m 29s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/1a63d565a9384f87bd3eb1b5e73790ac : SUCCESS in 53m 14s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/7d28378d617c41858aad767170bf48d6 : SUCCESS in 1h 01m 17s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/a8decbc99b2947f8be1cd539e857b011 : SUCCESS in 32m 20s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/1402270cf0e541dd80767c62d930abc5 : SUCCESS in 59m 59s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/0d96264bb4714c179751dfd7229158d3 : SUCCESS in 1h 03m 37s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/6a2f95e4008a403488f98c5d31c5cab4 : SUCCESS in 1h 05m 24s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/1f01df8c3ec7414a89ee914a901f91f9 : SUCCESS in 43m 25s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/b14159f4ba534700acf7ebaeec34554a : SUCCESS in 1h 02m 12s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/429230dc1fc24430b0b394841231a09f : FAILURE in 53m 00s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/5ed13440344444dc90011303615f9a5c : SUCCESS in 51m 18s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/7c882e2f1dd94ee8923bf8032f33ff0d : RETRY_LIMIT in 17m 03s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/1868bc26c2c741dca0f1e9356fb32d0d : SUCCESS in 53m 00s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/2618e18efda342b49ec3544c2367a9af : SUCCESS in 1h 18m 55s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/0a87c9b35db74ab0a4cabfe1ebcebf6a : FAILURE in 1h 45m 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8efd5b937adec0791adcf60847960ca222fb116","revisions":{"c8efd5b937adec0791adcf60847960ca222fb116":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:18:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/1004229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/29/1004229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/29/1004229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/29/1004229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/29/1004229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/29/1004229/1"}}},"commit":{"parents":[{"commit":"4a7cace19e06671aebed85f593ced618d5963a9e","subject":"Merge \"Stop useless passing preexec_fn to Popen\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4a7cace19e06671aebed85f593ced618d5963a9e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:18:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:18:46.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I0d0b403c7c7dbee79fb46623a47b89d5758f4ac2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/c8efd5b937adec0791adcf60847960ca222fb116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/c8efd5b937adec0791adcf60847960ca222fb116"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
