)]}'
{"id":"openstack%2Ftacker~1004230","triplet_id":"openstack%2Ftacker~master~I883d30ed1a48cd14ab6bd66e59a61d3ffcd05cd1","project":"openstack/tacker","branch":"master","hashtags":[],"change_id":"I883d30ed1a48cd14ab6bd66e59a61d3ffcd05cd1","subject":"Fix outdated default envlist","status":"NEW","created":"2026-09-05 18:19:56.000000000","updated":"2026-09-05 20:47:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"33e00c3496853d52f26be2845544606e074cb560","_number":1004230,"virtual_id_number":1004230,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-05 20:47:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 20:47:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33d7c61aa86f71ea59d5073588eeeb49234ce2b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:19:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33e00c3496853d52f26be2845544606e074cb560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 20:47:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/392968fe951c406f92962e3a948a39e0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/deee52075b23452885342b27ec2243cd : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43514072148a40ca96ebad65883752d9 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b781306097ba441589b255ad6370b28c : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea132a168b704e0ba3ab62a1baf708b4 : SUCCESS in 9m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f109fa06233f4c90b7c47c87e338e9e8 : SUCCESS in 6m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1645f25403c6414bba4a0b6a210df06d : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60c0a0d158041dd999e81e1ed3daa45 : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7faa291c80f54236981de01a0b3273f8 : SUCCESS in 6m 08s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/cc300b88abaa4979b1b756a9ae9b00b0 : SUCCESS in 51m 39s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/8f0f3013b68b485cb297f402ea2944fa : SUCCESS in 54m 30s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/498622b5740540558bd6bbdd9b36e060 : SUCCESS in 59m 25s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/63e47c2f26474e08a131c0847d0a6913 : SUCCESS in 1h 05m 14s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/db933b14621d4a14b1cf34dc2fdca12f : SUCCESS in 57m 04s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/37cb842479cb44e99a72bdb7d89ab213 : SUCCESS in 44m 39s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/3cdd91e87bdd471fb6fa843f67a425c6 : SUCCESS in 1h 01m 50s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/5eee95f48f784bd0834a16b220c8f6bd : SUCCESS in 56m 42s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/cc43a486218245c0889881edb337db8e : SUCCESS in 1h 03m 19s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/22f0d0f1cf8e4a399ed03867d02fda32 : FAILURE in 49m 05s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/d14878dc2a97488d9eece86e0d2e369b : SUCCESS in 1h 07m 01s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/039bc18ebd184ec7b8a2b12deb2ee011 : RETRY_LIMIT in 16m 45s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/146f29b80a80498182d9e6dc5e057f57 : SUCCESS in 54m 12s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/6535dd6f255a479c821b9d20cbac2129 : SUCCESS in 1h 16m 07s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/462c6728b89941edbbaf6b12da91542c : FAILURE in 2h 16m 06s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ade09a62c203a5d225f95180c0855452348564f5","revisions":{"ade09a62c203a5d225f95180c0855452348564f5":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:19:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/1004230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/30/1004230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/30/1004230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/30/1004230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/30/1004230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/30/1004230/1"}}},"commit":{"parents":[{"commit":"4a7cace19e06671aebed85f593ced618d5963a9e","subject":"Merge \"Stop useless passing preexec_fn to Popen\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4a7cace19e06671aebed85f593ced618d5963a9e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:19:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:19:48.000000000","tz":540},"subject":"Fix outdated default envlist","message":"Fix outdated default envlist\n\nSome of the versions described there are no longer supported. Use\nthe unversioned py3 env by default so that we don\u0027t have to update it\nregularly.\n\nChange-Id: I883d30ed1a48cd14ab6bd66e59a61d3ffcd05cd1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ade09a62c203a5d225f95180c0855452348564f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ade09a62c203a5d225f95180c0855452348564f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
