)]}'
{"id":"openstack%2Ftacker~1004248","triplet_id":"openstack%2Ftacker~master~Id8c0de185a2f1c77dc2f925dae10f551b03ed041","project":"openstack/tacker","branch":"master","hashtags":[],"change_id":"Id8c0de185a2f1c77dc2f925dae10f551b03ed041","subject":"tox: Drop redundant passenv","status":"NEW","created":"2026-09-05 18:39:16.000000000","updated":"2026-09-05 21:07:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"33e0fb8980e5303cd27b92cbf650279388f4ffa6","_number":1004248,"virtual_id_number":1004248,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-05 21:07:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 21:07:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43ea5fdcc04cabfd8be9af2730336cf31024cdc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:39:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33e0fb8980e5303cd27b92cbf650279388f4ffa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 21:07:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d76ebe97efe4c3f85a9894d0a1b2e34\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f7318615f07e4d04addb734df74bd053 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/218d82cedd9c44ebbceaa9ab25072bbf : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa3ff04075341fdaf4176f2ceedc2c0 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5077d66b1db644028b14635ac3c507d9 : SUCCESS in 9m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b999efd368454d33907ca4d84096ac5e : SUCCESS in 8m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/032dbd4d11614a78bba6c3f12b7f4609 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d6f7fee4d643169718a9bb21758cdb : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f339ddcd14064150a6193b7c216f4a57 : SUCCESS in 5m 06s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/5553af9f9b994e87a9d2965f999685a8 : SUCCESS in 50m 22s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/edda5b588e9c4fae90d2d995ddc2f5fa : SUCCESS in 37m 53s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/43bf225385df4e16afb0dcd4ab257b11 : SUCCESS in 55m 44s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/ee3f47bccbf94c5492dc185385b0a795 : SUCCESS in 46m 49s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/23ff5346d27543f794d95ac3c5bfb402 : SUCCESS in 28m 04s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/0bdf4d8055e6400cb82cfb1cfe07de02 : SUCCESS in 45m 27s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/820e06e3eb5b45e2ac2ae97292022199 : SUCCESS in 1h 00m 19s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/59c326ccfc0f49998afc344ad2415c1d : SUCCESS in 54m 13s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/457d0b62f9d245a88ff000e9712001ca : SUCCESS in 1h 03m 09s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/a8ceb797def34074bd640db280287b38 : FAILURE in 49m 38s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/1e837a2c31d949b9a3f528bde86f2037 : SUCCESS in 1h 04m 16s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/7e46937c1aee48e7b6d2dc4632642979 : RETRY_LIMIT in 11m 07s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/55167bdaf5424bb5811b0dd76b0abfb6 : SUCCESS in 50m 23s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/5909b150048d4a2cbf518da7016af2e2 : SUCCESS in 51m 30s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/12a621d6bc4c4deb8d327ce9a91c2d7b : FAILURE in 2h 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb7a2702c23258ec5ef460084cfb4f19de6823df","revisions":{"eb7a2702c23258ec5ef460084cfb4f19de6823df":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:39:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/1004248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/48/1004248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/48/1004248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/48/1004248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/48/1004248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/48/1004248/1"}}},"commit":{"parents":[{"commit":"ade09a62c203a5d225f95180c0855452348564f5","subject":"Fix outdated default envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ade09a62c203a5d225f95180c0855452348564f5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:37:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:37:22.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: Id8c0de185a2f1c77dc2f925dae10f551b03ed041\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/eb7a2702c23258ec5ef460084cfb4f19de6823df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/eb7a2702c23258ec5ef460084cfb4f19de6823df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
