)]}'
{"id":"openstack%2Ftacker~1004366","triplet_id":"openstack%2Ftacker~master~I2074635729b66c78f7856c2e4de2f92f2edf017c","project":"openstack/tacker","branch":"master","attention_set":{"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2026-09-06 23:55:40.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-07 15:02:55.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I2074635729b66c78f7856c2e4de2f92f2edf017c","subject":"Fix typo","status":"NEW","created":"2026-09-06 23:54:16.000000000","updated":"2026-09-07 15:02:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"311e3e15ab58ac04fe06b2213da2cbf0f7ee0235","_number":1004366,"virtual_id_number":1004366,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-07 02:17:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 23:55:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2026-09-07 02:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b22c7fc91e438efad6f08321475d44d6beef9077","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:54:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"840ba5178a9bdd4a276ef00e2055017835a81dd4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:55:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7774527bf8256bbad56dfb5ad3077e32ee9b5a48","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:55:40.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"d89edf123d3acd15afdd581c3ba9af50dc2a2452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 02:17:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d18c9a40534449768fd360e81a272d88\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a85edca073194f23a31b0126ee85a490 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b04915e9c8cc45caba43f0c12d3dabda : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1867b618d0c94a6fbc3dc8333f6994e5 : SUCCESS in 9m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/895648122ba44e429ba94fe937d3313a : SUCCESS in 5m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7d8056a22b84f29a833761a04081ce4 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163fb39878c541209e70670b4ae51efd : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5173e4e555d4a71afef5cc2736434a6 : SUCCESS in 5m 26s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/bc723bc2c93845249d0065e81a1e4524 : SUCCESS in 53m 42s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/06e1ac860c4d418b988f8c0e3dc6d0c0 : SUCCESS in 44m 31s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/8d887b877ab14fdb972b394aa69827db : SUCCESS in 56m 17s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/5c5a3008cf2d41e687845853a9edf541 : SUCCESS in 50m 32s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/26a76abbe1064f7fa6e37697b969d683 : SUCCESS in 34m 02s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/a4ec15e94cad44a7939eb5b1d2aed184 : SUCCESS in 55m 37s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/771c59f0244343519644b3805da45ba4 : FAILURE in 59m 19s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/96e216e54a094b2d9c37f397e21b4a9e : SUCCESS in 39m 00s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/f98ba561a6584f529c4f703016dde3e2 : SUCCESS in 1h 03m 44s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/a0bb950e010e4f968a9c132ac07df61b : FAILURE in 50m 25s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/ef7096d0f2cb4ba3ad9c01b4d87d6148 : SUCCESS in 1h 09m 48s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/dd8f68d6b3ab4495ab006e8b7a561cf4 : RETRY_LIMIT in 11m 10s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/ea1d84c271ec48998c9525c3d62034ea : SUCCESS in 53m 30s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/6cbf37f6a0af4feab716ca8a5b5a7600 : SUCCESS in 1h 16m 42s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/16b934dd39d549978e388ae4200097ce : FAILURE in 2h 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf7222722fe31d3357fd5c17ce8802a822f50cc9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 11:36:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8bc00f1963d0f02398675280fc07009388411dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:05:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de8def55d5344c2faf944c2a37d2b7bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a9e361b8d414c75a9e93fc81a9471fb : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49fab2665acc499f8424d4d6c043a0f1 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba5cc5ccfcc044eda255f9b85ccb06e0 : SUCCESS in 9m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c554d75336f4bb580de6f3fe177432a : SUCCESS in 8m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ef7cf8a9a53451c90a75d4d03b4be16 : SUCCESS in 11m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae187404d7c42cb8997e84cab62a247 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94ddd8d3ac1f4b46a7c81a3365506546 : SUCCESS in 6m 03s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/b4f4e1fc400643c6bb867402a8aa1eb8 : SUCCESS in 1h 17m 36s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/dc1f31fb074e42c9a842f8f7cbf17a3f : SUCCESS in 1h 06m 55s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/51ede5aefb794ce6957e6eb2e4fae0ff : SUCCESS in 46m 08s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/3335ea29c7d14688b2127ebc3c5f2399 : SUCCESS in 1h 02m 51s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/2e67e97a957b436ab359cf92db7febe8 : SUCCESS in 1h 02m 46s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/8adcc7ac937041489047871e251cff5a : FAILURE in 1h 01m 42s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/508cafc19f484324ae792e8057eedf9f : SUCCESS in 1h 01m 49s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/12afa3f3e1374a058b79154a8490752f : SUCCESS in 43m 52s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/650e025ba1b0454ba92100ec7a5fedba : SUCCESS in 1h 02m 53s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/d1d304f1619c4927b0a66a30370d17d1 : FAILURE in 53m 23s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/82aa8a4983a74515a4b189c495fd144a : SUCCESS in 1h 05m 47s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/f72c53bf2bb8427880f4362e5368a2f3 : RETRY_LIMIT in 22m 04s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/7f45229bb91d47969d2d63bb2c9664de : SUCCESS in 53m 41s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/679744bd22184a9888efb5c6cc9c5671 : SUCCESS in 1h 20m 05s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/6dadd2e442014885a7464e17147b02ee : FAILURE in 2h 19m 37s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8f10f20cf2ce533845abaf29c0760276d8a5d08","revisions":{"4afb49bbbc94e327d2cb5af6c2fa339aacc28898":{"kind":"REWORK","_number":1,"created":"2026-09-06 23:54:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/66/1004366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/66/1004366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/66/1004366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/66/1004366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/66/1004366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/66/1004366/1"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:54:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:54:16.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I2074635729b66c78f7856c2e4de2f92f2edf017c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4afb49bbbc94e327d2cb5af6c2fa339aacc28898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4afb49bbbc94e327d2cb5af6c2fa339aacc28898"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"a8f10f20cf2ce533845abaf29c0760276d8a5d08":{"kind":"REWORK","_number":2,"created":"2026-09-06 23:55:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/66/1004366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/66/1004366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/66/1004366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/66/1004366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/66/1004366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/66/1004366/2"}}},"commit":{"parents":[{"commit":"975f4fd54ff67c59a2826c01275c0de72dcfcb22","subject":"Merge \"Fix typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/975f4fd54ff67c59a2826c01275c0de72dcfcb22"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:54:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:55:22.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I2074635729b66c78f7856c2e4de2f92f2edf017c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/a8f10f20cf2ce533845abaf29c0760276d8a5d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/a8f10f20cf2ce533845abaf29c0760276d8a5d08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
