)]}'
{"id":"openstack%2Ftacker~1007079","triplet_id":"openstack%2Ftacker~stable%2F2026.2~I97ce1826cf200675d1d85ff4b6f74b4755921c4a","project":"openstack/tacker","branch":"stable/2026.2","attention_set":{},"removed_from_attention_set":{"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2026-09-24 10:06:32.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2026-09-24 09:03:13.000000000","reason":"\u003cGERRIT_ACCOUNT_17255\u003e replied on the change","reason_account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}}},"hashtags":[],"change_id":"I97ce1826cf200675d1d85ff4b6f74b4755921c4a","subject":"Validate VNF package artifact Source paths to prevent traversal","status":"MERGED","created":"2026-09-24 01:18:37.000000000","updated":"2026-09-24 10:22:11.000000000","submitted":"2026-09-24 10:20:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007079","meta_rev_id":"784d3fe317b3f622396b1621d1079047872796ef","_number":1007079,"virtual_id_number":1007079,"owner":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-24 10:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2026-09-24 10:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2026-09-24 09:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":1,"date":"2026-09-24 10:06:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}],"CC":[{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-24 01:18:37.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},"state":"CC"},{"updated":"2026-09-24 01:21:40.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2026-09-24 03:47:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21caeef02a3937819015bc8f92a97b76e28648c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-09-24 01:18:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c5343f2fab4398d0ed7eafbd939facb59502e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 03:47:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ba8891ba9b248ab9456876aa2b50425\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/037d9195cc9949d783f293880df8b835 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c4eac436664009a69018ba49abc51d : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/462416f7f3f04891912a1605b87532e4 : SUCCESS in 9m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69920b5edce9405583829fdaf776e988 : SUCCESS in 5m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5e8b757005e544b8a35416b3837a8a83 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e13ecea80ef49528d14507ce708c759 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7872609a3d04c6293d4a7592dc32c39 : SUCCESS in 6m 10s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/d6829d0ea73a443c993b99759383eefb : SUCCESS in 1h 18m 56s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/bce6f52344e0408788cf352869067387 : SUCCESS in 27m 36s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/99ae525f0e8443a8a545a30b98bad5e3 : SUCCESS in 1h 00m 27s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/5674e6a87668488ca0cbe6cdca28ebad : SUCCESS in 1h 04m 52s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/dbe0718e1df140df8d0837790f8509cc : SUCCESS in 1h 04m 09s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/682d3ffcb7a341eaba7240c5a67fb941 : SUCCESS in 58m 34s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/8dffa51d6f874897b37b36a8b7b96149 : SUCCESS in 1h 01m 37s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/df321f2c8d0243aa8efab24aadf8c9df : SUCCESS in 54m 14s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/9dbb52cb79b34b99b69ddee1a7afd9c1 : SUCCESS in 1h 08m 46s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/bff8537cce704b4eb42bf76058ec1118 : SUCCESS in 57m 34s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/55245caaf4b5466f99e6212b524b26c0 : SUCCESS in 1h 07m 13s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/8f4eee4783764b4e884faa525555fb2a : RETRY_LIMIT in 15m 18s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/6c11fb25150144c993398da7652429da : SUCCESS in 57m 48s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/be44b14cf00c405a88eac6790fde6985 : SUCCESS in 1h 24m 47s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/1a102fb1f7424d4ba287473fa1780079 : FAILURE in 2h 21m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc15c469bcfae91256ef0a7e48c4162ce35a2480","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2026-09-24 09:03:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c809c6c7791e2ae97c84be96383d8d46ad5436ab","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-09-24 10:06:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be4a7ddb3644ef0a59f1b0cd9b6a8df3b6552aa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 10:06:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48f74050adf64a462d67d19dcfa83ae2128b5514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 10:20:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6e9ef79900a49b3ab4f4e344360676f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ed67a1e3f594d0fadbd8bce7105f275 : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88df18b2ead64d938ec012a477037be5 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e91e52f03e340a6817770f6940af377 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/773e656ab8644664b9d2994ff247a7ce : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f17fe7ab9ced4682816b0bf6b782a2f1 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"376db37dcf71458a285d38eba0e89f8ed8a8c3f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 10:20:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"784d3fe317b3f622396b1621d1079047872796ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 10:22:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/649d7658df7448478518c5529f56776f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e99a66da5243a2b2794a016833a7aa : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2818812b374655c67e2915fed58517e96e5b1732","revisions":{"2818812b374655c67e2915fed58517e96e5b1732":{"kind":"REWORK","_number":1,"created":"2026-09-24 01:18:37.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/79/1007079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/79/1007079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/79/1007079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/79/1007079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/79/1007079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/79/1007079/1"}}},"commit":{"parents":[{"commit":"60f779af7943b0bd5ec7a2321d74f834dcf3bd04","subject":"Merge \"Delete eventlet URL from Tacker documents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/60f779af7943b0bd5ec7a2321d74f834dcf3bd04"}]}],"author":{"name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","date":"2026-08-25 20:33:40.000000000","tz":0},"committer":{"name":"Yasufumi Ogawa","email":"yasufumi.ogawa@ntt.com","date":"2026-09-24 01:14:36.000000000","tz":540},"subject":"Validate VNF package artifact Source paths to prevent traversal","message":"Validate VNF package artifact Source paths to prevent traversal\n\nManifest Source paths were not checked to ensure they stayed\nwithin the VNF package directory, allowing a crafted CSAR to\naccess files outside the package directory.\n\nHence, this patch adds a check during onboarding to validate\nthe artifact Source path and ensure it stays within the VNF\npackage directory.\n\nCloses-Bug: https://bugs.launchpad.net/tacker/+bug/2154540\nChange-Id: I97ce1826cf200675d1d85ff4b6f74b4755921c4a\nSigned-off-by: Shivam Shukla \u003cshivam.shukla3@india.nec.com\u003e\n(cherry picked from commit 830d6bc3a66a10d6d3543416bc47c7a5fa28a81a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/2818812b374655c67e2915fed58517e96e5b1732"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/2818812b374655c67e2915fed58517e96e5b1732"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
