)]}'
{"id":"openstack%2Ftacker~325711","triplet_id":"openstack%2Ftacker~master~Id7af7173c62acc498fa695cdd1fa25c0890d0180","project":"openstack/tacker","branch":"master","topic":"bug/1589343","hashtags":[],"change_id":"Id7af7173c62acc498fa695cdd1fa25c0890d0180","subject":"Refactor respawn workflow for monitor","status":"MERGED","created":"2016-06-06 03:58:39.000000000","updated":"2016-06-09 19:46:47.000000000","submitted":"2016-06-09 19:46:47.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"150bb1cdc6ddb76f4d2197a2256be7884bdfdb03","_number":325711,"virtual_id_number":325711,"owner":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-06-09 19:46:47.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13380,"name":"Sridhar Ramaswamy","email":"srics.r@gmail.com","username":"srics-r"},{"value":0,"_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},{"value":0,"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},{"value":0,"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13380,"name":"Sridhar Ramaswamy","email":"srics.r@gmail.com","username":"srics-r"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-06-09 19:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13380,"name":"Sridhar Ramaswamy","email":"srics.r@gmail.com","username":"srics-r"},{"value":2,"date":"2016-06-09 19:37:43.000000000","_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},{"value":0,"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},{"value":0,"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13380,"name":"Sridhar Ramaswamy","email":"srics.r@gmail.com","username":"srics-r"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-06-09 19:41:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13380,"name":"Sridhar Ramaswamy","email":"srics.r@gmail.com","username":"srics-r"},{"value":0,"_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},{"value":0,"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},{"value":0,"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},{"_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},{"_account_id":13380,"name":"Sridhar Ramaswamy","email":"srics.r@gmail.com","username":"srics-r"},{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-09 06:31:27.000000000","updated_by":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"reviewer":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"state":"REVIEWER"},{"updated":"2016-06-09 19:37:43.000000000","updated_by":{"_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},"reviewer":{"_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},"state":"REVIEWER"},{"updated":"2016-06-09 19:41:32.000000000","updated_by":{"_account_id":13380,"name":"Sridhar Ramaswamy","email":"srics.r@gmail.com","username":"srics-r"},"reviewer":{"_account_id":13380,"name":"Sridhar Ramaswamy","email":"srics.r@gmail.com","username":"srics-r"},"state":"REVIEWER"},{"updated":"2016-06-09 19:46:47.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"d71be29b2d5b171e36bd89f322dac8d67645e79c","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-06 03:58:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e87063e41e67f7c3a4c5aaca1f429e16385378c","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-06 04:04:34.000000000","message":"Patch Set 2: Published edit on patch set 1","accounts_in_message":[],"_revision_number":2},{"id":"44c7a0dd5a94a822c26a8225746d2f5a0a0fcb08","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-06 04:04:58.000000000","message":"Patch Set 2: Code-Review-1\n\nThere is more to be done to fix this bug. Setting workflow to -1.","accounts_in_message":[],"_revision_number":2},{"id":"163ef2b30ada3d66a7d7c02bf5d664ba3b818864","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-06 04:05:07.000000000","message":"Patch Set 2: -Code-Review Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"da6aa3e0b8bf22b94aa37f6d3be1861affc22c71","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 04:44:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tacker-requirements http://logs.openstack.org/11/325711/2/check/gate-tacker-requirements/b307a47/ : SUCCESS in 1m 31s\n- gate-tacker-docs http://docs-draft.openstack.org/11/325711/2/check/gate-tacker-docs/f934b0d//doc/build/html/ : SUCCESS in 2m 46s\n- gate-tacker-pep8 http://logs.openstack.org/11/325711/2/check/gate-tacker-pep8/40c6550/ : SUCCESS in 2m 52s\n- gate-tacker-python27-db http://logs.openstack.org/11/325711/2/check/gate-tacker-python27-db/95884c6/ : SUCCESS in 2m 44s\n- gate-tacker-dsvm-functional http://logs.openstack.org/11/325711/2/check/gate-tacker-dsvm-functional/a388667/ : SUCCESS in 35m 56s (non-voting)\n- gate-tacker-python34-db http://logs.openstack.org/11/325711/2/check/gate-tacker-python34-db/0f9b09c/ : SUCCESS in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"211619022f3b1e42e114a15c3d7838d6451a6722","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2016-06-06 04:51:05.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"aa5966be724d2b0ceacd20763ae5d25d88698491","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2016-06-06 05:08:24.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks for fixing it, it saved my day !","accounts_in_message":[],"_revision_number":3},{"id":"4afb6277c3d4d55fed4189afea9e59c5415b653b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 05:30:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tacker-requirements http://logs.openstack.org/11/325711/3/check/gate-tacker-requirements/05d3715/ : SUCCESS in 1m 37s\n- gate-tacker-docs http://docs-draft.openstack.org/11/325711/3/check/gate-tacker-docs/f64abab//doc/build/html/ : SUCCESS in 3m 10s\n- gate-tacker-pep8 http://logs.openstack.org/11/325711/3/check/gate-tacker-pep8/46c812e/ : SUCCESS in 3m 24s\n- gate-tacker-python27-db http://logs.openstack.org/11/325711/3/check/gate-tacker-python27-db/9a6ca95/ : SUCCESS in 3m 09s\n- gate-tacker-dsvm-functional http://logs.openstack.org/11/325711/3/check/gate-tacker-dsvm-functional/ada22d5/ : SUCCESS in 38m 01s (non-voting)\n- gate-tacker-python34-db http://logs.openstack.org/11/325711/3/check/gate-tacker-python34-db/fc86aed/ : SUCCESS in 3m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"83977828c95d6c1262fd890c1dbb67dd27992a59","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-08 01:38:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3c27377a88ddd1d0a5bb6e6794c2e3086cc86385","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-08 02:27:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tacker-requirements http://logs.openstack.org/11/325711/4/check/gate-tacker-requirements/75b8dc3/ : SUCCESS in 1m 40s\n- gate-tacker-docs http://docs-draft.openstack.org/11/325711/4/check/gate-tacker-docs/e6b0994//doc/build/html/ : SUCCESS in 3m 21s\n- gate-tacker-pep8 http://logs.openstack.org/11/325711/4/check/gate-tacker-pep8/f1e179f/ : SUCCESS in 3m 26s\n- gate-tacker-python27-db http://logs.openstack.org/11/325711/4/check/gate-tacker-python27-db/9cb79c2/ : SUCCESS in 3m 00s\n- gate-tacker-dsvm-functional http://logs.openstack.org/11/325711/4/check/gate-tacker-dsvm-functional/dbddf50/ : SUCCESS in 45m 20s (non-voting)\n- gate-tacker-python34-db http://logs.openstack.org/11/325711/4/check/gate-tacker-python34-db/6c797ee/ : SUCCESS in 2m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"52370b6892cd4c03cc4fc41c5c89e24d77e46916","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2016-06-08 09:00:01.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"bad50911bd2928ca1499bb1a43310112f0f1f08e","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2016-06-08 09:31:48.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5b17e8416fa53583f152612488140e41f2e4c226","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-08 09:54:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tacker-requirements http://logs.openstack.org/11/325711/5/check/gate-tacker-requirements/6270b31/ : SUCCESS in 2m 35s\n- gate-tacker-docs http://docs-draft.openstack.org/11/325711/5/check/gate-tacker-docs/6932a98//doc/build/html/ : SUCCESS in 3m 31s\n- gate-tacker-pep8 http://logs.openstack.org/11/325711/5/check/gate-tacker-pep8/678d726/ : SUCCESS in 3m 40s\n- gate-tacker-python27-db http://logs.openstack.org/11/325711/5/check/gate-tacker-python27-db/46adea7/ : SUCCESS in 3m 21s\n- gate-tacker-releasenotes http://docs-draft.openstack.org/11/325711/5/check/gate-tacker-releasenotes/fbfcce6//releasenotes/build/html/ : SUCCESS in 3m 38s\n- gate-tacker-dsvm-functional http://logs.openstack.org/11/325711/5/check/gate-tacker-dsvm-functional/f9e1ba7/ : SUCCESS in 42m 38s (non-voting)\n- gate-tacker-python34-db http://logs.openstack.org/11/325711/5/check/gate-tacker-python34-db/d6e3df2/ : SUCCESS in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9794e3508fa485e54d33d72f9dc77e13442c3e29","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2016-06-08 12:39:10.000000000","message":"Patch Set 5: Code-Review+1\n\nThanks Sripriya, Now all dependent db patches passed the functional test cases.","accounts_in_message":[],"_revision_number":5},{"id":"0d946a815779342a3fcf1bcac2e42b801bda0653","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-08 20:28:44.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9f3e0d29246b9fca21b57fab8b504f99234d0969","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-08 23:30:28.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"136738624ae84c640be462e958fa28a96d4fe0fe","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-09 00:05:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"890769f0ea41318c27e389bf73b2595b92954775","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 02:22:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tacker-docs http://docs-draft.openstack.org/11/325711/6/check/gate-tacker-docs/0268e48//doc/build/html/ : SUCCESS in 2m 49s\n- gate-tacker-pep8 http://logs.openstack.org/11/325711/6/check/gate-tacker-pep8/45d438d/ : SUCCESS in 2m 31s\n- gate-tacker-python27-db http://logs.openstack.org/11/325711/6/check/gate-tacker-python27-db/922edc8/ : SUCCESS in 3m 18s\n- gate-tacker-dsvm-functional http://logs.openstack.org/11/325711/6/check/gate-tacker-dsvm-functional/d7a8935/ : FAILURE in 1h 41m 32s (non-voting)\n- gate-tacker-python34-db http://logs.openstack.org/11/325711/6/check/gate-tacker-python34-db/15360e6/ : SUCCESS in 2m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7376cabf12c0828ba2de63f830397745ddf3c7db","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2016-06-09 03:25:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4a049b7e19e8d893053115a8c8950b8a0117108e","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2016-06-09 03:26:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1bab83691b2e12314991de756698b34101f98bcb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 04:01:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tacker-requirements http://logs.openstack.org/11/325711/7/check/gate-tacker-requirements/099e41d/ : SUCCESS in 1m 51s\n- gate-tacker-docs http://docs-draft.openstack.org/11/325711/7/check/gate-tacker-docs/fd2d4e4//doc/build/html/ : SUCCESS in 2m 44s\n- gate-tacker-pep8 http://logs.openstack.org/11/325711/7/check/gate-tacker-pep8/a85feed/ : SUCCESS in 1m 59s\n- gate-tacker-python27-db http://logs.openstack.org/11/325711/7/check/gate-tacker-python27-db/2328ca1/ : SUCCESS in 3m 02s\n- gate-tacker-releasenotes http://docs-draft.openstack.org/11/325711/7/check/gate-tacker-releasenotes/5d65322//releasenotes/build/html/ : SUCCESS in 2m 55s\n- gate-tacker-dsvm-functional http://logs.openstack.org/11/325711/7/check/gate-tacker-dsvm-functional/976aa40/ : SUCCESS in 34m 50s (non-voting)\n- gate-tacker-python34-db http://logs.openstack.org/11/325711/7/check/gate-tacker-python34-db/9641597/ : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"af7a005ed17fd008981abe12ce88b02786e02f12","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2016-06-09 06:31:27.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"e018ab68ec062296f6ce11329fbf4e0a2d7b9721","author":{"_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},"date":"2016-06-09 11:57:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b3a2120353cd4bb8faa929d091a66b0e548569cd","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-09 17:49:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b0dd6752fd754fd4a2967301fb1b4d7ca6751307","author":{"_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},"date":"2016-06-09 17:58:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5ad7009aeeea9d47cc2490c08edde1f318a7fbb8","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-09 18:28:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d0f5a419d2d2481d2d6b4f6bf0e63c37d9375361","author":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"date":"2016-06-09 18:37:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"409aeaa542a1deaa73d9efce72cfc59eac7ca230","author":{"_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},"date":"2016-06-09 18:38:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ab32a1570eac810b282c828392f088edd07c0e37","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 19:37:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tacker-docs http://docs-draft.openstack.org/11/325711/8/check/gate-tacker-docs/a729fe0//doc/build/html/ : SUCCESS in 5m 56s\n- gate-tacker-pep8 http://logs.openstack.org/11/325711/8/check/gate-tacker-pep8/94880cf/ : SUCCESS in 4m 12s\n- gate-tacker-python27-db http://logs.openstack.org/11/325711/8/check/gate-tacker-python27-db/ff07e8c/ : SUCCESS in 2m 58s\n- gate-tacker-dsvm-functional http://logs.openstack.org/11/325711/8/check/gate-tacker-dsvm-functional/8a09a2d/ : SUCCESS in 44m 23s (non-voting)\n- gate-tacker-python34-db http://logs.openstack.org/11/325711/8/check/gate-tacker-python34-db/7f36b7f/ : SUCCESS in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b972e32b0bcaa974299d52e542910dd0eb6382df","author":{"_account_id":12455,"name":"bharaththiruveedula","email":"bharath_ves@hotmail.com","username":"bharaththiruveedula"},"date":"2016-06-09 19:37:43.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"1cdd628495f2ef6bfa177931c53aebb0c0e5223d","author":{"_account_id":13380,"name":"Sridhar Ramaswamy","email":"srics.r@gmail.com","username":"srics-r"},"date":"2016-06-09 19:41:32.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"f007be0745a384cfcdb0982ed372227eb280cd19","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 19:42:18.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"d2733938d7dfec65abc75d925eb10748c954d113","date":"2016-06-09 19:46:47.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":8},{"id":"1463f87fa7e1750ab9836b038dc9a2a6a0e37562","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 19:46:47.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tacker-docs http://docs-draft.openstack.org/11/325711/8/gate/gate-tacker-docs/e7fea5a//doc/build/html/ : SUCCESS in 1m 55s\n- gate-tacker-pep8 http://logs.openstack.org/11/325711/8/gate/gate-tacker-pep8/abb17f2/ : SUCCESS in 2m 55s\n- gate-tacker-python27-db http://logs.openstack.org/11/325711/8/gate/gate-tacker-python27-db/e18a7c4/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"840d39120f6793e61c085016daee32da4fcf9b34","revisions":{"a634fb3bb5905351d0d73584fca5a9e91c963e99":{"kind":"REWORK","_number":1,"created":"2016-06-06 03:58:39.000000000","uploader":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"ref":"refs/changes/11/325711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/11/325711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/11/325711/1"}}},"commit":{"parents":[{"commit":"4a66645c1c0c80d2bb6e37d2f53ea4060d4d6a06","subject":"Merge \"Change default port into 9890\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4a66645c1c0c80d2bb6e37d2f53ea4060d4d6a06"}]}],"author":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-06 03:50:23.000000000","tz":-420},"committer":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-06 03:57:06.000000000","tz":-420},"subject":"Update new device id assignment in monitor respawn","message":"Update new device id assignment in monitor respawn\n\nReplace the \u003cuuid\u003e-RESPAWN-\u003ccount\u003e id format for new device with\njust the normal uuid in monitor respawn logic.\n\nThis is done to ensure resource ids in db always obey the uuid format.\n\nChange-Id: Id7af7173c62acc498fa695cdd1fa25c0890d0180\nCloses-Bug: #1589343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/a634fb3bb5905351d0d73584fca5a9e91c963e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/a634fb3bb5905351d0d73584fca5a9e91c963e99"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4a66645c1c0c80d2bb6e37d2f53ea4060d4d6a06","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f96794c5b64431ee588eab8f46cf588ef01e2393":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2016-06-06 04:04:34.000000000","uploader":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"ref":"refs/changes/11/325711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/11/325711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/11/325711/2"}}},"commit":{"parents":[{"commit":"4a66645c1c0c80d2bb6e37d2f53ea4060d4d6a06","subject":"Merge \"Change default port into 9890\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4a66645c1c0c80d2bb6e37d2f53ea4060d4d6a06"}]}],"author":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-06 03:50:23.000000000","tz":-420},"committer":{"name":"Sripriya Seetharam","email":"sseetha@brocade.com","date":"2016-06-06 04:04:05.000000000","tz":0},"subject":"[WIP] Update new device id assignment in monitor respawn","message":"[WIP] Update new device id assignment in monitor respawn\n\nReplace the \u003cuuid\u003e-RESPAWN-\u003ccount\u003e id format for new device with\njust the normal uuid in monitor respawn logic.\n\nThis is done to ensure resource ids in db always obey the uuid format.\n\nChange-Id: Id7af7173c62acc498fa695cdd1fa25c0890d0180\nCloses-Bug: #1589343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/f96794c5b64431ee588eab8f46cf588ef01e2393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/f96794c5b64431ee588eab8f46cf588ef01e2393"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4a66645c1c0c80d2bb6e37d2f53ea4060d4d6a06","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c16c1b15301ec50c824e3c7ceb88ce7766d5ce64":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-06-06 04:51:05.000000000","uploader":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"ref":"refs/changes/11/325711/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/11/325711/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/11/325711/3"}}},"commit":{"parents":[{"commit":"0e69bce55c76439ce86d56d018d5e7f40fee6ef4","subject":"Merge \"Remove unused model class from db layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/0e69bce55c76439ce86d56d018d5e7f40fee6ef4"}]}],"author":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-06 03:50:23.000000000","tz":-420},"committer":{"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","date":"2016-06-06 04:49:57.000000000","tz":330},"subject":"[WIP] Update new device id assignment in monitor respawn","message":"[WIP] Update new device id assignment in monitor respawn\n\nReplace the \u003cuuid\u003e-RESPAWN-\u003ccount\u003e id format for new device with\njust the normal uuid in monitor respawn logic.\n\nThis is done to ensure resource ids in db always obey the uuid format.\n\nChange-Id: Id7af7173c62acc498fa695cdd1fa25c0890d0180\nCloses-Bug: #1589343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/c16c1b15301ec50c824e3c7ceb88ce7766d5ce64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/c16c1b15301ec50c824e3c7ceb88ce7766d5ce64"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0e69bce55c76439ce86d56d018d5e7f40fee6ef4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"31886fd6f8349fbeb80638f7cced196d55e116ce":{"kind":"REWORK","_number":4,"created":"2016-06-08 01:38:26.000000000","uploader":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"ref":"refs/changes/11/325711/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/11/325711/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/11/325711/4"}}},"commit":{"parents":[{"commit":"0e69bce55c76439ce86d56d018d5e7f40fee6ef4","subject":"Merge \"Remove unused model class from db layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/0e69bce55c76439ce86d56d018d5e7f40fee6ef4"}]}],"author":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-06 03:50:23.000000000","tz":-420},"committer":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-08 01:34:40.000000000","tz":-420},"subject":"Refactor respawn workflow for monitor","message":"Refactor respawn workflow for monitor\n\nRemove the hacks from monitor and db files for keeping the\ndevice id unchanged.\n\nNew approach \u0027updates\u0027 existing device record instead of creating\nnew device record with old device id and deleting old record.\n\nChange-Id: Id7af7173c62acc498fa695cdd1fa25c0890d0180\nCloses-Bug: #1589343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/31886fd6f8349fbeb80638f7cced196d55e116ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/31886fd6f8349fbeb80638f7cced196d55e116ce"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0e69bce55c76439ce86d56d018d5e7f40fee6ef4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"54cb7e2d1cbe0aac5d540231897b4ee6cfa4a0b8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2016-06-08 09:00:01.000000000","uploader":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"ref":"refs/changes/11/325711/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/11/325711/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/11/325711/5"}}},"commit":{"parents":[{"commit":"962d7f8055ec08202371cb6b0109f364b52ad314","subject":"Merge \"oslo: remove usage of oslo.log verbose option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/962d7f8055ec08202371cb6b0109f364b52ad314"}]}],"author":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-06 03:50:23.000000000","tz":-420},"committer":{"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","date":"2016-06-08 08:55:41.000000000","tz":330},"subject":"Refactor respawn workflow for monitor","message":"Refactor respawn workflow for monitor\n\nRemove the hacks from monitor and db files for keeping the\ndevice id unchanged.\n\nNew approach \u0027updates\u0027 existing device record instead of creating\nnew device record with old device id and deleting old record.\n\nChange-Id: Id7af7173c62acc498fa695cdd1fa25c0890d0180\nCloses-Bug: #1589343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/54cb7e2d1cbe0aac5d540231897b4ee6cfa4a0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/54cb7e2d1cbe0aac5d540231897b4ee6cfa4a0b8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"962d7f8055ec08202371cb6b0109f364b52ad314","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"89f167b537254eb57b527df205b7b6e488abad6f":{"kind":"REWORK","_number":6,"created":"2016-06-09 00:05:05.000000000","uploader":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"ref":"refs/changes/11/325711/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/11/325711/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/11/325711/6"}}},"commit":{"parents":[{"commit":"1d67184b54bf049bd7f166d5981ce9ece566377d","subject":"Merge \"Log message correction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1d67184b54bf049bd7f166d5981ce9ece566377d"}]}],"author":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-06 03:50:23.000000000","tz":-420},"committer":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-09 00:04:50.000000000","tz":-420},"subject":"Refactor respawn workflow for monitor","message":"Refactor respawn workflow for monitor\n\nRemove the hacks from monitor and db files for keeping the\ndevice id unchanged.\n\nNew approach \u0027updates\u0027 existing device record instead of creating\nnew device record with old device id and deleting old record.\n\nChange-Id: Id7af7173c62acc498fa695cdd1fa25c0890d0180\nCloses-Bug: #1589343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/89f167b537254eb57b527df205b7b6e488abad6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/89f167b537254eb57b527df205b7b6e488abad6f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d67184b54bf049bd7f166d5981ce9ece566377d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"709db587c8cbb77d3ae55416526eb4e0e935c2f0":{"kind":"REWORK","_number":7,"created":"2016-06-09 03:25:05.000000000","uploader":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"ref":"refs/changes/11/325711/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/11/325711/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/11/325711/7"}}},"commit":{"parents":[{"commit":"1d67184b54bf049bd7f166d5981ce9ece566377d","subject":"Merge \"Log message correction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1d67184b54bf049bd7f166d5981ce9ece566377d"}]}],"author":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-06 03:50:23.000000000","tz":-420},"committer":{"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","date":"2016-06-09 03:06:35.000000000","tz":330},"subject":"Refactor respawn workflow for monitor","message":"Refactor respawn workflow for monitor\n\nRemove the hacks from monitor and db files for keeping the\ndevice id unchanged.\n\nNew approach \u0027updates\u0027 existing device record instead of creating\nnew device record with old device id and deleting old record.\n\nChange-Id: Id7af7173c62acc498fa695cdd1fa25c0890d0180\nCloses-Bug: #1589343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/709db587c8cbb77d3ae55416526eb4e0e935c2f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/709db587c8cbb77d3ae55416526eb4e0e935c2f0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d67184b54bf049bd7f166d5981ce9ece566377d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"840d39120f6793e61c085016daee32da4fcf9b34":{"kind":"REWORK","_number":8,"created":"2016-06-09 18:37:19.000000000","uploader":{"_account_id":13485,"name":"Sripriya Seetharam","email":"sseetha@brocade.com","username":"Sripriya"},"ref":"refs/changes/11/325711/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/11/325711/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/11/325711/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/11/325711/8"}}},"commit":{"parents":[{"commit":"1d67184b54bf049bd7f166d5981ce9ece566377d","subject":"Merge \"Log message correction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1d67184b54bf049bd7f166d5981ce9ece566377d"}]}],"author":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-06 03:50:23.000000000","tz":-420},"committer":{"name":"Sripriya","email":"sseetha@brocade.com","date":"2016-06-09 18:37:12.000000000","tz":-420},"subject":"Refactor respawn workflow for monitor","message":"Refactor respawn workflow for monitor\n\nRemove the hacks from monitor and db files for keeping the\ndevice id unchanged.\n\nNew approach \u0027updates\u0027 existing device record instead of creating\nnew device record with old device id and deleting old record.\n\nChange-Id: Id7af7173c62acc498fa695cdd1fa25c0890d0180\nCloses-Bug: #1589343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/840d39120f6793e61c085016daee32da4fcf9b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/840d39120f6793e61c085016daee32da4fcf9b34"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d67184b54bf049bd7f166d5981ce9ece566377d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
