)]}'
{"id":"openstack%2Ftacker~530306","triplet_id":"openstack%2Ftacker~master~Ib76f5ca1735d66bdcf5c25503dcecc6b536b59d3","project":"openstack/tacker","branch":"master","topic":"remove_unused_constants","hashtags":[],"change_id":"Ib76f5ca1735d66bdcf5c25503dcecc6b536b59d3","subject":"Remove unused constants","status":"MERGED","created":"2017-12-28 07:03:53.000000000","updated":"2018-01-12 04:14:14.000000000","submitted":"2018-01-12 04:14:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"530306-1515730454239-c1faceed","meta_rev_id":"5b08f7e3dc1f8766010313701dd1a8d49705ba20","_number":530306,"virtual_id_number":530306,"owner":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":2,"date":"2018-01-12 04:14:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"recommended":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":1,"date":"2018-01-11 09:29:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2018-01-11 09:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-01-11 09:57:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},{"value":1,"date":"2018-01-11 13:46:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2018-01-12 04:02:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-11 09:29:24.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2018-01-11 09:57:24.000000000","updated_by":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"reviewer":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"state":"REVIEWER"},{"updated":"2018-01-11 13:46:37.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2018-01-12 04:02:38.000000000","updated_by":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"reviewer":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"state":"REVIEWER"},{"updated":"2018-01-12 04:14:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93e283fa0c5ac686ed04ee3eff9092112183d5b2","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2017-12-28 07:03:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6521ac7d7f2ce3357625bcb4a09a30b056e21e12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-28 09:14:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/06/530306/1/check/openstack-tox-pep8/3f3093b/ : SUCCESS in 9m 06s\n- openstack-tox-py27 http://logs.openstack.org/06/530306/1/check/openstack-tox-py27/ff1a233/ : SUCCESS in 4m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/530306/1/check/build-openstack-sphinx-docs/ade1130/html/ : SUCCESS in 6m 43s\n- openstack-tox-py35 http://logs.openstack.org/06/530306/1/check/openstack-tox-py35/83949c7/ : SUCCESS in 8m 52s\n- tacker-functional-devstack http://logs.openstack.org/06/530306/1/check/tacker-functional-devstack/bb6225f/ : RETRY_LIMIT in 33m 15s (non-voting)\n- tacker-functional-devstack-python3 http://logs.openstack.org/06/530306/1/check/tacker-functional-devstack-python3/2b09810/ : RETRY_LIMIT in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10b29c9c05e016346e03e635815dd1278d7b3d0b","author":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"date":"2018-01-03 03:03:34.000000000","message":"Patch Set 1: Code-Review-1\n\nbut I cannot see it is used in code","accounts_in_message":[],"_revision_number":1},{"id":"8957907b05ed7ef30f71ef1a3b32aebff2ec995a","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-01-04 01:43:41.000000000","message":"Patch Set 1:\n\nHi gongysh,\n\nThank you for the review.\nI can see the constants from this module are not used anywhere.\nThis code was added in below commits at the initial phase of tacker when the code was imported from neutron:\n\nac91ef1445d190bcc77bd7108c2c23669c97d8a3\n8aff29509f1f6cfc73973b8da3e55ed559ae1112","accounts_in_message":[],"_revision_number":1},{"id":"c48e104d873e78fb9457fcfaa2ec0487bbfbba0e","author":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"date":"2018-01-08 07:31:27.000000000","message":"Patch Set 1:\n\nwe can just remove it.","accounts_in_message":[],"_revision_number":1},{"id":"8c32a1c33e15d4a9fe19e93a9c8ea6be8702f504","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-01-11 09:09:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30221495e5a052c1c13628d8c492d44873544762","author":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"date":"2018-01-11 09:16:21.000000000","message":"Patch Set 2: Code-Review+2\n\nwill merge after CI pass. thanks","accounts_in_message":[],"_revision_number":2},{"id":"8eb0e5f7247e9239993d30b4754fed6b4d739455","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-01-11 09:29:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"67478884b4526e26d22513b2ba7e05f1a0c514ca","author":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"date":"2018-01-11 09:57:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6a678abb21b02843945f5f78d2b669bd06483552","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2018-01-11 13:46:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6636b1567832c8686887a043ee0c3c599e8c6a7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-11 15:58:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/06/530306/2/check/openstack-tox-pep8/3588182/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/06/530306/2/check/openstack-tox-py27/3380d83/ : SUCCESS in 7m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/530306/2/check/build-openstack-sphinx-docs/5b182ab/html/ : SUCCESS in 6m 28s\n- openstack-tox-py35 http://logs.openstack.org/06/530306/2/check/openstack-tox-py35/5244158/ : SUCCESS in 9m 46s\n- tacker-functional-devstack http://logs.openstack.org/06/530306/2/check/tacker-functional-devstack/22c356f/ : POST_FAILURE in 2h 49m 10s (non-voting)\n- tacker-functional-devstack-python3 http://logs.openstack.org/06/530306/2/check/tacker-functional-devstack-python3/ffa4858/ : POST_FAILURE in 2h 48m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"482174b1bf38d534f067b7b35b59712f331383e7","author":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"date":"2018-01-12 04:02:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d0ef13ff62f6113fec58a6aa90334009344bfebf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-12 04:02:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d106fb4983676cafa434ca2596103a335209da76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-12 04:14:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/06/530306/2/gate/openstack-tox-pep8/3b89035/ : SUCCESS in 6m 24s\n- openstack-tox-py27 http://logs.openstack.org/06/530306/2/gate/openstack-tox-py27/f795c60/ : SUCCESS in 6m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/530306/2/gate/build-openstack-sphinx-docs/6ab11fa/html/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/06/530306/2/gate/openstack-tox-py35/a9a5736/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"72d4b87244035e45be4cb11e6dee04909509ea44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-12 04:14:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea03d156e5e72a08c9a5d16fd048c555ea06305d","revisions":{"3dbd8009831f242f4d5cc7c466af4945fe08c5f5":{"kind":"REWORK","_number":1,"created":"2017-12-28 07:03:53.000000000","uploader":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"ref":"refs/changes/06/530306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/06/530306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/530306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/530306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/530306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/06/530306/1"}}},"commit":{"parents":[{"commit":"583f2eb840b44f885cbbb57b20e4d3df32e72dec","subject":"Merge \"Add reno note for symmetric bug fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/583f2eb840b44f885cbbb57b20e4d3df32e72dec"}]}],"author":{"name":"dineshbhor","email":"dinesh.bhor@nttdata.com","date":"2017-12-28 06:48:49.000000000","tz":540},"committer":{"name":"dineshbhor","email":"dinesh.bhor@nttdata.com","date":"2017-12-28 07:03:49.000000000","tz":540},"subject":"Trivial: Fix spelling mistake","message":"Trivial: Fix spelling mistake\n\nThis patch fixes a spelling mistake for the constant defined below:\nATTR_NAME_SCHEDULER_HINTS \u003d \u0027sheculer_hints\u0027\n\nThe \u0027sheculer_hints\u0027 is changed to \u0027scheduler_hints\u0027.\n\nTrivialFix\n\nChange-Id: Ib76f5ca1735d66bdcf5c25503dcecc6b536b59d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/3dbd8009831f242f4d5cc7c466af4945fe08c5f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/3dbd8009831f242f4d5cc7c466af4945fe08c5f5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"583f2eb840b44f885cbbb57b20e4d3df32e72dec","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ea03d156e5e72a08c9a5d16fd048c555ea06305d":{"kind":"REWORK","_number":2,"created":"2018-01-11 09:09:57.000000000","uploader":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"ref":"refs/changes/06/530306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/06/530306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/530306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/530306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/530306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/06/530306/2"}}},"commit":{"parents":[{"commit":"23b47b8588fe641ecb5e92399c7cbd3515719761","subject":"Merge \"remove tacker horizon during unstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/23b47b8588fe641ecb5e92399c7cbd3515719761"}]}],"author":{"name":"dineshbhor","email":"dinesh.bhor@nttdata.com","date":"2017-12-28 06:48:49.000000000","tz":540},"committer":{"name":"dineshbhor","email":"dinesh.bhor@nttdata.com","date":"2018-01-11 08:50:08.000000000","tz":540},"subject":"Remove unused constants","message":"Remove unused constants\n\nThere are constants defined in below module which are not used\nanywhere in the tacker code:\n    tacker.common.constants.py\n\nThis patch removes such unused constants.\n\nTrivialFix\nChange-Id: Ib76f5ca1735d66bdcf5c25503dcecc6b536b59d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ea03d156e5e72a08c9a5d16fd048c555ea06305d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ea03d156e5e72a08c9a5d16fd048c555ea06305d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"23b47b8588fe641ecb5e92399c7cbd3515719761","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
