)]}'
{"id":"openstack%2Ftacker~751965","triplet_id":"openstack%2Ftacker~master~I93dbd84cec252bb93257b852aa171d9ca89665c2","project":"openstack/tacker","branch":"master","topic":"fix-env-ft","hashtags":[],"change_id":"I93dbd84cec252bb93257b852aa171d9ca89665c2","subject":"Temporarily remove ceilometer install","status":"MERGED","created":"2020-09-15 04:14:46.000000000","updated":"2020-09-18 07:47:03.000000000","submitted":"2020-09-18 07:41:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751965-1600414889346-193fd9bf","meta_rev_id":"c19b2730510a86b95c7f73272758c9be7e13b67b","_number":751965,"virtual_id_number":751965,"owner":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"_account_id":30901,"name":"Masazumi ota","email":"masazumi.ota@ntt.com"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"date":"2020-09-18 01:59:56.000000000","_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 07:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"recommended":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"all":[{"value":1,"date":"2020-09-18 06:56:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},{"value":2,"date":"2020-09-18 07:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":1,"date":"2020-09-18 01:59:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30901,"name":"Masazumi ota","email":"masazumi.ota@ntt.com"},{"value":1,"date":"2020-09-18 06:51:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-18 01:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":2,"date":"2020-09-18 07:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},{"value":1,"date":"2020-09-18 07:26:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"_account_id":30901,"name":"Masazumi ota","email":"masazumi.ota@ntt.com"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"value":0,"date":"2020-09-18 01:56:30.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},{"_account_id":30901,"name":"Masazumi ota","email":"masazumi.ota@ntt.com"},{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 00:53:13.000000000","updated_by":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"reviewer":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"state":"REVIEWER"},{"updated":"2020-09-18 01:56:30.000000000","updated_by":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"reviewer":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"state":"REVIEWER"},{"updated":"2020-09-18 01:58:35.000000000","updated_by":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"reviewer":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"state":"REVIEWER"},{"updated":"2020-09-18 01:59:40.000000000","updated_by":{"_account_id":30901,"name":"Masazumi ota","email":"masazumi.ota@ntt.com"},"reviewer":{"_account_id":30901,"name":"Masazumi ota","email":"masazumi.ota@ntt.com"},"state":"REVIEWER"},{"updated":"2020-09-18 06:51:55.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2020-09-18 06:56:05.000000000","updated_by":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"reviewer":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"state":"REVIEWER"},{"updated":"2020-09-18 07:06:06.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2020-09-18 07:26:24.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-09-18 07:41:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b1620c666a705400f5539973d6d52d15a729996","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-15 04:14:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddb4eebebd8e4330baf791ac282a3eeadb007d1a","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-15 06:08:25.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks! I understand it\u0027s WIP, but suggest one thing so that it should be merged quickly.","accounts_in_message":[],"_revision_number":1},{"id":"8dba9816948ecafd341df6d77b2089ce30193184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 07:30:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93ab8d0947d247dabe2e69533251f68b : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f6aac7a12a7492ba98000e5dd28290a : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201dcd0072c543a0a54e21ec168472e9 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37c3659d48464bb8aa7cf309241ba35f : FAILURE in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/719566b47f8744b2b422b9f14bab72fe : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d060f292ff401f97c07720f6bb9e94 : SUCCESS in 8m 08s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/8a4826126dd649be890dfadbd1f6f021 : RETRY_LIMIT in 32m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"809850dffc1c6019e73003e17ff99d755e72611f","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-15 08:30:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d2ee97b1bbf92e77b226e6bd25215986d03d832","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-15 18:38:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5e06736fa61258fa13a8d7b7251b6204b67312a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 21:27:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ce1e07842cd440680b9eb0cc3844b5a : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3618bda349804735b93b16653197d96d : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428584f2bcb549bab1b5e70cdcfefb5a : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/362950a25289461cbbdea4e906f0cefb : SUCCESS in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/877c7345ef9a4ce5a8ed5e4ef283906c : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c33fe6d73a54f3c8d692969b14aea86 : SUCCESS in 5m 36s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/8bd4897d6e9c4857ac4d91980b0c5242 : RETRY_LIMIT in 47m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc77a26c094037b8d8f9d8fed447129938892b47","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2020-09-16 00:53:13.000000000","message":"Patch Set 1: Code-Review-1\n\nAs you can see the log of functional test, installing barbican faied with mysql error. There may be problems other than gnocchi.","accounts_in_message":[],"_revision_number":1},{"id":"f5c80ab4f7a10e4c5f91802501bdb4dbba10fa27","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-16 01:24:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ac761bb6783016fad0d17b7f9bc938b83827e2a","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-16 01:32:53.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5b96af43bfd12142a5cd50dce394b13641a3ab59","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-16 01:37:27.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"295f1243c7fd63d7595bb7d599391186a2fd8d3b","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-16 01:48:45.000000000","message":"Patch Set 4:\n\n(1 comment)\n\n\u003e As you can see the log of functional test, installing barbican\n \u003e faied with mysql error. There may be problems other than gnocchi.\n\nThank you for the comment.\nI\u0027m trying to remove installing barbican too.\nThis is also issues which OpenStack members know.","accounts_in_message":[],"_revision_number":4},{"id":"cc1b596872bb610adc5d1d1899d996f75b62dcf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 04:18:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8baba06c337f4c6985e70ad0b4975ab7 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf564b3197ce4fca98f1ca29d1ec3cac : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b47f422caf44ed785ae730de0214628 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb9cd40e04d64effae50725a627a471f : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d5771f4d22c4650a0d69dbc2edf8cae : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d879570f5a714885a3477fd856c21f76 : SUCCESS in 5m 39s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/719a21d45dba478bbc160ebb4ce8bd86 : RETRY_LIMIT in 46m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"88e39020f807ce87501e6f6b7024e093e0dcdce2","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-16 06:39:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"81b1ff6d0a9c41d1b092489b33462280da9a182f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 09:24:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/295896a8ed2243199cacf0b67f3aa467 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56cd91f08c9a4121a55faabe1f3cebd8 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49cb628849314e3a83b02d4acff6f2a1 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03860a4f8a80484c9ea3ac4df00e9490 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a60790ef4034ce7a43a68e21d5fef45 : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2daea0af244e0eb07d41059ba30317 : SUCCESS in 6m 59s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/d73c0863030a45629b64dd7aacfb38cf : FAILURE in 1h 48m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"54c252706bc2005944c20d1cf667dba63ac0c521","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-16 09:34:41.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"85efb11cc626d5673e1c1da974d635ac943419cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 11:30:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d39f24a76f1b46f9927a4768a3621ba7 : SUCCESS in 10m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11d6922115304ed0b5ccd8eeec4d685e : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1237efbe08bb4b88ab0c263095909372 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31d8f2aa7e784f2aaa01a01117e53b74 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee30b2d8309740ff85ca1d133092af38 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38ca580f847c423ab730f1344e53dc31 : SUCCESS in 8m 30s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/f8d329cd33274e5a962f09f6d7453dda : FAILURE in 1h 47m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7c9782e400213beca5c819f6b6fd2758ed799b8f","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-09-16 16:06:33.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"d057f8aa34fe97f369ff7d2cf558d9bd8872db43","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-18 00:54:46.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f7d4bb7ae7161703e6a0fe6c1a89ce54c0645239","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-09-18 01:02:36.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"1eb191f3728b19ee2d09e8ae5f093430d60eb10b","author":{"_account_id":30901,"name":"Masazumi ota","email":"masazumi.ota@ntt.com"},"date":"2020-09-18 01:06:43.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9d0eda6bd699535fa2e23ec217f8c5a410cdee92","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-18 01:07:21.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7efdd9c93ca3a0ce44d5a008ffa1ceceb12c2542","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2020-09-18 01:16:37.000000000","message":"Patch Set 6:\n\nI think this patch makes some functional-tests FAILURE.\nIs it clear which test will be FAILURE?\nIf YES, you can skip test as follows.\n  @ unittest.skip (\"Until BUG 1889732\")\nIf NO, this is a temporary fix and I think it isn\u0027t necessary to investigate which test will be FAILURE, so I think it\u0027s okay as it is.\nPlease tell me your opinion.","accounts_in_message":[],"_revision_number":6},{"id":"79529b165aa36cc1b92fe7c5affa92d11dd6abab","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-18 01:29:24.000000000","message":"Patch Set 6:\n\n\u003e I think this patch makes some functional-tests FAILURE.\n \u003e Is it clear which test will be FAILURE?\n \u003e If YES, you can skip test as follows.\n \u003e @ unittest.skip (\"Until BUG 1889732\")\n \u003e If NO, this is a temporary fix and I think it isn\u0027t necessary to\n \u003e investigate which test will be FAILURE, so I think it\u0027s okay as it\n \u003e is.\n \u003e Please tell me your opinion.\n\nIn my opinion, the answer is NO.\nSo we should proceed with this *temporary fix*.\n\nIt seems very difficult to investigate, because we can not immediately clarify which of \"Ubuntu version upgrade\" or \"uninstall components\" lead to failure. But this investigation in parallel is necessary .","accounts_in_message":[],"_revision_number":6},{"id":"5d7527a1d99c1252a0719334908ac341bd155b00","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-18 01:36:24.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6744548dc2b6d99053ddd5424acb79311acbc7c3","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2020-09-18 01:36:28.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"4c64e808b9639922302b4ba64065df344b5f3383","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-18 01:55:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1020358ae0efbd594a93e6149670de435b8d42bf","author":{"_account_id":27180,"name":"Keiko Kuriu","email":"kuriu.keiko@gmail.com","username":"keiko-k"},"date":"2020-09-18 01:58:35.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"aa1a6992dffcd125a152d62bd9812f686a5a1728","author":{"_account_id":30901,"name":"Masazumi ota","email":"masazumi.ota@ntt.com"},"date":"2020-09-18 01:59:40.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"bcc8dfdce2d65316b004878b30ed024c4f1a00aa","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-09-18 01:59:56.000000000","message":"Patch Set 7:\n\nAdded related URL to all comments in order to issue track.\nI also add gmann\u0027s fix URL because it is not shown in Story and Launchpad Bug.","accounts_in_message":[],"_revision_number":7},{"id":"509c5163f482397394d989be8d2f81c6bf99e84b","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-18 03:26:41.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"ba544e307f851d19c240ee943cb164dc979d6e45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 04:00:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c514fff9db3c46f9923ae92f51ac4980 : FAILURE in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0bc03a5b5494e9fab01e35b11f2e38d : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/850fd26239cb41798c5ad0d3958be4e9 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78b5af2baf904b1fb350f937d452d7f7 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/161922e4cf464f479d2f3a7b0e88bfcc : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89e3fd0c2d4a469bb01db0928ef7a9e1 : SUCCESS in 5m 21s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/fae0cd310c5845718eaaf8b9490a1b6b : FAILURE in 1h 58m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"66c0f7c306df5e60bf4186171cebf8be9204348f","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-18 04:29:56.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"3656b418f43cc00480fee4074f7816b8498eb199","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-18 04:51:51.000000000","message":"Patch Set 7: -Code-Review\n\nOops, something failure is happened in the zuul check.","accounts_in_message":[],"_revision_number":7},{"id":"a5b387bfc49140060f172b38f4d4b1be029beaaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 06:29:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13c5265e302a4b8a9ff2888e9a66bbd9 : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bd61ea7f0344c4fb84c3625bd37539f : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8bc94ae50e4fc88de12e512f8175d9 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/285820dd98f0486f867b2abb14f4bf49 : SUCCESS in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f13fd0455cf84e96b6dff8dbf399c836 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ad08294d064c889458999bd392518b : SUCCESS in 6m 29s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/ac4678db5b2c4e2e882f3d031f5cc58f : FAILURE in 1h 57m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"92e872c03930c4109bbb63c551f73dfde3aa846c","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-09-18 06:51:55.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"fddc9bed7faed81478b93b59a8e3502383728536","author":{"_account_id":31072,"name":"Hiroo Kitamura","email":"hiroo.kitamura@ntt-at.co.jp","username":"Hiroo.Kitamura"},"date":"2020-09-18 06:56:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"dc4093aacad438fd4c54608b1562ff860aa24ad1","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2020-09-18 07:06:06.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"39f46ce073a439e8d8c913ccbd33602e1ee284e8","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-18 07:26:24.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nIt looks passing again after recheck. Although there are some unknown failures increased in the result of FT, we should merge this update at once to avoid more critical issues. Thanks, takahashi-tsc!","accounts_in_message":[],"_revision_number":7},{"id":"dd6261654c103908c4afaf4a598ff5947e54c4e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:26:39.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4d3d16247970240c460ac9173a6bec4b3d0254d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:41:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0def5924ee644c32a42953c786a42dbf : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78256fde85741969862abd0cd037c74 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5eea1328a1e14e14813897a72b32c505 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/574931103c014d27bd0afc4f816ea5bf : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0670999a7f754c99bd5853b8b3aac308 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":7},{"id":"a5327f22dafd2906f1fe7a0b400714aec6fcb5b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:41:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"763bfac9671a5538df1891c0c8af1173694dcdac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:47:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b25cc7394d4137a1c8082b7c6bea0b : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"959da510b56bba982179f5f08b2cb1bca05cbc1e","revisions":{"ff7d3e1830b42e24f7b138f1bc2989c273b036fb":{"kind":"REWORK","_number":1,"created":"2020-09-15 04:14:46.000000000","uploader":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"ref":"refs/changes/65/751965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/751965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/751965/1"}}},"commit":{"parents":[{"commit":"8c6d7085532c6518ae1415fb8e4261db35ea5567","subject":"Merge \"Add introduction and architecture doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8c6d7085532c6518ae1415fb8e4261db35ea5567"}]}],"author":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-15 04:10:09.000000000","tz":540},"committer":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-15 04:14:00.000000000","tz":540},"subject":"[WIP] Temporarily remove ceilometer install","message":"[WIP] Temporarily remove ceilometer install\n\nWe temporarily remove ceilometer install to execute FT\nbecause Gnocchi cannot be installed by some error.\n\nChange-Id: I93dbd84cec252bb93257b852aa171d9ca89665c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ff7d3e1830b42e24f7b138f1bc2989c273b036fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ff7d3e1830b42e24f7b138f1bc2989c273b036fb"}]},"branch":"refs/heads/master"},"5a2ceb9103f87d70c0e806c22dffb049ba8f8fa1":{"kind":"REWORK","_number":2,"created":"2020-09-16 01:24:51.000000000","uploader":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"ref":"refs/changes/65/751965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/751965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/751965/2"}}},"commit":{"parents":[{"commit":"8c6d7085532c6518ae1415fb8e4261db35ea5567","subject":"Merge \"Add introduction and architecture doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8c6d7085532c6518ae1415fb8e4261db35ea5567"}]}],"author":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-15 04:10:09.000000000","tz":540},"committer":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-16 01:23:43.000000000","tz":540},"subject":"[WIP] Temporarily remove ceilometer install","message":"[WIP] Temporarily remove ceilometer install\n\nWe temporarily remove ceilometer install to execute FT\nbecause Gnocchi cannot be installed by some error.\n\nChange-Id: I93dbd84cec252bb93257b852aa171d9ca89665c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/5a2ceb9103f87d70c0e806c22dffb049ba8f8fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/5a2ceb9103f87d70c0e806c22dffb049ba8f8fa1"}]},"branch":"refs/heads/master"},"01ebc4e2228f37d680ed538a977e0d92b2e39a72":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-16 01:32:53.000000000","uploader":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"ref":"refs/changes/65/751965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/751965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/751965/3"}}},"commit":{"parents":[{"commit":"8c6d7085532c6518ae1415fb8e4261db35ea5567","subject":"Merge \"Add introduction and architecture doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8c6d7085532c6518ae1415fb8e4261db35ea5567"}]}],"author":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-15 04:10:09.000000000","tz":540},"committer":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-16 01:29:51.000000000","tz":540},"subject":"[WIP] Temporarily remove ceilometer install","message":"[WIP] Temporarily remove ceilometer install\n\nWe temporarily remove ceilometer install to execute FT\nbecause Gnocchi cannot be installed by some error.\n\nRelated to https://storyboard.openstack.org/#!/story/2007732\nRelated to https://storyboard.openstack.org/#!/story/2008121\nRelated-Bug: #1889732\nChange-Id: I93dbd84cec252bb93257b852aa171d9ca89665c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/01ebc4e2228f37d680ed538a977e0d92b2e39a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/01ebc4e2228f37d680ed538a977e0d92b2e39a72"}]},"branch":"refs/heads/master"},"e738dc3994eef6c277de97790a609642ee93e5f1":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-09-16 01:37:27.000000000","uploader":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"ref":"refs/changes/65/751965/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/751965/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/751965/4"}}},"commit":{"parents":[{"commit":"8c6d7085532c6518ae1415fb8e4261db35ea5567","subject":"Merge \"Add introduction and architecture doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8c6d7085532c6518ae1415fb8e4261db35ea5567"}]}],"author":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-15 04:10:09.000000000","tz":540},"committer":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-16 01:36:38.000000000","tz":540},"subject":"[WIP] Temporarily remove ceilometer install","message":"[WIP] Temporarily remove ceilometer install\n\nWe temporarily remove Ceilometer and Barbican install to execute FT\nbecause Gnocchi and Barbican cannot be installed by some error.\n\nRelated to https://storyboard.openstack.org/#!/story/2007732\nRelated to https://storyboard.openstack.org/#!/story/2008121\nRelated-Bug: #1889732\nChange-Id: I93dbd84cec252bb93257b852aa171d9ca89665c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e738dc3994eef6c277de97790a609642ee93e5f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e738dc3994eef6c277de97790a609642ee93e5f1"}]},"branch":"refs/heads/master"},"cd5ed97603daff426537453acf57972262c01070":{"kind":"REWORK","_number":5,"created":"2020-09-16 06:39:33.000000000","uploader":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"ref":"refs/changes/65/751965/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/751965/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/751965/5"}}},"commit":{"parents":[{"commit":"8c6d7085532c6518ae1415fb8e4261db35ea5567","subject":"Merge \"Add introduction and architecture doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8c6d7085532c6518ae1415fb8e4261db35ea5567"}]}],"author":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-15 04:10:09.000000000","tz":540},"committer":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-16 06:38:42.000000000","tz":540},"subject":"[WIP] Temporarily remove ceilometer install","message":"[WIP] Temporarily remove ceilometer install\n\nWe temporarily remove Ceilometer and Barbican install to execute FT\nbecause Gnocchi and Barbican cannot be installed by some error.\n\nRelated to https://storyboard.openstack.org/#!/story/2007732\nRelated to https://storyboard.openstack.org/#!/story/2008121\nRelated-Bug: #1889732\nChange-Id: I93dbd84cec252bb93257b852aa171d9ca89665c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/cd5ed97603daff426537453acf57972262c01070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/cd5ed97603daff426537453acf57972262c01070"}]},"branch":"refs/heads/master"},"ad74b3bca680717ee497e4dabb5e77f35c6f2c7c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-09-16 09:34:41.000000000","uploader":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"ref":"refs/changes/65/751965/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/751965/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/751965/6"}}},"commit":{"parents":[{"commit":"8c6d7085532c6518ae1415fb8e4261db35ea5567","subject":"Merge \"Add introduction and architecture doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8c6d7085532c6518ae1415fb8e4261db35ea5567"}]}],"author":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-15 04:10:09.000000000","tz":540},"committer":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-16 09:33:49.000000000","tz":540},"subject":"Temporarily remove ceilometer install","message":"Temporarily remove ceilometer install\n\nWe temporarily remove Ceilometer and Barbican install to execute FT\nbecause Gnocchi and Barbican cannot be installed by some error.\n\nRelated to https://storyboard.openstack.org/#!/story/2007732\nRelated to https://storyboard.openstack.org/#!/story/2008121\nRelated-Bug: #1889732\nChange-Id: I93dbd84cec252bb93257b852aa171d9ca89665c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ad74b3bca680717ee497e4dabb5e77f35c6f2c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ad74b3bca680717ee497e4dabb5e77f35c6f2c7c"}]},"branch":"refs/heads/master"},"959da510b56bba982179f5f08b2cb1bca05cbc1e":{"kind":"REWORK","_number":7,"created":"2020-09-18 01:55:37.000000000","uploader":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"ref":"refs/changes/65/751965/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/65/751965/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/65/751965/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/65/751965/7"}}},"commit":{"parents":[{"commit":"8c6d7085532c6518ae1415fb8e4261db35ea5567","subject":"Merge \"Add introduction and architecture doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8c6d7085532c6518ae1415fb8e4261db35ea5567"}]}],"author":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-15 04:10:09.000000000","tz":540},"committer":{"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","date":"2020-09-18 01:54:30.000000000","tz":540},"subject":"Temporarily remove ceilometer install","message":"Temporarily remove ceilometer install\n\nWe temporarily remove Ceilometer and Barbican install to execute FT\nbecause Gnocchi and Barbican cannot be installed by some error.\n\nRelated to https://storyboard.openstack.org/#!/story/2007732\nRelated to https://storyboard.openstack.org/#!/story/2008121\nRelated-Bug: #1889732\nChange-Id: I93dbd84cec252bb93257b852aa171d9ca89665c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/959da510b56bba982179f5f08b2cb1bca05cbc1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/959da510b56bba982179f5f08b2cb1bca05cbc1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
