)]}'
{"id":"openstack%2Ftacker~755603","triplet_id":"openstack%2Ftacker~stable%2Fvictoria~Icccaa523d33115e8b7ec0a6fd7e9c5f177d0a363","project":"openstack/tacker","branch":"stable/victoria","topic":"bug/1886124","hashtags":[],"change_id":"Icccaa523d33115e8b7ec0a6fd7e9c5f177d0a363","subject":"Running the VNF LCM instantiate using the sample vnf package fails","status":"MERGED","created":"2020-10-01 17:29:24.000000000","updated":"2020-10-02 16:23:32.000000000","submitted":"2020-10-02 16:23:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755603-1601655812223-ec8743fb","meta_rev_id":"6dcba922ca886a3a5dad054877dd7b5c674d11e2","_number":755603,"virtual_id_number":755603,"owner":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 16:23:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},{"value":2,"date":"2020-10-02 15:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2020-10-02 13:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},{"value":1,"date":"2020-10-02 15:57:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 17:29:24.000000000","updated_by":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"reviewer":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"state":"REVIEWER"},{"updated":"2020-10-02 13:44:48.000000000","updated_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2020-10-02 16:23:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff156ae72c0ca00e4c668fe0aff6701f45ed09dc","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-10-01 17:29:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21c4934693700140896251cd24ca22a043d4a4fb","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-10-01 17:43:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"6614d0565d5c3d60d153f2b2ce1c1f32a6414477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 21:04:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7166020676c43098c4bea2117a36355 : SUCCESS in 7m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f32fa4b66a349e891618be0cae3888f : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c99996636b343329dbc092b076eb895 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1623dcd560c248ff86ebe0a4237d8288 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82b521482edf4874ad54e6d029e1e3ca : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1033ffe01954f3991ce9ec7b9b3a135 : SUCCESS in 6m 16s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/a72453f5798840cc94d3d064ea6853ea : FAILURE in 1h 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcbccbc3170728069ad56ce31752fb53cac880d4","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-10-02 06:44:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"9420af30b2927b6244dea2705640905ac75995c5","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-10-02 09:24:16.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"275a52efd00467ee27ec4f294b88e5ee69cdd633","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-10-02 09:27:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"363ef2f0c4ffe7ede76fa77e2c3e519dc622c5f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:43:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92b3f847d1454409bfabb28197223e4a : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1327c69d1a064b729ba72fd89cc1b946 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6118c8ead346cabc35a6ecd251ea9e : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3af29fac6f48bc8429a88a7a626090 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e18ddfe56ed04cd3b3ca2c3818244b0d : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d1e9607ffa4751bea9af4bddbe1b05 : SUCCESS in 6m 50s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/4fe4c953623b4168940f6a50bc991c15 : FAILURE in 1h 45m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"261c9b41f669612dc0a22bde0895c2875cc8f973","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-10-02 13:44:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d99b07e3c1c7b02a883bef8b5df37711879400ba","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-10-02 15:57:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c346c1ba392ae962d2af941b9a08bb10e26453b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 15:57:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b44271a7d64d62cddad22df4ea81bbb0e8931c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:23:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/490ba20a6c054d2d98f0a25fd1c99f24 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e19cb50ddbf45cb98dddcd1df07ccd9 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/601e0651f5d64c5c9fde7181a45dc70c : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d13734f178954959b8a66c795be832d1 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a9b5fcfa0384616a98cd1972ed3294e : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":2},{"id":"15f146bee3cf1847284ba871cebff9274fb4cfef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:23:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"96ca43483deffd582666f8c717c9671e443352e6","revisions":{"16540339abd67c3b0b649e9ffd398d898ef9d0ef":{"kind":"REWORK","_number":1,"created":"2020-10-01 17:29:24.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/03/755603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/03/755603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/03/755603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/03/755603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/03/755603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/03/755603/1"}}},"commit":{"parents":[{"commit":"ca0bb75de9fb6d5a124818d410bee06ea416b6de","subject":"Remove obsolete files and the unused coding style modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ca0bb75de9fb6d5a124818d410bee06ea416b6de"}]}],"author":{"name":"Naoaki Horie","email":"horie.naoaki@ntt-tx.co.jp","date":"2020-07-07 09:30:33.000000000","tz":0},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2020-10-01 17:25:44.000000000","tz":540},"subject":"Running the VNF LCM instantiate using the sample vnf package fails","message":"Running the VNF LCM instantiate using the sample vnf package fails\n\nThe CSAR file for VNF package registered as a sample fails to run\nthe VNF LCM instantiate. The cause of this issue is typo and invalid\ndefinition for the VNFD. I think that the csar zip file for uploaded VNF\npackage should be able to run the VNF LCM instantiate normally.\nTherefore, fix typo and invalid definition for the VNFD in the CSAR file.\n\nAlso, fix to the following settings that has a dependent on environment\nbecause the current VNFD can only be run in the limited environment.\n\n* Change hash algorithm from sha-256 to sha-512 according\n  the Glance default hash algorithm\n* Change from 8192 MB to 256 MB in the `min_ram` size of\n  the VirtualStorage to work even in a low specification environment\n* Delete `CP1.properties.vnic_type` to work even under a virtual\n  environment\n\nChange-Id: Icccaa523d33115e8b7ec0a6fd7e9c5f177d0a363\nCloses-Bug: #1886124\n(cherry picked from commit b397d01a3e4fb77809c8c48e78eebe8cf2a3fcf4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/16540339abd67c3b0b649e9ffd398d898ef9d0ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/16540339abd67c3b0b649e9ffd398d898ef9d0ef"}]},"branch":"refs/heads/stable/victoria"},"96ca43483deffd582666f8c717c9671e443352e6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-02 09:27:24.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/03/755603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/03/755603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/03/755603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/03/755603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/03/755603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/03/755603/2"}}},"commit":{"parents":[{"commit":"3962c23ab07ebdfd67718c21583196a857a436e2","subject":"Implementation Fenix plugin in Tacker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/3962c23ab07ebdfd67718c21583196a857a436e2"}]}],"author":{"name":"Naoaki Horie","email":"horie.naoaki@ntt-tx.co.jp","date":"2020-07-07 09:30:33.000000000","tz":0},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2020-10-02 09:26:01.000000000","tz":540},"subject":"Running the VNF LCM instantiate using the sample vnf package fails","message":"Running the VNF LCM instantiate using the sample vnf package fails\n\nThe CSAR file for VNF package registered as a sample fails to run\nthe VNF LCM instantiate. The cause of this issue is typo and invalid\ndefinition for the VNFD. I think that the csar zip file for uploaded VNF\npackage should be able to run the VNF LCM instantiate normally.\nTherefore, fix typo and invalid definition for the VNFD in the CSAR file.\n\nAlso, fix to the following settings that has a dependent on environment\nbecause the current VNFD can only be run in the limited environment.\n\n* Change hash algorithm from sha-256 to sha-512 according\n  the Glance default hash algorithm\n* Change from 8192 MB to 256 MB in the `min_ram` size of\n  the VirtualStorage to work even in a low specification environment\n* Delete `CP1.properties.vnic_type` to work even under a virtual\n  environment\n\nChange-Id: Icccaa523d33115e8b7ec0a6fd7e9c5f177d0a363\nCloses-Bug: #1886124\n(cherry picked from commit b397d01a3e4fb77809c8c48e78eebe8cf2a3fcf4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/96ca43483deffd582666f8c717c9671e443352e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/96ca43483deffd582666f8c717c9671e443352e6"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
