)]}'
{"id":"openstack%2Ftacker~759608","triplet_id":"openstack%2Ftacker~master~I218cb0f176762d766fad5fd1f60fdfa3da628f25","project":"openstack/tacker","branch":"master","topic":"minor-fixes-and-improvements","hashtags":[],"change_id":"I218cb0f176762d766fad5fd1f60fdfa3da628f25","subject":"Minor fixes and improvements","status":"ABANDONED","created":"2020-10-26 05:15:39.000000000","updated":"2020-10-29 04:00:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"afba5097c0d099d662600725e08629730aa769e4","_number":759608,"virtual_id_number":759608,"owner":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"date":"2020-10-26 08:55:38.000000000","_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-26 07:21:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"date":"2020-10-26 05:26:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-26 05:18:51.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-10-26 05:18:57.000000000","updated_by":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"reviewer":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"state":"REVIEWER"},{"updated":"2020-10-26 07:21:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7326312c0e7bf3e11756df4b2e5f365aaf11bbbb","author":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"date":"2020-10-26 05:15:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddc3bb45d7bf0583f973be8f1d850a9be5a42988","author":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"date":"2020-10-26 05:21:30.000000000","message":"Patch Set 1:\n\nNote that we need not to wait until the change 756703 is finally accepted and merged, even though I have been looking into the FT results with it being cherry-picked in 758954.","accounts_in_message":[],"_revision_number":1},{"id":"78386f1643a7cc5e26e02cb161f27b7e6158fb2b","author":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"date":"2020-10-26 05:26:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"147734e9a7af200b46d16451e1309be47ca47e14","author":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"date":"2020-10-26 05:27:18.000000000","message":"Patch Set 2:\n\nPatchSet1 -\u003e PatchSet2: just fixed the author email.","accounts_in_message":[],"_revision_number":2},{"id":"c42202972c740eb9a9c90eb7bbde17ab9cd5a3ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 07:21:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a20df6bcf4624312a89bd2e81326f818 : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27cb0f088d7847a9840ae2c06701043e : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5d10421e9394e24b8bab3049a495d59 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd2a72a6b5194b18a301eed2e8db5547 : SUCCESS in 12m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06d2ec9b65824bbcb7a1de99c1f5777e : SUCCESS in 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09bc1d10ab094d318bd0cab53ba168ef : SUCCESS in 6m 04s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/296fa289f31f40379ce7d923d720ff3d : FAILURE in 1h 49m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e317a7630a1e321291a134b8597a5a18e72bfa27","author":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"date":"2020-10-26 08:55:38.000000000","message":"Patch Set 2:\n\nCurrent situation:\n* master(\u003d756602/5): 15 failures\n* 756703/8(rebased to master): TIMED_OUT\n* 759608/2(rebased to master): 12 failures\n* 758954/8(rebased to master): 3 failures\n\nNow,\n* 758954/8 \u003d\u003d 756703/8 + 759608/2(this patch).\n* As I commented above, this change (759608/2) can be merged before 756703/8 and vice versa.\n\nSo please don\u0027t be deterred by the failure count 12.","accounts_in_message":[],"_revision_number":2},{"id":"48e40e7b2fc428ed8fbd733ba6ec6f0283b17181","author":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"date":"2020-10-29 04:00:21.000000000","message":"Abandoned\n\nsplit this into multiple patches.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"222198a1d97706be67d8d3fe93e2c3755abb460e","revisions":{"8dd46b87fdb6712770ebe3c3507c4358e5e87351":{"kind":"REWORK","_number":1,"created":"2020-10-26 05:15:39.000000000","uploader":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"ref":"refs/changes/08/759608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/08/759608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/08/759608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/08/759608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/08/759608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/08/759608/1"}}},"commit":{"parents":[{"commit":"313eb776c4af53fd857ef4a0375448c2b32ffcca","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/313eb776c4af53fd857ef4a0375448c2b32ffcca"}]}],"author":{"name":"Koichiro Den","email":"den@klaipeden.com","date":"2020-10-19 06:04:30.000000000","tz":540},"committer":{"name":"Koichiro Den","email":"den@valinux.co.jp","date":"2020-10-26 05:14:48.000000000","tz":540},"subject":"Minor fixes and improvements","message":"Minor fixes and improvements\n\nLooking to eliminate FT errors\n\n* objects:\n  - drop redundant __import__\n  - VnfLcmSubscription: fix typo (operationTypese -\u003e operationTypes)\n  - VnfDeploymentFlavour: set default instantiation_levels to empty list\n* tacker-conductor-server:\n  - fix logging error\n  - fix missing vnfc resource updating on vnf respawn via SOL003-style\n    Heal API call\n* FT codebase:\n  - VnfPackageTest: set total sizes to its real full size\n  - Keystone: introduce client initialization retry\n* tosca/utils:\n  - fix a python3 issue\n* plugins:\n  - fenix: avoid an erroneous heat API call\n\nChange-Id: I218cb0f176762d766fad5fd1f60fdfa3da628f25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8dd46b87fdb6712770ebe3c3507c4358e5e87351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8dd46b87fdb6712770ebe3c3507c4358e5e87351"}]},"branch":"refs/heads/master"},"222198a1d97706be67d8d3fe93e2c3755abb460e":{"kind":"NO_CHANGE","_number":2,"created":"2020-10-26 05:26:03.000000000","uploader":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"ref":"refs/changes/08/759608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/08/759608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/08/759608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/08/759608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/08/759608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/08/759608/2"}}},"commit":{"parents":[{"commit":"313eb776c4af53fd857ef4a0375448c2b32ffcca","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/313eb776c4af53fd857ef4a0375448c2b32ffcca"}]}],"author":{"name":"Koichiro Den","email":"den@valinux.co.jp","date":"2020-10-19 06:04:30.000000000","tz":540},"committer":{"name":"Koichiro Den","email":"den@valinux.co.jp","date":"2020-10-26 05:25:38.000000000","tz":540},"subject":"Minor fixes and improvements","message":"Minor fixes and improvements\n\nLooking to eliminate FT errors\n\n* objects:\n  - drop redundant __import__\n  - VnfLcmSubscription: fix typo (operationTypese -\u003e operationTypes)\n  - VnfDeploymentFlavour: set default instantiation_levels to empty list\n* tacker-conductor-server:\n  - fix logging error\n  - fix missing vnfc resource updating on vnf respawn via SOL003-style\n    Heal API call\n* FT codebase:\n  - VnfPackageTest: set total sizes to its real full size\n  - Keystone: introduce client initialization retry\n* tosca/utils:\n  - fix a python3 issue\n* plugins:\n  - fenix: avoid an erroneous heat API call\n\nChange-Id: I218cb0f176762d766fad5fd1f60fdfa3da628f25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/222198a1d97706be67d8d3fe93e2c3755abb460e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/222198a1d97706be67d8d3fe93e2c3755abb460e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
