)]}'
{"id":"openstack%2Ftacker~760278","triplet_id":"openstack%2Ftacker~master~I940fdc289701023a3f37abb01c3f37dd312f2f7f","project":"openstack/tacker","branch":"master","topic":"fix/minor-cleanups","hashtags":[],"change_id":"I940fdc289701023a3f37abb01c3f37dd312f2f7f","subject":"Minor fixes and clean-ups","status":"MERGED","created":"2020-10-29 03:51:42.000000000","updated":"2020-10-30 15:44:24.000000000","submitted":"2020-10-30 15:42:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760278-1604072562144-d96e2c5f","meta_rev_id":"7159ba470a6d1669c045e0d4e53a238ec18f0dbf","_number":760278,"virtual_id_number":760278,"owner":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-30 15:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":2,"date":"2020-10-30 01:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2020-10-30 11:49:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-30 02:08:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2020-10-30 15:20:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"all":[{"value":0,"date":"2020-10-29 04:22:40.000000000","_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2020-10-30 15:20:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 04:22:40.000000000","updated_by":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"reviewer":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"state":"REVIEWER"},{"updated":"2020-10-30 01:40:36.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-10-30 02:08:55.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-10-30 11:49:00.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2020-10-30 15:20:30.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2020-10-30 15:42:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac66f98544ca6f3c62e4a9b4b6e470bf0cf5174a","author":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"date":"2020-10-29 03:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2857db26dbf2b7207dc67db63b0475ab45a93b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 06:04:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77d42f54d87a450bbcacf93cbba2ebb0 : SUCCESS in 7m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff7312b01661401ba7e1bdba2538f424 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb9a6fdb1a754d21b2d519349f6b8b0c : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c07fb6937c244c6fb1140159492092ed : SUCCESS in 13m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8532781dcb23465d84b4e1b8fdbc0b90 : SUCCESS in 13m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7feb2a9e874c45963e7f78764884b5 : SUCCESS in 6m 29s\n- tacker-functional-devstack-multinode-python3 https://zuul.opendev.org/t/openstack/build/7d6268b33f5942479bf5130332ebd14e : FAILURE in 1h 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"784a32858992ece698d3b65a9469388e6d69fe06","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-10-30 01:40:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d1078ded786e5591e8ae001e6214662591d5dfb","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-30 02:08:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"895714e066fe165c4f72f8d802cc508b3c143ab8","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-10-30 11:49:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ccb6a2d7a352ef57b0f1fa3f94968006cee2621f","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2020-10-30 15:20:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de561cb1df362353ec64a8c20d55abcb56919b72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 15:20:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b44d37e71a959d81a3bd88288ed7386ecd2d824","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 15:42:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3823afd41dae4ffa923b4d3c78d2ff8b : SUCCESS in 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3f354fd2aaa4b10a166916bca657982 : SUCCESS in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34e38cead574406db065f93861f26a10 : SUCCESS in 12m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/971e081e3c2a41dab756d2f320e60821 : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0040a3fd8342dc928494556550f164 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"68427c4fd8a3fd086c2fa2de5bb5cbbddf311640","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 15:42:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a2ba9218315fe79719aeb0f4fb20cc48338b1a70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 15:44:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9432ee732c4d1db52aa38dcff5e269 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"180ff8ae4ac5f4f6ca130a11ad875c9e7ae6d7f9","revisions":{"180ff8ae4ac5f4f6ca130a11ad875c9e7ae6d7f9":{"kind":"REWORK","_number":1,"created":"2020-10-29 03:51:42.000000000","uploader":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"ref":"refs/changes/78/760278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/78/760278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/78/760278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/78/760278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/78/760278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/78/760278/1"}}},"commit":{"parents":[{"commit":"1a98ddf42beadc8cd803485332e03948c92921ee","subject":"Merge \"Fix warnings for deprecated JSON policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1a98ddf42beadc8cd803485332e03948c92921ee"}]}],"author":{"name":"Koichiro Den","email":"den@valinux.co.jp","date":"2020-10-29 03:47:34.000000000","tz":540},"committer":{"name":"Koichiro Den","email":"den@valinux.co.jp","date":"2020-10-29 03:47:34.000000000","tz":540},"subject":"Minor fixes and clean-ups","message":"Minor fixes and clean-ups\n\n* eliminate redundant __import__\n* fix typo\n* fix python3 issue\n* FT: set total sizes to its real full size\n\nRelated-Bug: #1886213\nChange-Id: I940fdc289701023a3f37abb01c3f37dd312f2f7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/180ff8ae4ac5f4f6ca130a11ad875c9e7ae6d7f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/180ff8ae4ac5f4f6ca130a11ad875c9e7ae6d7f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
