)]}'
{"id":"openstack%2Ftacker~760502","triplet_id":"openstack%2Ftacker~master~I2543fa0610bc421341ba46d166cf510e34e7971a","project":"openstack/tacker","branch":"master","topic":"revise-doc","hashtags":[],"change_id":"I2543fa0610bc421341ba46d166cf510e34e7971a","subject":"Add docs of Instantiate CNF and Artifacts","status":"MERGED","created":"2020-10-30 06:09:59.000000000","updated":"2020-11-11 15:01:28.000000000","submitted":"2020-11-11 14:59:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760502-1605106770699-b3c3e99b","meta_rev_id":"ed83ed3ed13499cc52ddbf349c119109385ca71b","_number":760502,"virtual_id_number":760502,"owner":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-11 14:59:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 05:16:43.000000000","_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},{"value":2,"date":"2020-11-11 14:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2020-11-11 12:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},{"value":1,"date":"2020-11-11 14:44:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-02 15:47:21.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2020-11-05 04:58:03.000000000","updated_by":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"reviewer":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"state":"REVIEWER"},{"updated":"2020-11-05 06:22:16.000000000","updated_by":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"reviewer":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"state":"REVIEWER"},{"updated":"2020-11-11 12:09:08.000000000","updated_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2020-11-11 14:44:56.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-11-11 14:59:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a85b97b8ce80088ea1955cb9d9038a8ad01bb0c","author":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"date":"2020-10-30 06:09:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edaefeccf6c54c4ea6eec3b596b459bb2341a415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 06:41:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a987b6d68340518ba041e6530bc687 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/358419be3bc949a79370188be0b29732 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2fec9be29f50285fdcd560110db7f07f74b8ca6f","author":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"date":"2020-11-01 22:49:06.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nCurrently `Topic` is empty. so I think it\u0027s better to describe something in `Topic`.","accounts_in_message":[],"_revision_number":1},{"id":"777a8d2c9990f512c1abc1b618b228be077b4b62","author":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"date":"2020-11-02 05:06:20.000000000","message":"Topic set to revise-doc","accounts_in_message":[],"_revision_number":1},{"id":"83b902d3d663b74aa3e5c69f94ebacffd01b4e66","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2020-11-02 15:47:21.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3b70f69daa9d93fc5d25f929df74182d259e6ca6","author":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"date":"2020-11-05 01:23:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d13dc7cdb44d7ab3187a06c0fab0e4ec7da8c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 01:30:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f96a1ed57bda49b79683755e06ff628d : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd02c7f6156d48bba00cfc903178fae4 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"8db7bc551cad5ab4b5e560fd3e0258d93d6a23cb","author":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"date":"2020-11-05 02:18:39.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"27b75c2e25dc0838ed56109a42c97cc6d7348180","author":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"date":"2020-11-05 04:58:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9ce5ff54176dfe4aea5addc7ffba71a285995df2","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2020-11-05 06:22:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1cbdda0ce037ed84b5f8f637dd6a286b8dd4f9f4","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-11-05 17:21:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84da123b63bb17a241a7aad618327ea562faf694","author":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"date":"2020-11-11 05:03:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b123d6800be26162885beedcfc9953a0f595b8fe","author":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"date":"2020-11-11 05:16:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3b93841e1c673b241359fd37648e5ac0114dcd7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 05:28:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/942d37a6d36e4e349df3e3c49854cab2 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb4f339ab8547358c6e1772b9de9db5 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":3},{"id":"dd415cb4c62e1df74743cb4024b99c3614cb8984","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2020-11-11 12:09:08.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks! LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"893a11aa30c459e2060ddfae68678a39bdc8ead7","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-11-11 14:44:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d7add4eefccbb5ba0cc7b6c3aa63f256c8d9c029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 14:45:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4d4f7818040b06be9091b7f12f20690faf0d5102","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 14:59:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0b587af7d79f57e554ddeacfbab9c1c7b8eb3881","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 14:59:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d71e2acbd741ab8dd1619f618dd1b9 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce27418e980f462f853d1d6479c61e54 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":3},{"id":"ed83ed3ed13499cc52ddbf349c119109385ca71b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 15:01:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b23794b60143129c76f88a5b996b16 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"531803dbd89f86bf34649875de13da975f9ec16b","revisions":{"7c1e206811067a2e73f5d24dc754b9ac015c9e9a":{"kind":"REWORK","_number":1,"created":"2020-10-30 06:09:59.000000000","uploader":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"ref":"refs/changes/02/760502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/02/760502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/02/760502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/02/760502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/02/760502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/02/760502/1"}}},"commit":{"parents":[{"commit":"1a98ddf42beadc8cd803485332e03948c92921ee","subject":"Merge \"Fix warnings for deprecated JSON policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1a98ddf42beadc8cd803485332e03948c92921ee"}]}],"author":{"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","date":"2020-10-30 06:05:04.000000000","tz":540},"committer":{"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","date":"2020-10-30 06:09:47.000000000","tz":540},"subject":"Add docs of Instantiate CNF and Artifacts","message":"Add docs of Instantiate CNF and Artifacts\n\nUsage guides for how to deploy ETSI NFV-SOL containerized VNF\nis added in user docs. Command-Line Help for Fetch VNF Package\nArtifacts is added in cli docs.\n\nChange-Id: I2543fa0610bc421341ba46d166cf510e34e7971a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/7c1e206811067a2e73f5d24dc754b9ac015c9e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/7c1e206811067a2e73f5d24dc754b9ac015c9e9a"}]},"branch":"refs/heads/master"},"90b56da822423df4ef95ab586c7c9052ce9f1598":{"kind":"REWORK","_number":2,"created":"2020-11-05 01:23:13.000000000","uploader":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"ref":"refs/changes/02/760502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/02/760502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/02/760502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/02/760502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/02/760502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/02/760502/2"}}},"commit":{"parents":[{"commit":"1a98ddf42beadc8cd803485332e03948c92921ee","subject":"Merge \"Fix warnings for deprecated JSON policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1a98ddf42beadc8cd803485332e03948c92921ee"}]}],"author":{"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","date":"2020-10-30 06:05:04.000000000","tz":540},"committer":{"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","date":"2020-11-05 01:22:16.000000000","tz":540},"subject":"Add docs of Instantiate CNF and Artifacts","message":"Add docs of Instantiate CNF and Artifacts\n\nUsage guides for how to deploy ETSI NFV-SOL containerized VNF\nis added in user docs. Command-Line Help for Fetch VNF Package\nArtifacts is added in cli docs.\n\nChange-Id: I2543fa0610bc421341ba46d166cf510e34e7971a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/90b56da822423df4ef95ab586c7c9052ce9f1598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/90b56da822423df4ef95ab586c7c9052ce9f1598"}]},"branch":"refs/heads/master"},"531803dbd89f86bf34649875de13da975f9ec16b":{"kind":"REWORK","_number":3,"created":"2020-11-11 05:03:22.000000000","uploader":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"ref":"refs/changes/02/760502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/02/760502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/02/760502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/02/760502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/02/760502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/02/760502/3"}}},"commit":{"parents":[{"commit":"1a98ddf42beadc8cd803485332e03948c92921ee","subject":"Merge \"Fix warnings for deprecated JSON policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1a98ddf42beadc8cd803485332e03948c92921ee"}]}],"author":{"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","date":"2020-10-30 06:05:04.000000000","tz":540},"committer":{"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","date":"2020-11-11 05:00:30.000000000","tz":540},"subject":"Add docs of Instantiate CNF and Artifacts","message":"Add docs of Instantiate CNF and Artifacts\n\nUsage guides for how to deploy ETSI NFV-SOL containerized VNF\nis added in user docs. Command-Line Help for Fetch VNF Package\nArtifacts is added in cli docs.\n\nChange-Id: I2543fa0610bc421341ba46d166cf510e34e7971a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/531803dbd89f86bf34649875de13da975f9ec16b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/531803dbd89f86bf34649875de13da975f9ec16b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
