)]}'
{"id":"openstack%2Ftacker~773286","triplet_id":"openstack%2Ftacker~master~I6ffdadc196cf2ddcec3625c6bd99b5133e37af4b","project":"openstack/tacker","branch":"master","topic":"bug/1913978","hashtags":[],"change_id":"I6ffdadc196cf2ddcec3625c6bd99b5133e37af4b","subject":"Close HOT template files after read","status":"MERGED","created":"2021-02-01 05:16:21.000000000","updated":"2021-02-09 02:11:20.000000000","submitted":"2021-02-09 02:11:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773286-bug/1913978","meta_rev_id":"336a7b6bc193cee858ce21a39cb3c4a2250e7556","_number":773286,"virtual_id_number":773286,"owner":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 02:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2021-02-05 00:52:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2021-02-03 04:36:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":1,"date":"2021-02-02 12:32:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-09 01:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-09 01:52:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 07:37:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-02 00:44:51.000000000","updated_by":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"reviewer":{"_account_id":32395,"name":"Koichiro Den","email":"koichiro.den@gmail.com","username":"kden"},"state":"REVIEWER"},{"updated":"2021-02-02 00:44:51.000000000","updated_by":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"reviewer":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"state":"REVIEWER"},{"updated":"2021-02-02 00:44:51.000000000","updated_by":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2021-02-03 04:36:22.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2021-02-09 01:52:24.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"}],"messages":[{"id":"8ffd2e5885f67113abc70d1aa39346328891399a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2021-02-01 05:16:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b5ff54d797935b3928019fba30548b589cade4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 07:37:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91e5af7889804d36a5e6b13c7326f0f2 : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66f48b4c488f4a72bf3ecd5781affc87 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17cf4cdf20ba40f7b34b383a79fede94 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4235480ef39f4537a88d67cb693f05ef : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31ca45aecedd4735b05d2651b8dacfcb : SUCCESS in 13m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf5b7bcc0a5b40f6a1ad234ece443b5f : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a22411678d4d91a75f1cfa9667f36b : SUCCESS in 6m 40s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/e203f1538d8047d3891eeec9c4ef6b22 : SUCCESS in 2h 14m 08s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/0788c6849fcf4f97b47c4d352c09ce57 : SUCCESS in 1h 47m 39s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/0520b73cfb46472dbae97dc30d9f587e : FAILURE in 1h 04m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9756ad9b393aa340bf6e01c6670f985e9fb2ec25","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2021-02-01 21:57:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3ea3b0bcbe7905313767991b5025711f6369c9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 23:55:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/884dc8e721dd4f0f9a5be2f4278b160c : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ca68805745f43b18971a16ae0907d8e : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed538aa13b948129181338dabda212d : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1922396daff94c2c9a01c67bdac88256 : SUCCESS in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf901dbd82d4401396ccc1f4fcca1fc2 : SUCCESS in 15m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c86d304b20cb4fd8b92f7d2569043d72 : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e523024d91456ea230d6dfb97dff47 : SUCCESS in 6m 22s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/16b5cc2de2d04288883fc4bc6dc6a31c : SUCCESS in 1h 50m 34s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/13379a95ed3e462eb75749c71ef13ca6 : SUCCESS in 1h 42m 51s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/6c74c902750e4cc9a9bd54a0d9e1c277 : SUCCESS in 1h 02m 23s","accounts_in_message":[],"_revision_number":1},{"id":"59c38ea30e29c602251b49e21dcc8051fd095a80","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2021-02-02 12:32:00.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":1},{"id":"b7b687cf6c4456652e078cec14fd32f9044711ee","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-02-03 04:36:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e854dff944ea496aca5473e78f07196b33ff755b","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-02-05 00:52:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"437f59896483b5148723607d922c97f6200a853c","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2021-02-09 01:52:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cea366d2a19cb13fa88c9a752045d5df21b2ff9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 01:52:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d150062114a14692428185dc44b7001f144d6db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 02:11:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e67d0da3592747bc9adb95f99c2b2daa : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a94374024214a9fa0cc6a366e1c77fc : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2cac6089abf442ea1804b5510391a6d : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/429284c4c9bf4ad5a4d0d114d9257311 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5b029c7f944c189a9cdef5198365d0 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"336a7b6bc193cee858ce21a39cb3c4a2250e7556","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 02:11:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54aca142b09b8249531ec558e119e9b7e0db035f","revisions":{"54aca142b09b8249531ec558e119e9b7e0db035f":{"kind":"REWORK","_number":1,"created":"2021-02-01 05:16:21.000000000","uploader":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"ref":"refs/changes/86/773286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/86/773286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/86/773286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/86/773286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/86/773286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/86/773286/1"}}},"commit":{"parents":[{"commit":"a190526e47f23312049b06ef3991b9648ab11962","subject":"Merge \"Add new FT set for tests with separated NFVO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/a190526e47f23312049b06ef3991b9648ab11962"}]}],"author":{"name":"Itsuro Oda","email":"oda@valinux.co.jp","date":"2021-02-01 04:39:13.000000000","tz":540},"committer":{"name":"Itsuro Oda","email":"oda@valinux.co.jp","date":"2021-02-01 05:08:12.000000000","tz":540},"subject":"Close HOT template files after read","message":"Close HOT template files after read\n\nPreviously get_base_nest_hot_dict() used by tacker-conductor\nopens and reads HOT template files but does not close them.\nAs a result, tacker-conductor will reach the limit of open\nfile descriptor while it is running long time.\n\nThis patch fixes to close HOT template files after read\nthem.\n\nThis patch makes some clean-ups too.\n* use existing method to make path construction and make it easy to\n  understand.\n* remove _get_base_hot_dict() because it was replaced to\n  get_base_nest_hot_dict() and is not used anywhere now.\n  And it is considerd that it will not be used any more.\n* remove meaningless array to tupple translation.\n* same bug fix and clean-ups are applied in the same file.\n\nCloses-Bug: #1913978\n\nChange-Id: I6ffdadc196cf2ddcec3625c6bd99b5133e37af4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/54aca142b09b8249531ec558e119e9b7e0db035f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/54aca142b09b8249531ec558e119e9b7e0db035f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
