)]}'
{"id":"openstack%2Ftacker~777275","triplet_id":"openstack%2Ftacker~master~Ia1bb10daf426820b44487c85fbffd3b7ff5c0f39","project":"openstack/tacker","branch":"master","topic":"fix-ft-error","hashtags":[],"change_id":"Ia1bb10daf426820b44487c85fbffd3b7ff5c0f39","subject":"Fix FT error for legacy anti-affinity test","status":"MERGED","created":"2021-02-24 02:36:34.000000000","updated":"2021-02-25 11:00:03.000000000","submitted":"2021-02-25 10:54:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777275-fix-ft-error","meta_rev_id":"475096c3e3d36309d19ee8a1b148c89f87a8f126","_number":777275,"virtual_id_number":777275,"owner":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-25 10:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2021-02-25 07:56:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2021-02-25 09:40:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":1,"date":"2021-02-25 01:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-25 01:10:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":2,"date":"2021-02-25 10:34:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":1,"date":"2021-02-25 10:34:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 04:51:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-25 00:08:17.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2021-02-25 00:08:17.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2021-02-25 00:08:17.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"state":"REVIEWER"},{"updated":"2021-02-25 00:08:17.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2021-02-25 01:10:57.000000000","updated_by":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"reviewer":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"state":"REVIEWER"},{"updated":"2021-02-25 09:40:17.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"}],"messages":[{"id":"c42507ef405e39efeaf53cec4d6317540952b537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-02-24 02:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3dfc9f5d33d7d0a3c244e700849731b1d6586f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:51:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1035df4a08041849226b54d492c1c2d : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cf263a1652a43909b670a092e45ca44 : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e022bd2bba374dc3bd64e1c9571eb0a7 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7246ce46518142ad8f7fd81051936150 : SUCCESS in 13m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe117c3736db41afbb5fa9288e592452 : SUCCESS in 12m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f61209f958b4824824b3db633644f57 : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae37439e1024035a9903977fac3c02a : SUCCESS in 6m 44s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/eabccda079ed41ab9c6736c2316fa350 : SUCCESS in 1h 45m 45s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/5945b6d2e9ac434abb0185c87043c4b0 : SUCCESS in 2h 08m 23s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/8f017ba473c2420e9be0ed533b6f86b3 : SUCCESS in 1h 18m 25s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/b14b0b8ce90e4bfd8bec40d6d7a6a98c : SUCCESS in 1h 08m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f014c7a6b0e40e6e2569efcc1ec3c21d41443ec2","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-02-24 04:53:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a0797a2a1a144edc3d37796675cd7449f33ca01d","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-02-24 05:42:31.000000000","message":"Patch Set 1:\n\nI will recheck a few times to see if the Legacy FT anti-afinity test fails.","accounts_in_message":[],"_revision_number":1},{"id":"ae77fe59ccb39a35a15e0ea2c64ff00212a12b77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 06:52:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cdd2296fe364e199a3b5063a51eb502 : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b20d2354f054c6d8f30f95a86987d15 : SUCCESS in 13m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2415b8ee637e445a85028f11ff593a2a : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a91fcffb2594672a112d5e598239335 : SUCCESS in 11m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e5eb353a5d44b2b9ef221ad3107297c : SUCCESS in 11m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/213bac59156b48caa8dee55f7fafb478 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c775c5ff4442188c7f0a1fc1fba1eb : SUCCESS in 6m 59s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/462f16365cd54e868d481bd18831834f : SUCCESS in 1h 52m 46s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/7785d85c34dd400391d77e4f7bbcb7c8 : SUCCESS in 1h 54m 21s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/a8a89f47bcc749f1b59aca86ec9aa645 : SUCCESS in 1h 03m 37s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/66edcd632ee64066a2c1b4bb990af0e7 : SUCCESS in 1h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf17a80af6e80c7ceb9cc8aeeb206b73d19b1906","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-02-24 07:19:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"396e36c0995f7307f2f4a22ee29ae069598ef2b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-02-24 07:38:50.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a4dfffa47de125208cca2f997f94b16042dc8eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-02-24 07:39:32.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5ff5800f5ec091cf8515888cbd014c2163db5719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 09:59:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc7fee7e570146ebb98f95497cd0fa28 : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d1892721512486daac5a153cef2d09f : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d03a070d374cdb98f02f401632b569 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67e970c9432a476c9a786624e4977654 : SUCCESS in 10m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23e4dbbec16f4d07937b279eb9433dce : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53314a2b348b483da16fb25045802ce9 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69bbfb511514474e971ae8b28b3b71a6 : SUCCESS in 6m 34s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/b2272d32aec84590b2cd02d4c38a6a24 : SUCCESS in 1h 37m 45s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/1df680e6ca7c4aaaacd7873c7f13aecd : SUCCESS in 1h 56m 59s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/1a1d94e164cd4ef3ad8078600482d268 : SUCCESS in 1h 02m 00s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/6ddd2cc977ba4373bf60ec4ed1623444 : SUCCESS in 1h 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09bbf7120c740466600f4d44d4225db089f33030","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-02-24 10:03:18.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"cbe816b15999a1ad39e5b917a13f40f11fe9a884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 12:25:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a608f4096ea34e9d998ef2f8eda3fd77 : SUCCESS in 11m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf407aeacfc3453bbc57937c47d281f7 : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae115313b5bc4714a11630eb3409d181 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07b89c8c58634b75acec8ee610de765d : SUCCESS in 12m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3191bfd23394458c885638ae74a7805f : SUCCESS in 11m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55a10db2b4e44e7cac8649aa743b2fdf : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64dd410fc22f424992b0e60f3fe70469 : SUCCESS in 7m 20s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/98b5f33e21bf457ab3c39b725c1933a5 : SUCCESS in 1h 50m 27s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/6c80c26e9fcf4e89a22cbe5f56fd0bbd : FAILURE in 2h 14m 21s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/ea31751134bf4e4690247dc236141359 : SUCCESS in 1h 06m 56s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/3dc82f6c30124835a510d77a22088f2e : SUCCESS in 1h 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"383b7a1c36aabea64d07825a46d5fcd0cca2c020","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2021-02-24 14:00:53.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"07a07519fb988d962727d1c52255e74d729192ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:33:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d1bdcd950ca4372a41092e46b254731 : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86bb21bc826744c3ba08e7b4cf971b09 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5716f030994ed6881898c84c2a7806 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ccab831fec24bea9dfc8b8e55762e98 : SUCCESS in 13m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2810c0a41319426ca29d8a2d03b53c15 : SUCCESS in 11m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b096cbb7fce14410a6a9717b38db6019 : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91302653ec4c40a1936944f5573c79f0 : SUCCESS in 7m 48s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/280728a0179d46e09cfd303d60a92371 : SUCCESS in 1h 48m 30s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/44be895fb18c41b6a916ca3fc97e93b8 : SUCCESS in 1h 52m 39s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/2415bba74bf14e6eb373daf9fb2dcd1d : SUCCESS in 52m 57s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/0c50a544d9c94a7480d0ecc7c8ddf962 : SUCCESS in 1h 19m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"87fadc735ed4dc001fc79d9bb7ec85aff630ad5d","author":{"_account_id":27880,"name":"Yoshito Ito","email":"yoshito.itou.dr@hco.ntt.co.jp","username":"yoshito-ito"},"date":"2021-02-25 01:03:04.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM. An error is observed at the third time FT but it seems another cause.","accounts_in_message":[],"_revision_number":4},{"id":"849edc7f9701292d8ad39fad54661a839b7802d7","author":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"date":"2021-02-25 01:10:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ee34140c89e45c9b516578a4a2fbc541a159d3a2","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-02-25 07:56:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"17ea796f99a85a49a832cb6975d1348bd5a4d39e","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-02-25 09:40:17.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"29512875455f12a0ad6428e4a7abf4b90d7d7757","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2021-02-25 10:34:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"566894698f6b0220e7d3820e98e2bb1a0e995883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 10:35:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fdff086b9efef19dfc3532276663fb72a0e1f2f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 10:54:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81542ee8c3d54783aab98339629450d3 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a5c22397a349a796371c87b1275c90 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/337564cc90df409a8032606a7b4cc4df : SUCCESS in 12m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6fee145c4d14dce96d96e4aa1c60a5e : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a42b7fae0554a2d91bb3a98a89062ff : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":4},{"id":"c5fec9325e8e8ec3d7c1fe6d1c3dbb52f3b2ad68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 10:54:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"475096c3e3d36309d19ee8a1b148c89f87a8f126","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 11:00:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3191a893f54e45c3922dbde1cb6262b6 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"811d47f0c06035f66fd7679dbe6ae19fc44c6d89","revisions":{"dfd24963e6c717bb929665365cf0b817e2f52e66":{"kind":"REWORK","_number":1,"created":"2021-02-24 02:36:34.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/75/777275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/75/777275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/75/777275/1"}}},"commit":{"parents":[{"commit":"22d69f971aec396f4b8f04f814cc86f2781d900b","subject":"Merge \"Implement SOL001 features to MgmtDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/22d69f971aec396f4b8f04f814cc86f2781d900b"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2021-02-24 02:34:16.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2021-02-24 02:34:16.000000000","tz":0},"subject":"DNM: Fix FT error for legacy anti-affinity test","message":"DNM: Fix FT error for legacy anti-affinity test\n\nIn the following patch, I solved the FT error by removing\n\"CELLSV2_SETUP: singleconductor\" and setup as a superconductor, but\nafter merging the patches, the error is become to occur in the\nanti-affinity test in the legacy FT.\n[1] https://review.opendev.org/c/openstack/tacker/+/775958\n\nBecause anti-affinity may not work well in a multi-cell environment\nwith nova installed [2].\n[2] https://opendev.org/openstack/tacker/commit/b20812678bdb5266f4483b2e55c582167a605e\n\nAnd db dependency error that occurred when setting single cell have\nbeen resolved by merging the following patches of devstack [3].\n[3] https://review.opendev.org/c/openstack/devstack/+/775940\n\nSo, this patch set config `CELLSV2_SETUP: singleconductor` in\n.zuul.yaml for resolve error of legacy anti-affinity test.\n\nChange-Id: Ia1bb10daf426820b44487c85fbffd3b7ff5c0f39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/dfd24963e6c717bb929665365cf0b817e2f52e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/dfd24963e6c717bb929665365cf0b817e2f52e66"}]},"branch":"refs/heads/master"},"f17bb00b25e4ba93d73fed22a996a83fc5527439":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-24 07:38:50.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/75/777275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/75/777275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/75/777275/2"}}},"commit":{"parents":[{"commit":"22d69f971aec396f4b8f04f814cc86f2781d900b","subject":"Merge \"Implement SOL001 features to MgmtDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/22d69f971aec396f4b8f04f814cc86f2781d900b"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2021-02-24 02:34:16.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2021-02-24 07:38:50.000000000","tz":0},"subject":"DNM: Fix FT error for legacy anti-affinity test","message":"DNM: Fix FT error for legacy anti-affinity test\n\nIn the following patch, FT error was solved by removing\n\"CELLSV2_SETUP: singleconductor\" and setup as a superconductor, but\nafter merging the patches, other error is become to occur in the\nanti-affinity test in the legacy FT.\n[1] https://review.opendev.org/c/openstack/tacker/+/775958\n\nBecause, anti-affinity may not work well in a environment with nova\ninstalled as multi-cell [2].\n[2] https://opendev.org/openstack/tacker/commit/b20812678bdb5266f4483b2e55c582167a605e\n\nAnd db dependency error that occurred when setting single cell before\nhave been resolved by merging the following patches of devstack [3].\n[3] https://review.opendev.org/c/openstack/devstack/+/775940\n\nTherefore, this patch revert `CELLSV2_SETUP: singleconductor` in\n.zuul.yaml to resolve error of legacy anti-affinity test.\n\nChange-Id: Ia1bb10daf426820b44487c85fbffd3b7ff5c0f39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/f17bb00b25e4ba93d73fed22a996a83fc5527439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/f17bb00b25e4ba93d73fed22a996a83fc5527439"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9b56c75cea5bc985b7692b4c5a103e1d3f726658":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-24 07:39:32.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/75/777275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/75/777275/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/75/777275/3"}}},"commit":{"parents":[{"commit":"22d69f971aec396f4b8f04f814cc86f2781d900b","subject":"Merge \"Implement SOL001 features to MgmtDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/22d69f971aec396f4b8f04f814cc86f2781d900b"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2021-02-24 02:34:16.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2021-02-24 07:39:32.000000000","tz":0},"subject":"DNM: Fix FT error for legacy anti-affinity test","message":"DNM: Fix FT error for legacy anti-affinity test\n\nIn the following patch [1], FT error was solved by removing\n\"CELLSV2_SETUP: singleconductor\" and setup as a superconductor, but\nafter merging the patch, other error is become to occur in the\nanti-affinity test in the legacy FT.\n[1] https://review.opendev.org/c/openstack/tacker/+/775958\n\nBecause, anti-affinity may not work well in a environment with nova\ninstalled as multi-cell [2].\n[2] https://opendev.org/openstack/tacker/commit/b20812678bdb5266f4483b2e55c582167a605e\n\nAnd db dependency error that occurred when setting single cell before\nhave been resolved by merging the following patches of devstack [3].\n[3] https://review.opendev.org/c/openstack/devstack/+/775940\n\nTherefore, this patch revert `CELLSV2_SETUP: singleconductor` in\n.zuul.yaml to resolve error of legacy anti-affinity test.\n\nChange-Id: Ia1bb10daf426820b44487c85fbffd3b7ff5c0f39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/9b56c75cea5bc985b7692b4c5a103e1d3f726658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/9b56c75cea5bc985b7692b4c5a103e1d3f726658"}]},"branch":"refs/heads/master","description":"Edit commit message"},"811d47f0c06035f66fd7679dbe6ae19fc44c6d89":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-02-24 10:03:18.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/75/777275/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/75/777275/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/75/777275/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/75/777275/4"}}},"commit":{"parents":[{"commit":"22d69f971aec396f4b8f04f814cc86f2781d900b","subject":"Merge \"Implement SOL001 features to MgmtDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/22d69f971aec396f4b8f04f814cc86f2781d900b"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2021-02-24 02:34:16.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2021-02-24 10:03:18.000000000","tz":0},"subject":"Fix FT error for legacy anti-affinity test","message":"Fix FT error for legacy anti-affinity test\n\nIn the following patch [1], FT error was solved by removing\n\"CELLSV2_SETUP: singleconductor\" and setup as a superconductor, but\nafter merging the patch, other error is become to occur in the\nanti-affinity test in the legacy FT.\n[1] https://review.opendev.org/c/openstack/tacker/+/775958\n\nBecause, anti-affinity may not work well in a environment with nova\ninstalled as multi-cell [2].\n[2] https://opendev.org/openstack/tacker/commit/b20812678bdb5266f4483b2e55c582167a605e\n\nAnd db dependency error that occurred when setting single cell before\nhave been resolved by merging the following patches of devstack [3].\n[3] https://review.opendev.org/c/openstack/devstack/+/775940\n\nTherefore, this patch revert `CELLSV2_SETUP: singleconductor` in\n\".zuul.yaml\" to resolve error of legacy anti-affinity test.\n\nChange-Id: Ia1bb10daf426820b44487c85fbffd3b7ff5c0f39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/811d47f0c06035f66fd7679dbe6ae19fc44c6d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/811d47f0c06035f66fd7679dbe6ae19fc44c6d89"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
