)]}'
{"id":"openstack%2Ftacker~837081","triplet_id":"openstack%2Ftacker~master~I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3","project":"openstack/tacker","branch":"master","topic":"bug/1968103","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-04-22 02:48:50.000000000","reason":"Ayumu Ueha replied on the change"},"31730":{"account":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"last_update":"2022-04-26 05:42:17.000000000","reason":"Change was submitted"},"33492":{"account":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"last_update":"2022-04-26 05:42:17.000000000","reason":"Change was submitted"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-04-26 04:59:39.000000000","reason":"Yasufumi Ogawa replied on the change"},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2022-04-26 05:19:11.000000000","reason":"Toshiaki Takahashi replied on the change"},"33292":{"account":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"last_update":"2022-04-26 05:42:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3","subject":"Fix validate error when k8s resource init","status":"MERGED","created":"2022-04-08 04:45:07.000000000","updated":"2022-04-26 05:44:51.000000000","submitted":"2022-04-26 05:42:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":41,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"837081-bug/1968103","meta_rev_id":"70a7f03ff0fdd4820f759877cb190208360e1ec7","_number":837081,"virtual_id_number":837081,"owner":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-26 05:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2022-04-26 04:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2022-04-26 05:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2022-04-26 05:19:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 11:45:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-12 07:54:22.000000000","updated_by":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"reviewer":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"state":"REVIEWER"},{"updated":"2022-04-12 07:54:22.000000000","updated_by":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2022-04-12 07:54:22.000000000","updated_by":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"reviewer":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"state":"REVIEWER"},{"updated":"2022-04-12 07:54:22.000000000","updated_by":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2022-04-12 07:54:22.000000000","updated_by":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"42a554444c1ff906c4c5818dfa0e59ca6242b841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-08 04:45:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"244ad9853ddeb2834848046c6b68ff9362a209af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-08 05:10:52.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6ceed5614a72805f127be96e5bbcf8fbcc2828ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-08 07:48:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"812424474c16ee948210818f59a45ba21ea361a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-08 08:37:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1338c22f02618582f4994bfe6ed627f780184a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 11:45:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d89ce5941d7c497b8f9353c5b18c6582 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb4da524c84f400bbe77eca9c5ef7559 : SUCCESS in 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31f9968cb4f342ea8f7c165af81bfbc0 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cffa3233f5d048f9adab50c579e2b5bd : SUCCESS in 17m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9c0b601b4044925b822d6736592e825 : SUCCESS in 17m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e563a3a5d63f4f6d811458950d4d8429 : SUCCESS in 19m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65306bb30f974e3dbbd884a0f6a11651 : SUCCESS in 7m 16s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/42787ced6f3a442f9a0294731e86ce7d : SUCCESS in 1h 54m 41s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/d707be33fb224b6ea5d7f1b6d8e43082 : SUCCESS in 2h 38m 53s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/d202c7d0859545958e0ce1403ef7b732 : SUCCESS in 1h 23m 22s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/56ab9e1a24ec4b3d8baaa0f88f0d142e : FAILURE in 2h 14m 00s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/f97d83f2a5574faea5e257e9e28fb5cc : SUCCESS in 1h 52m 52s\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/acc9f16018a04f9eb475df75dfb97cbe : SUCCESS in 2h 04m 45s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/00a652209cac4b47bf5bfc1484e275b7 : FAILURE in 2h 15m 02s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/237db1e16b504f2b82f2ef73f44afa31 : SUCCESS in 1h 20m 54s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/77bb2349a9e642f18450c4bf40227384 : SUCCESS in 1h 18m 22s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/461c679c17844c78bdf93130b99def74 : FAILURE in 3h 04m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1666a418bb485929fafd479c22477455775a8505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-12 01:11:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6824c2dc85b4e2f04cc59a9a8b93822bcf555104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 04:17:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ce4b81acbb242d486654ce264a398a7 : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27d00ee8fb8948cca521452ee5d30e80 : SUCCESS in 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b83219ad6c6491381408c203a503760 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13a6fa20683c4348afa77d01c456b7fd : SUCCESS in 17m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5829bee32cdd4077bcc5c725f8de89a6 : SUCCESS in 17m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96c8650d36c64804ab6b43527ceb47e2 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a4f935eb18b42268d214c155216b9e3 : SUCCESS in 7m 46s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/77aac280b80e4bd68ea1e977575d2225 : SUCCESS in 1h 39m 17s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/280cdc353a224c13af708f314bf2fbf1 : SUCCESS in 2h 20m 32s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/9181d6aef76d440ea69a3e708c4863a9 : SUCCESS in 1h 03m 10s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/88178700c8ec40df82a6bcb6eca7467c : SUCCESS in 2h 01m 34s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/460bd9adee464a11afcb0eef0ea37e53 : SUCCESS in 1h 43m 23s\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/014f6ff6ae834963be90e32172b2b7a6 : SUCCESS in 2h 06m 26s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/98a62efa2b7644ff90c07b6dd911547b : SUCCESS in 2h 15m 14s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/ca4a663504384508b5b59fa2cf59f675 : SUCCESS in 1h 08m 56s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/7ae77c9ef2994bfab835c2a413ccca15 : SUCCESS in 1h 38m 26s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/7e7c70cdbc2f4d6f8fdc2bd199914ddb : FAILURE in 3h 02m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"910b7640a9d96c231ea5db8e1ebb8d53dcbf47d5","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-04-14 00:54:04.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"bf3883a5bee6831467080511b06abd8ee416d212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-18 06:03:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bd3392acb9f71a97361a4d94bbf1cd4587fd493d","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-18 06:18:06.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d0644389a1a9bdd8b768af65a7c272c71fcb3944","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-04-18 08:58:53.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"980eb4e14ae741a898ed0dd88aa1325215fa86ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 09:16:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd8d3df84f54084bb86c3db337e34df : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/626c054abee54ebd87f7192e130ee0fe : SUCCESS in 18m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64c2975a9b9441a1ac086e5fe91d25e3 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdfba475ecb04a8599166863cb4707ad : SUCCESS in 16m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffe22f4391364c50a0d3ab73271d0156 : SUCCESS in 16m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6913127de9bf4a3f873d80f10dab9256 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/035f57479b104ebebb7713eb011fc46c : SUCCESS in 6m 36s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/80d191ff44714015bbee0b422835b406 : SUCCESS in 1h 28m 56s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/f474948d0d114233b01a627384d41925 : SUCCESS in 2h 03m 50s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/2e96ff20f7a4493c90fe65e48fb099d0 : SUCCESS in 59m 15s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/04c02de914d9401f8f63128f8ea20dc8 : SUCCESS in 1h 33m 00s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/30cc894564064f81abcbe5cfa06cfa08 : SUCCESS in 1h 32m 21s\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/731f6e931b9540ddb1e1f3c8e333af4e : SUCCESS in 2h 13m 51s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/82515055ef514fde9d225f0d7be12874 : SUCCESS in 2h 27m 19s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/b2b08e28d55f480eb1101e52692f5ae7 : SUCCESS in 1h 14m 41s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/418cefdbeb4f4e86af0f489235e45780 : SUCCESS in 1h 10m 05s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/2f1d247f885b4d5eb0a64ff8b6ff2468 : TIMED_OUT in 3h 04m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2d4179763509257db16cd0844d54845c9cc86f7d","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-04-18 11:06:58.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b013b6c33dc0978143859292b818427b5a25a62d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-20 07:23:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a2898584ff0089cae3cf639c4d9a998c73a1869c","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-20 07:25:02.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a3be1f0164c64e7da9ed99d4a91366a7536941b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 10:27:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64037a4b56024dc6bd5cb3e96471ad04 : SUCCESS in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9adc628751584b7fb93afd7526d93f41 : SUCCESS in 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387dfc5c0eea4724bb45a1bab178dd9f : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64b273e922ae4823a9ea2277a14b0aa5 : SUCCESS in 17m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e153ddbf9d3d4837835e412a925572a5 : SUCCESS in 17m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c6f0422a93f48c7af07e091a1a95db6 : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9bb8351fbf4987b519b1edf9a5e4d0 : SUCCESS in 6m 59s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/93dd3dca1f8b4d45b040f9befeed532c : SUCCESS in 1h 48m 17s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/ef7e5c2575f4433dad2a6e33fe352c8c : SUCCESS in 2h 10m 03s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/22f8ad5ffb9a4579b5bdb90664300ef2 : SUCCESS in 1h 20m 12s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/479ea275874341618c783ea9a62b5aac : SUCCESS in 1h 51m 04s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/3e5602601cba497b9268e36f84423df0 : SUCCESS in 1h 37m 21s\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/9ddbde81c97c4ca09d7c101d1f1957f2 : SUCCESS in 2h 01m 05s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/cd1fc388711343f08b063ca4d766d7d2 : SUCCESS in 2h 03m 13s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/a5b64c3ead624c37b1982d7c9740cf2f : SUCCESS in 56m 11s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/7115231323a646008011684b87f6d458 : SUCCESS in 57m 29s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/9bc766ef43ac4c7db29de33c1e84a9a4 : FAILURE in 3h 01m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fb937b51f7d4378d64af1a8935c8d3e96699da84","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-04-21 09:06:23.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"723593a372554935eca3a9531339fc82a72517fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-22 01:29:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b9dea143055eb3464067c8e270d1bc4f25a93be1","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-22 01:31:52.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"013bde38208e3551db1e1fec02419db69f868f26","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-04-22 02:48:50.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8478bd9861805365ef0ba65f6701bf40219d2800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 04:39:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f9062ae66504900a46afd4a33eca80c : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6109e7fea7df4701a08ef8f71e025c81 : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f04a29d5f34b40a65f019c561447d9 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29be5af1f8a14a7599273df6e4d03e2b : SUCCESS in 17m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97296ba6865f414ba677680af89767ce : SUCCESS in 17m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bd3708af6244552a76202de4c9a65e2 : SUCCESS in 17m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f184d797eb74a87a6d8eb144f773207 : SUCCESS in 7m 45s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/51f221c9086c40ba99687a248867f599 : SUCCESS in 1h 31m 28s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/d58c7d692e59480f894918c844e5b5ca : SUCCESS in 2h 21m 59s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/7d40168fd2a240a78642796da3398090 : SUCCESS in 1h 14m 39s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/5b74d281d7684f49a343da8f6b11454f : SUCCESS in 2h 02m 43s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/fb7335d85c39434c8d886abf706be55a : SUCCESS in 1h 46m 58s\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/52904e22d9554ae29076456117ca2ef3 : SUCCESS in 1h 59m 46s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/f173d845108a433788ef1f7c10ef4896 : SUCCESS in 2h 21m 26s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/12c215e766e94522bd8e8cd4a5a270e6 : SUCCESS in 1h 12m 42s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/7ff803af70884439ad32d06ebcf0f74b : SUCCESS in 1h 33m 07s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/c5bab8ff0f244955b7ccb1139a2ed3ed : TIMED_OUT in 3h 04m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"41e6638e5c8dbbeff1fc76aeddae67cc9f36048c","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-04-22 06:23:42.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4b5f545d972a8e51d793bd5826ff21cb613e8496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-25 07:34:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"044b0485c0c978c9ffa54249e1bace368fb775b5","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-25 07:36:34.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c5bdee6685e5dfa442cd909500f49455d32e695e","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-04-25 09:13:59.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"91997e52635e876722696fb4d9468e7d125d2605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 10:42:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/602661fd8ab14b45a95d20d7217d5e39 : SUCCESS in 7m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cf23bb2930b4e6e98eb3bb64a763856 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934b517c82e44fdf9fa1963bf5a44c46 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20761ffbf17941f7bfe9848215f9ca61 : SUCCESS in 17m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bbf204f7a2145ac858611053c4586be : SUCCESS in 17m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb1cd8b231d54205b2746cae10e9163d : SUCCESS in 18m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898a68301af54e2cbb0d5d9419e854a2 : SUCCESS in 7m 16s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/68ed0128d94842a2aec4a23892be5b65 : SUCCESS in 1h 52m 37s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/3164fdfd854c4dcf8e23d8a3000a86e3 : SUCCESS in 1h 52m 38s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/797d140cea664f27a90900c86b60ee37 : SUCCESS in 1h 17m 03s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/8f06818018fe4daea7b97aab6475d9f9 : SUCCESS in 1h 49m 26s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/bc47085ef1734697b15701c98993e6b7 : SUCCESS in 1h 35m 29s\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/31ca6745b6c44f5d9ed9300c446f935d : SUCCESS in 1h 53m 36s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/c89462955dab4119be13d60a3c5c9582 : SUCCESS in 2h 29m 32s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/6e0a873ce4a14f48ba5aa199225de4a1 : SUCCESS in 1h 03m 44s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/1f917192b89848f683533b27b2713c30 : SUCCESS in 1h 18m 38s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/54bcd4dc839143f5aec057da5ee3b391 : TIMED_OUT in 3h 05m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"490dd1b30ae949dd862c948baf5608d19ce23546","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-26 01:36:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9ce027b5ef419af9bdb12391c47daf4ac3fd784c","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2022-04-26 01:39:40.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f3f12aff45a0e3855f244daf30d1d79797b2f6ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 04:47:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67bb82662b994e0aaa75134a8e8112a3 : SUCCESS in 10m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b012f2ae2cb94589b7c3e2c06b226627 : SUCCESS in 17m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68be3abd0d9f4ee48519a380f59049e6 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ba07d74590b48d693e3fc2c1fbceb69 : SUCCESS in 17m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4101a89f4efa4209ad03bc978ea0e1da : SUCCESS in 17m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5afc7a3b3f44cb6968e3f7ebe5c9cc6 : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff323c30a5548faa3563da307b99a45 : SUCCESS in 7m 22s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/3e83fc5bb3f847138b16751b7e35fb56 : SUCCESS in 2h 00m 49s\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/1567f2a56126409aa262d30f97a35bf2 : SUCCESS in 2h 30m 19s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/f1fd535a05964330925c64a826159dff : SUCCESS in 1h 16m 48s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/fc086f80a7184419863a9436e7b8efe1 : SUCCESS in 2h 02m 41s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/7a3559d6f730484892a3feb63e690df5 : SUCCESS in 1h 49m 26s\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/487273a781804d6ca45985c635ad3e68 : SUCCESS in 2h 02m 17s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/1932260119a347f280814864098b3564 : SUCCESS in 2h 21m 35s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/bca1fba3f08f404b835650f1efff078b : SUCCESS in 1h 12m 26s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/f1afce0872c04a009a32389bd9be33c8 : SUCCESS in 1h 17m 11s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/4d8463a966854d83bae62793adaea5cf : TIMED_OUT in 3h 05m 02s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8bf6cc3ba2356fe3af872803730bec3095b43f25","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-04-26 04:59:39.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e39fe3cfea120f03de6b35845913d83565a29f9c","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2022-04-26 05:19:11.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"75a1adcd8f36c5226922c4f90729bcfc6fad5c7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 05:19:36.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"817fcbd5d8d29441286b5ffc810cb12fc7da42b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 05:42:16.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f48112be458447a9f772de57fa49aab : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15322ad203a344f19f38763dc5ea4b52 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3953bfdf29a84726a546124d9b590707 : SUCCESS in 17m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d4f154a66b64577a29b2efad2020fb0 : SUCCESS in 17m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f9cf7d3405d43a99cd78c61c346b65a : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":10},{"id":"e0cca329cfeb4c517793effe3778dfa5fa79b72b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 05:42:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"70a7f03ff0fdd4820f759877cb190208360e1ec7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 05:44:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17bbc3a4105146d08acb9e931aa3af70 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e2db1c7a4032d316e5b655b858b023ace8435cd8","revisions":{"f2d4057bd377ee10943654bbb1ce1ce6b13a3558":{"kind":"REWORK","_number":1,"created":"2022-04-08 04:45:07.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/1"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 04:45:44.000000000","tz":540},"subject":"Fix validate error when k8s resource init Since the k8s version has been upgraded from v22.6.0 to v23.3.0, the initial value of some fields must be the specified string, which causes the initialization of the current k8s resource to fail.","message":"Fix validate error when k8s resource init\nSince the k8s version has been upgraded from v22.6.0 to v23.3.0,\nthe initial value of some fields must be the specified string,\nwhich causes the initialization of the current k8s resource to fail.\n\nThis patch fixes the issue and skips the validation of k8s object\ninitialization by setting the `client_side_validation` parameter.\nIn order to pass the validation, the must_param variable was added\nto set the initialized value. This modification skips the\nvalidation and no longer uses the variable, so the related processing\nof the variable is deleted.\n\nAt the same time, due to the rapid iteration of the k8s version,\nthe must_param variable needs to be repaired in each iteration to\nadapt to the new validation, which requires repeated maintenance\nand high maintenance costs, so it is no longer used.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/f2d4057bd377ee10943654bbb1ce1ce6b13a3558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/f2d4057bd377ee10943654bbb1ce1ce6b13a3558"}]},"branch":"refs/heads/master"},"8f46814b053de176490ad986f74744434dd01972":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-08 05:10:52.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/2"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 05:10:52.000000000","tz":0},"subject":"Fix validate error when k8s resource init","message":"Fix validate error when k8s resource init\n\nSince the k8s version has been upgraded from v22.6.0 to v23.3.0,\nthe initial value of some fields must be the specified string,\nwhich causes the initialization of the current k8s resource to fail.\n\nThis patch fixes the issue and skips the validation of k8s object\ninitialization by setting the `client_side_validation` parameter.\nIn order to pass the validation, the must_param variable was added\nto set the initialized value. This modification skips the\nvalidation and no longer uses the variable, so the related processing\nof the variable is deleted.\n\nAt the same time, due to the rapid iteration of the k8s version,\nthe must_param variable needs to be repaired in each iteration to\nadapt to the new validation, which requires repeated maintenance\nand high maintenance costs, so it is no longer used.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8f46814b053de176490ad986f74744434dd01972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8f46814b053de176490ad986f74744434dd01972"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ae1e7f1abc6f235d147ab2e88dd8aba763e9f112":{"kind":"REWORK","_number":3,"created":"2022-04-08 07:48:19.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/3"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 07:42:42.000000000","tz":540},"subject":"Fix validate error when k8s resource init","message":"Fix validate error when k8s resource init\n\nSince the k8s version has been upgraded from v22.6.0 to v23.3.0,\nthe initial value of some fields must be the specified string,\nwhich causes the initialization of the current k8s resource to fail.\n\nThis patch fixes the issue and skips the validation of k8s object\ninitialization by setting the `client_side_validation` parameter.\nIn order to pass the validation, the must_param variable was added\nto set the initialized value. This modification skips the\nvalidation and no longer uses the variable, so the related processing\nof the variable is deleted. At the same time, even if the\ninitialization validation is skipped, the fields in the yaml file\nwill still be verified when the create method is called, and an\nexception will be thrown and output to the log if the fields are\nmissing.\n\nAt the same time, due to the rapid iteration of the k8s version,\nthe must_param variable needs to be repaired in each iteration to\nadapt to the new validation, which requires repeated maintenance\nand high maintenance costs, so it is no longer used.\n\nIn kubernetes v23.3.0, \u0027available_replicas\u0027 must be set into status\nof `StatefulSet`, so this patch added this attribute into statefulset\nyaml used in FT.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ae1e7f1abc6f235d147ab2e88dd8aba763e9f112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ae1e7f1abc6f235d147ab2e88dd8aba763e9f112"}]},"branch":"refs/heads/master"},"867e63713cf04c0f68a9a96094bd7877a7a06883":{"kind":"REWORK","_number":4,"created":"2022-04-08 08:37:44.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/4"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 08:37:56.000000000","tz":540},"subject":"Fix validate error when k8s resource init","message":"Fix validate error when k8s resource init\n\nSince the k8s version has been upgraded from v22.6.0 to v23.3.0,\nthe initial value of some fields must be the specified string,\nwhich causes the initialization of the current k8s resource to fail.\n\nThis patch fixes the issue and skips the validation of k8s object\ninitialization by setting the `client_side_validation` parameter.\nIn order to pass the validation, the must_param variable was added\nto set the initialized value. This modification skips the\nvalidation and no longer uses the variable, so the related processing\nof the variable is deleted. At the same time, even if the\ninitialization validation is skipped, the fields in the yaml file\nwill still be verified when the create method is called, and an\nexception will be thrown and output to the log if the fields are\nmissing.\n\nAt the same time, due to the rapid iteration of the k8s version,\nthe must_param variable needs to be repaired in each iteration to\nadapt to the new validation, which requires repeated maintenance\nand high maintenance costs, so it is no longer used.\n\nIn kubernetes v23.3.0, \u0027available_replicas\u0027 must be set into status\nof `StatefulSet`, so this patch added this attribute into statefulset\nyaml used in FT.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/867e63713cf04c0f68a9a96094bd7877a7a06883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/867e63713cf04c0f68a9a96094bd7877a7a06883"}]},"branch":"refs/heads/master"},"0d3a9ba33f229f9ddf9557c24f79462edeb013b9":{"kind":"REWORK","_number":5,"created":"2022-04-12 01:11:23.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/5"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-12 01:11:18.000000000","tz":540},"subject":"Fix validate error when k8s resource init","message":"Fix validate error when k8s resource init\n\nSince the k8s version has been upgraded from v22.6.0 to v23.3.0,\nthe initial value of some fields must be the specified string,\nwhich causes the initialization of the current k8s resource to fail.\n\nThis patch fixes the issue and skips the validation of k8s object\ninitialization by setting the `client_side_validation` parameter.\nIn order to pass the validation, the must_param variable was added\nto set the initialized value. This modification skips the\nvalidation and no longer uses the variable, so the related processing\nof the variable is deleted. At the same time, even if the\ninitialization validation is skipped, the fields in the yaml file\nwill still be verified when the create method is called, and an\nexception will be thrown and output to the log if the fields are\nmissing.\n\nAt the same time, due to the rapid iteration of the k8s version,\nthe must_param variable needs to be repaired in each iteration to\nadapt to the new validation, which requires repeated maintenance\nand high maintenance costs, so it is no longer used.\n\nIn kubernetes v23.3.0, \u0027available_replicas\u0027 must be set into status\nof `StatefulSet`, so this patch added this attribute into statefulset\nyaml used in FT.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/0d3a9ba33f229f9ddf9557c24f79462edeb013b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/0d3a9ba33f229f9ddf9557c24f79462edeb013b9"}]},"branch":"refs/heads/master"},"c52343a974b05fe880bd01977bd82641cc0f2a8c":{"kind":"REWORK","_number":6,"created":"2022-04-18 06:03:49.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/6"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-18 06:01:25.000000000","tz":540},"subject":"Fix validate error when k8s resource init","message":"Fix validate error when k8s resource init\n\nSince the k8s version has been upgraded from v22.6.0 to v23.3.0,\nthe initial value of some fields must be the specified string,\nwhich causes the initialization of the current k8s resource to fail.\n\nThis patch fixes the issue and skips the validation of k8s object\ninitialization by setting the `client_side_validation` parameter.\nIn order to pass the validation, the must_param variable was added\nto set the initialized value. This modification skips the\nvalidation and no longer uses the variable, so the related processing\nof the variable is deleted. At the same time, even if the\ninitialization validation is skipped, the fields in the yaml file\nwill still be verified when the create method is called, and an\nexception will be thrown and output to the log if the fields are\nmissing.\n\nAt the same time, due to the rapid iteration of the k8s version,\nthe must_param variable needs to be repaired in each iteration to\nadapt to the new validation, which requires repeated maintenance\nand high maintenance costs, so it is no longer used.\n\nIn kubernetes v23.3.0, \u0027available_replicas\u0027 must be set into status\nof `StatefulSet`, so this patch add a kuryr-kubernetes versioned\nparameter temporarily to .zuul.yaml to make the response returned\nby kubernetes correct.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/c52343a974b05fe880bd01977bd82641cc0f2a8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/c52343a974b05fe880bd01977bd82641cc0f2a8c"}]},"branch":"refs/heads/master"},"6a0ba25f122ad3944c4a07f28f7893b85070c6be":{"kind":"REWORK","_number":7,"created":"2022-04-20 07:23:02.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/7"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-20 07:22:20.000000000","tz":540},"subject":"Fix validate error when k8s resource init","message":"Fix validate error when k8s resource init\n\nThis patch fixes the issue that k8s failed to initialize object due\nto version upgrade. Skip initial validation by setting the\n`client_side_validation` parameter and remove the must_param\nvariable that was previously set in order to pass the validation.\n\nIn kubernetes v23.3.0, \u0027available_replicas\u0027 must be set into status\nof `StatefulSet`, so this patch also add a kuryr-kubernetes versioned\nparameter temporarily to .zuul.yaml to make the response returned\nby kubernetes correct.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/6a0ba25f122ad3944c4a07f28f7893b85070c6be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/6a0ba25f122ad3944c4a07f28f7893b85070c6be"}]},"branch":"refs/heads/master"},"ee16eca0c3c113ae49fb28e200371df20d167480":{"kind":"REWORK","_number":8,"created":"2022-04-22 01:29:08.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/8"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-22 01:29:46.000000000","tz":540},"subject":"Fix validate error when k8s resource init","message":"Fix validate error when k8s resource init\n\nThis patch fixes the issue that k8s failed to initialize object due\nto version upgrade. Skip initial validation by setting the\n`client_side_validation` parameter and remove the must_param\nvariable that was previously set in order to pass the validation.\n\nIn kubernetes v23.3.0, \u0027available_replicas\u0027 must be set into status\nof `StatefulSet`, so this patch also add a kuryr-kubernetes versioned\nparameter temporarily to .zuul.yaml to make the response returned\nby kubernetes correct.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ee16eca0c3c113ae49fb28e200371df20d167480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ee16eca0c3c113ae49fb28e200371df20d167480"}]},"branch":"refs/heads/master"},"d3f9e025980fbccca8f9ff26e91aacf58b7a1c45":{"kind":"REWORK","_number":9,"created":"2022-04-25 07:34:31.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/9"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-25 07:35:33.000000000","tz":540},"subject":"Fix validate error when k8s resource init","message":"Fix validate error when k8s resource init\n\nThis patch fixes the issue that k8s failed to initialize object due\nto version upgrade. Skip initial validation by setting the\n`client_side_validation` parameter and remove the must_param\nvariable that was previously set in order to pass the validation.\n\nIn kubernetes v23.3.0, \u0027available_replicas\u0027 must be set into status\nof `StatefulSet`, so this patch also add a kuryr-kubernetes versioned\nparameter temporarily to .zuul.yaml to make the response returned\nby kubernetes correct.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/d3f9e025980fbccca8f9ff26e91aacf58b7a1c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/d3f9e025980fbccca8f9ff26e91aacf58b7a1c45"}]},"branch":"refs/heads/master"},"e2db1c7a4032d316e5b655b858b023ace8435cd8":{"kind":"REWORK","_number":10,"created":"2022-04-26 01:36:24.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/81/837081/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/81/837081/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/81/837081/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/81/837081/10"}}},"commit":{"parents":[{"commit":"acf84a168acd4dc239c9a55f21a1b19d96377b81","subject":"Merge \"Fix failure of testing privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/acf84a168acd4dc239c9a55f21a1b19d96377b81"}]}],"author":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-08 02:43:34.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2022-04-26 01:37:28.000000000","tz":540},"subject":"Fix validate error when k8s resource init","message":"Fix validate error when k8s resource init\n\nThis patch fixes the issue that k8s failed to initialize object due\nto version upgrade. Skip initial validation by setting the\n`client_side_validation` parameter and remove the must_param\nvariable that was previously set in order to pass the validation.\n\nIn kubernetes v23.3.0, \u0027available_replicas\u0027 must be set into status\nof `StatefulSet`, so this patch also add a kuryr-kubernetes versioned\nparameter temporarily to .zuul.yaml to make the response returned\nby kubernetes correct.\n\nCloses-Bug: #1968103\nChange-Id: I9495ce0f0893e5f9a1d6c52b98c3db3928bd95a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e2db1c7a4032d316e5b655b858b023ace8435cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e2db1c7a4032d316e5b655b858b023ace8435cd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
