)]}'
{"id":"openstack%2Ftacker~872132","triplet_id":"openstack%2Ftacker~master~I8edfa6e543103eb5c20993398f53443d3b13f407","project":"openstack/tacker","branch":"master","topic":"dnm-fix-free5gc-vnf","attention_set":{"31730":{"account":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"last_update":"2023-01-31 08:47:34.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8edfa6e543103eb5c20993398f53443d3b13f407","subject":"[DNM] Fix free5GC provided as a sample for VNF does not work","status":"NEW","created":"2023-01-31 05:57:52.000000000","updated":"2023-09-28 03:57:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"320a510ae4c533284796942bb701f87f59438920","_number":872132,"virtual_id_number":872132,"owner":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-24 06:56:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 08:47:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 05:26:17.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"state":"REVIEWER"},{"updated":"2023-09-28 03:57:17.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"state":"REMOVED"}],"messages":[{"id":"903f31122d43f52f55d7a8dd9858a02024c15af0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"date":"2023-01-31 05:57:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f0ea4a008f6d9956adc01ba4b4ec0946d2418ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 08:47:34.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9301e87064934aa88661e9124505a3fe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/563e9f3fe75c4cbeb250d669ba1276fc : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aae200a8cd6480495b0277fd8562739 : FAILURE in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80972816bcfc42d7a0a394f427a28725 : SUCCESS in 9m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a944fe0ceaa5410f84d4b4a633e76b8a : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09fb0072ca4475c990d1f26c7a1f19d : SUCCESS in 7m 19s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/b1cf3c25bd454052ad13419e041973ff : SUCCESS in 1h 33m 03s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/19ba7f9fd5a54edc84e9d38ee0679a1a : FAILURE in 2h 45m 02s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/3fe444df28c54614b27cfa1fe7e53ed0 : SUCCESS in 47m 49s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/e8843119f43b4bb682af901f9906efd5 : SUCCESS in 1h 12m 37s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/9cef9b17b7a943999b450ded1921eca4 : FAILURE in 1h 48m 49s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/a3fb5e11cb624b19afb0116ce8832dc3 : SUCCESS in 1h 53m 31s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/3824917c1eb44474bce2b636d844b40f : SUCCESS in 1h 14m 25s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/8a3c278638ba43cfa92c9d3bb2515af0 : SUCCESS in 1h 21m 21s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/cb251bcfe3f641dca0d3010e4ff99c42 : SUCCESS in 53m 08s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/d8cbdd5f3d47448e84f0e258f0c3eb38 : SUCCESS in 57m 59s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/11f1703754a246438d81a7d26c97c961 : SUCCESS in 1h 20m 55s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/b4d5e6f5f09c42c2898268ed7ea2edde : FAILURE in 46m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5da35bcbdabdf2603434d1bf041907916f9077dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"date":"2023-03-24 04:21:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"741bbcbeb44162328ba41e4aff533ddcdf0d2ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 06:56:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68b12fa14dce4cc794639488dfb72762\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1baf1a62621406f8da6c34051aab2bc : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174cc3f0364441ea906458b7f1035c9f : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d8a270506ac4150b2dc46378dd522d0 : SUCCESS in 9m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6db89dff62e45afb456251b74e65bb0 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90563d2ad3894ebc92e794d104b6b726 : SUCCESS in 6m 23s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/b25105a432c34a0589fab29bb14b4532 : SUCCESS in 1h 39m 00s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/bf23b05b2bf743adaa6a946504e1c767 : SUCCESS in 1h 47m 49s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/fda3a4b809d44b1abd508a8029b6e52f : SUCCESS in 43m 49s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/87e9d376d98043b1b861133e35b2c8c9 : SUCCESS in 1h 07m 55s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/420e816be00f450d9467384f5826e082 : SUCCESS in 1h 43m 15s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/7221e3f3a23645d0b6c56b514c1f7c5a : SUCCESS in 2h 22m 38s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/bffe06bd8e764817a6a9c629243c4fc7 : SUCCESS in 1h 21m 10s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b1ee485cc9647daadeaf8041f4ce8f3 : SUCCESS in 2h 26m 47s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/084c35046c7148eea200a62937e1130d : SUCCESS in 1h 34m 43s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/6ae4f2caaf334cd9b2aa4a95dac39dda : SUCCESS in 1h 09m 48s\n- tacker-functional-devstack-multinode-sol-https-v2 https://zuul.opendev.org/t/openstack/build/792c06d603d64cd2a89d1431f4600f89 : SUCCESS in 1h 07m 25s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/8fc0ac06fbd34f97a6b36fee835bb451 : SUCCESS in 1h 24m 23s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/3f9a4f5444cc4c5d88cb3f3d62f63e3d : SUCCESS in 1h 03m 53s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/42385fead11a4faeb16b7deed3b6b296 : SUCCESS in 56m 25s\n- tacker-functional-devstack-enhanced-policy-sol https://zuul.opendev.org/t/openstack/build/d2156c7510fe479c863bcd97469be102 : SUCCESS in 1h 16m 34s\n- tacker-functional-devstack-enhanced-policy-sol-kubernetes https://zuul.opendev.org/t/openstack/build/93b5053d63c349619de7cd33ddc841bb : SUCCESS in 1h 31m 07s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/ac0d36ecde25427083c13789b0c9d3e3 : FAILURE in 56m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ce98720739993656280f44948b7a8420a61d71c9","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-26 05:26:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"320a510ae4c533284796942bb701f87f59438920","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-28 03:57:17.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"931b158036208c0c6814edd27b22387bb99d7dce","revisions":{"361fd9d561b257e0d20582c467d66a6d2de74fa4":{"kind":"REWORK","_number":1,"created":"2023-01-31 05:57:52.000000000","uploader":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"ref":"refs/changes/32/872132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/32/872132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/32/872132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/32/872132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/32/872132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/32/872132/1"}}},"commit":{"parents":[{"commit":"01af921686f8e0af23c892b4536dcbdc2a9f2cf4","subject":"Switch the Legacy FT non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/01af921686f8e0af23c892b4536dcbdc2a9f2cf4"}]}],"author":{"name":"Naoaki Horie","email":"horie.naoaki@ntt-tx.co.jp","date":"2023-01-31 05:55:40.000000000","tz":540},"committer":{"name":"Naoaki Horie","email":"horie.naoaki@ntt-tx.co.jp","date":"2023-01-31 05:55:40.000000000","tz":540},"subject":"[DNM] Fix free5GC provided as a sample for VNF does not work","message":"[DNM] Fix free5GC provided as a sample for VNF does not work\n\nThis update is for the fix as follows.\n  - Change IP address in the MgmtDriver and the install script\n    from the static values to the default values of the free5GC.\n  - Fixed issue that the webconsole doesn\u0027t work.\n  - Fixed issue that the UPF process doesn\u0027t work in scaled VM.\n  - Delete the sudoers setup since it has already been set up in VM image.\n\nChange-Id: I8edfa6e543103eb5c20993398f53443d3b13f407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/361fd9d561b257e0d20582c467d66a6d2de74fa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/361fd9d561b257e0d20582c467d66a6d2de74fa4"}]},"branch":"refs/heads/master"},"931b158036208c0c6814edd27b22387bb99d7dce":{"kind":"REWORK","_number":2,"created":"2023-03-24 04:21:09.000000000","uploader":{"_account_id":31730,"name":"Naoaki Horie","email":"naoaki.h.opst@gmail.com","username":"naoaki-horie"},"ref":"refs/changes/32/872132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/32/872132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/32/872132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/32/872132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/32/872132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/32/872132/2"}}},"commit":{"parents":[{"commit":"01af921686f8e0af23c892b4536dcbdc2a9f2cf4","subject":"Switch the Legacy FT non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/01af921686f8e0af23c892b4536dcbdc2a9f2cf4"}]}],"author":{"name":"Naoaki Horie","email":"horie.naoaki@ntt-tx.co.jp","date":"2023-01-31 05:55:40.000000000","tz":540},"committer":{"name":"Naoaki Horie","email":"horie.naoaki@ntt-tx.co.jp","date":"2023-03-24 04:15:49.000000000","tz":540},"subject":"[DNM] Fix free5GC provided as a sample for VNF does not work","message":"[DNM] Fix free5GC provided as a sample for VNF does not work\n\nThis patch fixes the following:\n  - Change IP address in the MgmtDriver and the install script\n    from the static values to the default values of the free5GC.\n  - Fixed an issue that the webconsole doesn\u0027t work.\n  - Fixed an issue that the UPF process doesn\u0027t work in scaled VM.\n  - Delete the sudoers setup since it has already been set up\n    in VM image.\n  - Fixed an issue that the SMF configuration file was not updated\n    when executing scale-out.\n\nChange-Id: I8edfa6e543103eb5c20993398f53443d3b13f407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/931b158036208c0c6814edd27b22387bb99d7dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/931b158036208c0c6814edd27b22387bb99d7dce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
