)]}'
{"id":"openstack%2Ftacker~872252","triplet_id":"openstack%2Ftacker~master~Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022","project":"openstack/tacker","branch":"master","topic":"bp/add-sample-coordinate-script","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2023-03-02 05:29:35.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"34226":{"account":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"last_update":"2023-03-03 00:20:04.000000000","reason":"Change was submitted"},"31668":{"account":{"_account_id":31668,"deleted":true},"last_update":"2023-03-02 07:29:42.000000000","reason":"\u003cGERRIT_ACCOUNT_31668\u003e replied on the change","reason_account":{"_account_id":31668,"deleted":true}},"35476":{"account":{"_account_id":35476,"name":"taiki kimura","email":"taiki.kimura@ntt-at.co.jp","username":"tkimura"},"last_update":"2023-02-22 01:22:39.000000000","reason":"\u003cGERRIT_ACCOUNT_35476\u003e replied on the change","reason_account":{"_account_id":35476,"name":"taiki kimura","email":"taiki.kimura@ntt-at.co.jp","username":"tkimura"}},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2023-03-03 00:02:15.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"4149":{"account":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"last_update":"2023-02-28 02:43:02.000000000","reason":"\u003cGERRIT_ACCOUNT_4149\u003e replied on the change","reason_account":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2023-03-03 00:20:04.000000000","reason":"Change was submitted"},"34712":{"account":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"last_update":"2023-02-28 00:14:07.000000000","reason":"\u003cGERRIT_ACCOUNT_34712\u003e replied on the change","reason_account":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"}}},"hashtags":[],"change_id":"Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022","subject":"Support VNF LCM Coordination IF in change_vnfpkg","status":"MERGED","created":"2023-01-31 03:41:21.000000000","updated":"2023-03-03 00:22:52.000000000","submitted":"2023-03-03 00:20:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":123,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"872252-bp/add-sample-coordinate-script","meta_rev_id":"06ca68da4882ce226e5c43ec6634f342610a602f","_number":872252,"virtual_id_number":872252,"owner":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},{"value":0,"_account_id":31668},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-03 00:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2023-03-02 14:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},{"value":1,"date":"2023-03-02 09:48:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31668},{"value":2,"date":"2023-03-03 00:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2023-03-02 23:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},{"value":0,"_account_id":31668},{"value":1,"date":"2023-03-03 00:02:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"CC":[{"_account_id":35476,"name":"taiki kimura","email":"taiki.kimura@ntt-at.co.jp","username":"tkimura"}],"REVIEWER":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31668,"deleted":true},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 03:58:50.000000000","updated_by":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"reviewer":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"state":"REVIEWER"},{"updated":"2023-01-31 05:57:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 23:51:38.000000000","updated_by":{"_account_id":35476,"name":"taiki kimura","email":"taiki.kimura@ntt-at.co.jp","username":"tkimura"},"reviewer":{"_account_id":35476,"name":"taiki kimura","email":"taiki.kimura@ntt-at.co.jp","username":"tkimura"},"state":"CC"},{"updated":"2023-02-27 00:27:25.000000000","updated_by":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"reviewer":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"state":"REVIEWER"},{"updated":"2023-02-28 00:15:33.000000000","updated_by":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2023-02-28 00:15:33.000000000","updated_by":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2023-02-28 00:15:33.000000000","updated_by":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2023-03-01 09:58:52.000000000","updated_by":{"_account_id":31668,"deleted":true},"reviewer":{"_account_id":31668,"deleted":true},"state":"CC"},{"updated":"2023-03-02 05:24:25.000000000","updated_by":{"_account_id":31668,"deleted":true},"reviewer":{"_account_id":31668,"deleted":true},"state":"REVIEWER"}],"messages":[{"id":"7b1bdde2941a02dca9a39000b5cdb2fe99db6e68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-01-31 03:41:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a00853d6951be1f025503d79e7810894180589f5","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-01-31 03:42:53.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b5e0841dd6ed502847092df7f131ae34512a7586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 05:57:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2eb90600354456987a1ca0636b44b13\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99d40d7194ff4487a59d2d1f2610cf5f : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb352f98b9fc40b99ae0f454b6c6be32 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87d6d2abb0bc43b7a85a468f11c8f997 : SUCCESS in 9m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84143e7746a442a38195a558d14751fa : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499b951e57dc44b8818803fe6c2d2097 : SUCCESS in 6m 48s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/a974b7a45a544f588874ed33e2b81b10 : SUCCESS in 1h 50m 28s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/57249845b1814b8da494a472766ecce3 : SUCCESS in 2h 01m 52s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/ae0a935e160e49719079924f9d7fa965 : SUCCESS in 54m 45s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/0f7b17a7ca214ed584766f89b448fbca : SUCCESS in 1h 16m 21s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/a79e32cc45124853afa7351ebe4c0f8d : SUCCESS in 1h 45m 22s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/54d12493bd0246698ad9acefb52666e2 : FAILURE in 2h 02m 36s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/368e3592419342b29726781349a8a3d0 : SUCCESS in 1h 13m 20s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/7db6c39e954241d38d33bf462e1417ff : SUCCESS in 1h 26m 03s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/4b5242f4872141be9d3971f551703843 : SUCCESS in 1h 07m 59s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/8d0cf7a47fde44519ee8596cd3252767 : SUCCESS in 53m 19s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/e5b48c4984e04346932c0553f94b1a6c : SUCCESS in 1h 07m 56s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/1601e848aebc4cc88e3946dd94fc6ba3 : SUCCESS in 43m 20s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/e70a5c809cb142d1b4297a652e262a75 : FAILURE in 59m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8e56ee3e1136d0925e9459d2b9866f10c260ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-01-31 06:23:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0ecc2b6d562eaf20b977298e465b76afaea59195","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-01-31 06:24:05.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"8e61b44ff8d87da7d5d7d51f89ceea616421ccc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 08:59:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0017424649d0437f9f7c765bdbaa4d3d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e875e29868ea4957adf790004d3c74de : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf5803686de49cdb3dd55f5ddfa34cb : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0ff53d5abf8405c94ac84aface4cdf9 : SUCCESS in 8m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/555bbf49d53748fc95a925de6862f4bb : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a522dee8eeb44da6af861b39e87fa707 : SUCCESS in 6m 41s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/4984e32393544122b6e3ae6b6899ee38 : SUCCESS in 1h 37m 10s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/8a8416378b894bc4adec8fc6ff52bc74 : FAILURE in 2h 20m 18s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/afff32f646be482498169233b132b071 : SUCCESS in 56m 12s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/5acaf81cff1b4241b53ec596fd626330 : SUCCESS in 1h 39m 02s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/e89fb91a639e40b99d3fedbfe830ac2f : SUCCESS in 1h 27m 57s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/618974309fc44445b0b9cda600957ca4 : SUCCESS in 2h 02m 47s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/f6973e03644045d0bca9813ba9082eb2 : SUCCESS in 1h 21m 58s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/0addc8002e9b40a39fbc1bba8b994571 : SUCCESS in 1h 29m 49s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/65cada45d7e245b8a869efac1990fbce : SUCCESS in 57m 58s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/bacd973e848e4fecb7dc1184bf476ed0 : SUCCESS in 46m 45s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/acefe00f48064d4aa8312fa21babfaa5 : SUCCESS in 1h 08m 30s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/e1cb4e1a3ed74005aae82cc40f65a671 : SUCCESS in 1h 09m 11s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/d2d371d50a5641c2bab6b3cfc0f7da9e : FAILURE in 55m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ccc36bd2980b86e11a9783a1f4f0bad917b0381","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-02-08 01:27:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8af757a9ebf6aaa941227e45693e8fb0e862f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 03:54:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/207268fa9cab48dfb8f57a3da34492a7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cab657ad0c5b42a5a10dd37bf24b9462 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba0ce5507f34e4db9c340cba057a03e : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/211e2b8d79bb4be08f7b0d8389e0e9d1 : SUCCESS in 9m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6e97bb035144453b3345500d7f411f4 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/000e7a1007134867b21c6fa8e9432e15 : SUCCESS in 7m 15s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/2e89795639a14920bb7110af7798de0c : FAILURE in 1h 42m 30s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/dbd6a64a5c6544e7bf3a387005b74984 : SUCCESS in 2h 15m 32s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/a4f160add3ea4c859b2e44c5909f3bd1 : SUCCESS in 1h 08m 08s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/3307c24ac59a4469bbfd6eec98026f50 : SUCCESS in 1h 33m 55s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/e85662e456a743909b0b0f68cfa73391 : SUCCESS in 2h 08m 45s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/ddbaf5398d0f467a89b40d4a11ffd9ed : FAILURE in 2h 18m 17s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/b7f5d61c2772442fbd6859cb7d2fb314 : SUCCESS in 1h 15m 16s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/b04d810f97b847ddaa3f5b132e51c6ac : SUCCESS in 1h 36m 25s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/631af33abc984bcf86227d42641cb192 : SUCCESS in 51m 42s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/a3970b5e4cdb48bfad98700f4f50c7f8 : SUCCESS in 1h 02m 51s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/6f19708ebd8c412cb274df2378abf533 : SUCCESS in 1h 05m 59s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/606d15f39cf6476b922eace1ea579ff1 : SUCCESS in 49m 10s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/4c0e0ab6d18d4c9b8d660d395c2cb544 : FAILURE in 59m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1720a1b128c53cb00ee7f11081919d3cd26e05a3","author":{"_account_id":35476,"name":"taiki kimura","email":"taiki.kimura@ntt-at.co.jp","username":"tkimura"},"date":"2023-02-14 23:51:38.000000000","message":"Patch Set 3:\n\n(16 comments)","accounts_in_message":[],"_revision_number":3},{"id":"220223303a41ecdad7c8aa429a345c9f2a292126","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-02-15 02:42:09.000000000","message":"Patch Set 3:\n\n(14 comments)","accounts_in_message":[],"_revision_number":3},{"id":"26f3954158e1383fee8791bdeb054dd3192cdabe","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-02-15 06:35:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4e6609c42c7ebd6333af1799550d23859a7e7ce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-02-16 07:21:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"86ae8e2e7b70db31df6949ff02d0ea523726f02d","author":{"_account_id":35476,"name":"taiki kimura","email":"taiki.kimura@ntt-at.co.jp","username":"tkimura"},"date":"2023-02-16 07:56:20.000000000","message":"Patch Set 4:\n\n(14 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7a70cc9b574b14c8408be04d762ed7156ab55990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 10:00:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d66db0fb4d4f3695a24707e66ae5c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/256b09d34f7c45ea8dd7b59bda232c77 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/402c5c51fe7d490f8b6bd3f6fed1eba9 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c56e0a49f4741cf9cf11c6ded71e1e4 : SUCCESS in 10m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e0de598aa1d4d9db224571be3f217c3 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30f053ebac94abe8967b68d97690c2b : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baef5a49b24e4c47860d31cc2bc49adb : SUCCESS in 4m 14s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/fb2d5fa9802e4691a735a8356641e922 : SUCCESS in 1h 40m 14s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/fa261e95a33646088f139b0b251520bf : SUCCESS in 2h 12m 59s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/03bc65deb0f245cd92ec66c722cbab43 : SUCCESS in 1h 00m 06s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/7b8f012889b74e18afa3f2a4248b1903 : SUCCESS in 1h 29m 40s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/221414c8ec9a43b5bb801ffa92eb0298 : SUCCESS in 1h 35m 07s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/1895daee95284a1785ce31bc471f4959 : SUCCESS in 2h 16m 57s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/e00308925c9942078a834e31b08a066a : SUCCESS in 1h 16m 19s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c766cd5045744acaab08c2eac9ef0d8 : SUCCESS in 2h 33m 01s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/e659413214714f9498e82c079b33f875 : SUCCESS in 1h 20m 27s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/45973b9569b04ce4a4bc25754137e9b2 : SUCCESS in 53m 42s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/6e11a8a47e604b5686076b8e5186152b : SUCCESS in 1h 12m 27s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/a92de213c97a4e98b7e687a502eb7409 : SUCCESS in 1h 14m 01s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/e5407f0fe75349ddb657a7e538410b9e : SUCCESS in 51m 12s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/6e6f548899b94b75b5d74ef7226ddd56 : FAILURE in 54m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"399bd7dd4de1f3915835d6cfa779ceb338187886","author":{"_account_id":35476,"name":"taiki kimura","email":"taiki.kimura@ntt-at.co.jp","username":"tkimura"},"date":"2023-02-21 02:59:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"32840d0444156a2b3d8aea209c93ecdda3da0c36","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-02-21 08:09:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"532a89878f40d67bffa6ff1d35f8be7492be86a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-02-21 08:56:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f151e1b21538e849a9e6a1cd05988605eedbf27e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 12:09:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bf6bedc899a4e51b9b1ad930e99d2e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c89667c52184531b4cabe41c80117a4 : FAILURE in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90565a7842b749caa928cba40f44279f : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/438be23766c64e30934cd9cd02f12f62 : SUCCESS in 9m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/222f26eb22fc49b2b83467be9eba5ab2 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7ca8e84dff4df0bcbe0751dba63817 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20b3e6f18c4342b1b6b8539c16a2165a : SUCCESS in 3m 55s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/c8f0f8d43612470f8eb27ee2da602c93 : FAILURE in 1h 52m 20s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/7bd0d1d372be4170bf4258e8e416b256 : TIMED_OUT in 3h 03m 25s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/2acb2d9d3fd0471388d7ea91fd931d27 : SUCCESS in 1h 09m 28s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/c068a9d0a9e449bfb68c775c14f1e615 : SUCCESS in 1h 37m 23s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/a5caa11bacf54fc487e73d081e2c1fb6 : FAILURE in 1h 49m 15s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/4f841eb4e0ea4ea4b5c088fb27c8655a : FAILURE in 1h 44m 15s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/a9962998295d4dfe96f3f9d66666d35a : SUCCESS in 1h 25m 44s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e296f1285e36456b8f7c999806454f9a : FAILURE in 1h 45m 05s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/f71de2264bfc4e6382df24051640249a : SUCCESS in 1h 21m 57s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/16a0b2513f714b04b7a5ff473dae990a : SUCCESS in 1h 11m 14s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/49929784766947dc98d6c6bdf5afb949 : SUCCESS in 1h 12m 45s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/2507f61fdbe04e47901ca50b999f140c : SUCCESS in 1h 16m 36s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/2dc3e4d2a85342838ad93e54757b942d : SUCCESS in 56m 50s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/95045801a29941bfa6703fb964c3db2c : FAILURE in 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5c18ee24d094f16b674509c3acc97fbb180441fd","author":{"_account_id":35476,"name":"taiki kimura","email":"taiki.kimura@ntt-at.co.jp","username":"tkimura"},"date":"2023-02-22 01:22:39.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"786233c24a56776e4f1a407db38da3bf46292e74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2023-02-23 22:25:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"8c091d4a6f86d81f753209443a403fb13b7782b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 01:37:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b11ec5d7abce4d7696d85422d7d13062\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7da5fe28310c474caa2330f50e6b3400 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3932c6827b1448999869176f9d138d7 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20775fc88f3640308a5787b93557004f : SUCCESS in 9m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/524431d83db84e37b7b3881f8a6ed510 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e68ecf2243ca48f286838648cbc447d2 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d2a9877fe914385af451ce69adf2f5a : SUCCESS in 5m 43s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/cc2643e9289c4d73ac0063cf5aa5c23a : SUCCESS in 2h 36m 12s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/d329c25b4e514f3691c44a044e1f6e09 : SUCCESS in 2h 04m 51s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/ba11caad6a2d4f3fa96902d0fe0893a7 : SUCCESS in 1h 28m 29s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/f9238b851e52493c8e210957ffc531d5 : SUCCESS in 1h 18m 15s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/1d5f4e5fff0946bbb37e7b78284f7f43 : SUCCESS in 1h 25m 55s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/7744c8e2abc54b07b869face7d6c13ab : SUCCESS in 1h 55m 35s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/7595e9b56e2a49958b353ae1c85ff6d5 : SUCCESS in 1h 46m 16s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3608b57cf7b4cac841ebc4f45e27e98 : SUCCESS in 2h 01m 02s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/689d80d4d9bb4f01b6e0c605786be8cb : SUCCESS in 1h 04m 27s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/ddc50e8816a845bb8aa5cf828f6a19fb : SUCCESS in 1h 15m 32s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/aeb4e58f136846c58733f84b1e91fd57 : SUCCESS in 56m 53s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/aa50ecdfc40949548ac9676db558428c : SUCCESS in 55m 32s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/54a065beada646049d69ba27a88c81a4 : SUCCESS in 1h 17m 52s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/e4f51647b072458ea67b7499ebc01362 : FAILURE in 49m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8eb05d4c8e6bd6dc4993558c331d7be9fd1bc805","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2023-02-27 00:27:25.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ae036ef9d617a72356c3ac8e022acbfe99c233ac","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-02-27 04:17:57.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6c3a63e692a340f71ca3afdcf3d8d95cf78e47a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-02-27 05:38:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"798160862e4a0ea22be5ea056934fe842452d840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 08:22:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35d531de07a94680a35e73f066ff8b09\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/385d6712cc4a4c95ba5f6e4f0464a996 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/342be92fbd5b40e080746c8b357435a1 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a867290c668c456a80f51c7ff6e064a4 : RETRY_LIMIT in 1m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3770b65058f94f188c3de227110f8bfa : RETRY_LIMIT in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f54a41818f84ecb9cf9f5a32d0087ba : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/175ee1ed131d4bb9a8068be40c69e824 : SUCCESS in 4m 28s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/f2bada1b401b42d293404218d964045c : SUCCESS in 2h 20m 11s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/e772044d526347bb99026b8c2d36e5a0 : SUCCESS in 2h 08m 24s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/868e77f0c1bf4e1fbe98440c5fda1294 : SUCCESS in 1h 11m 08s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/c51fb942648f4aefb385c6443af00e28 : RETRY_LIMIT in 50m 02s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/17e237508df4423f9bcd1a01ad288aef : SUCCESS in 1h 35m 48s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/9ffaf41296cd426d927fdd4078805840 : SUCCESS in 1h 57m 53s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/1b6f96807e97491e849f2b43acd5673a : SUCCESS in 1h 09m 13s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f2c17bdf74d4fa9b9047d4ac4d5aeb8 : SUCCESS in 2h 26m 37s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/1277630ca8cc4f319797cccfe245b874 : SUCCESS in 1h 26m 34s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/e73e7e8dbaec4e92bdf93cd74dd0ad8d : SUCCESS in 1h 05m 11s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/68ef0028ab024f7a8e2f063b1832dc19 : RETRY_LIMIT in 4m 46s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/7b02c7810bfd43e8ba819f4623cd6d51 : RETRY_LIMIT in 4m 44s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/702a04a1b71b441f9402e3714e6deb0b : SUCCESS in 1h 01m 00s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/49192792b1364c87a253cd525184fe36 : FAILURE in 58m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"162adfdf6652069e851bac7507c9bc989bf7415e","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2023-02-28 00:14:07.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b251dbb65ed610a0b0443a6f08436e568d5bf10c","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2023-02-28 00:15:33.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"361cb62744a8c12a76468938186b91beb44fc02c","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2023-02-28 02:43:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2d34a1a4a78fe4a3a3885dfe1c0c4d7e53833ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 06:07:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/227d1dece07c4592aeff6fcb190e5ad5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fef7fe5074454c07918e499d4eeea3cd : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6939e90b904498a4b931bac5efec0c : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f96004225b249ff88fbd1e96c696aec : SUCCESS in 8m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94dc35af2e4e48c9986f3a04d6553d35 : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96f761d30b964d409298e03d16ccb099 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91a059d495c74cd9ab15b71fc4e875c0 : SUCCESS in 3m 47s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/0c9a4f38817f43a29b0f81ee48581f5b : SUCCESS in 1h 31m 10s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/fd3d6d4c96db4478af2c12160a288bfe : SUCCESS in 2h 06m 02s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/14d8be9945584c5292c21f3c558e0deb : SUCCESS in 52m 10s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/e5096ed685e14e23bdcc150da8ecfeec : SUCCESS in 1h 40m 42s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/af61b9d4ab584f4a9a92532521c6287b : SUCCESS in 1h 46m 51s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/9941adbb51eb4f3ca2e372e0af2f9a04 : SUCCESS in 2h 18m 19s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/a65f1006512c494798307aa74132c5b3 : SUCCESS in 1h 16m 25s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6be555420194ee0a5e01f8efe929d59 : SUCCESS in 2h 13m 49s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/80ce1b1f295a402e8659fa805d5c37a1 : SUCCESS in 1h 11m 58s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/85f42059dd2a41899060307bfb2de127 : SUCCESS in 1h 06m 52s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/139d33508a244a269a58423419abaad3 : SUCCESS in 1h 03m 09s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/67ee988e4e6d4c9aa6d499623a94be6d : SUCCESS in 1h 20m 59s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/7eb416473b634bee8f2e2cbb42c29e9b : SUCCESS in 58m 28s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/2feefbd8e93f44f9a2337e8175bc1d12 : FAILURE in 1h 05m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f5ca85925b73f14c0db52227724a2199f42132f7","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2023-02-28 07:07:16.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6053eb442be3a423c9ca60d3f3b4ff35210a9911","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-02-28 16:32:38.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"adc8cc4d454f477034f791cd9a04c47ac47780ce","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-03-01 02:09:14.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5f0ebe09f925f131d2a74f554c93b1e1639b45cd","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-01 04:25:49.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c91a3f276252a64f26ea6700c1fd1a0f61e4c01f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-01 05:16:58.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6640fbe7966fd141903a415c1a09cfd28a945751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 07:35:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29987d6b16ea473f8cc50f01379f47f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8456487bb2e41a0a1d22882b618bc42 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dba0c353f364ed092da253e903d8712 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a56658cbed441339bfde4bdcca484f5 : SUCCESS in 9m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/387b08cbdac646bba2ba67f111de4271 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f5031766c847fba979536b3b3d3499 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/553761c39dd5405384b4aff37e563e15 : SUCCESS in 3m 27s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/1bfdc7a258e04f789e69e4dcef4f56e5 : SUCCESS in 1h 49m 15s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/4ca32ba23c664519a880f72a4182c84a : SUCCESS in 1h 57m 16s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/7f117ff45f4e44e6b893af4d820fb26d : SUCCESS in 45m 35s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/3cf9211c20d145b98dde3c9642a54957 : SUCCESS in 1h 32m 04s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/ffa48e755b4c4ce3bbe861da34a0702a : SUCCESS in 1h 28m 13s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/4e53b2523d294bf7a3f19c771315578e : SUCCESS in 2h 07m 41s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/59013b43165f4ad9b1cd2f01d6110b0a : SUCCESS in 1h 18m 23s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a2703181a9a4d1eaedb2c1b2bbd2200 : SUCCESS in 2h 10m 32s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/6e8a572d0a5640bead17601dce32ad96 : SUCCESS in 1h 05m 53s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/96b56344868f4435897773f7801df1ed : SUCCESS in 49m 58s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/6dfa635230fb4bdfa48ddfc0ec45f7a2 : SUCCESS in 1h 13m 46s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/45d1fc1a35b24a44adcce03f503b926c : SUCCESS in 1h 07m 27s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/bd549caa3173472291de645ff0f6f95a : SUCCESS in 50m 28s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/f432f540934c419eb6cbef9956251cf5 : FAILURE in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"12b80186c1e8aba8fb5f9dc44d2f04422629c246","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-03-01 07:53:39.000000000","message":"Patch Set 8: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"86a4845a1919c280cc351690c430a198a0f83526","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-01 08:30:26.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1d927dce3ec9594dab232bd01d5dcc3b34c51b7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-01 08:37:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3e493a38ecc4649068b861f0e85e062469a65c24","author":{"_account_id":31668,"deleted":true},"date":"2023-03-01 09:58:52.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"6a0e98997fb8574914105d9d393c209a4e3bdfa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 11:25:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01d78b2c5972476aa85fc6a72d0ae9c4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b73803609f1845a2853fcff3c1f41b8c : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d80d468059c4e48b44f4b0841851a21 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29536a0650ad4b21b459b6ba0d3813d8 : SUCCESS in 8m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88cd38ba93994583bbff38b810996e77 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c681edd493da49b09e26caf57374f979 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddffdd9df76f4c64b463912c7ce60ca4 : SUCCESS in 3m 33s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/d6a63f4ef61949b5b1d73e3582ebd451 : SUCCESS in 1h 38m 45s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/dd407639946d46aa8a6614546014638a : SUCCESS in 2h 12m 53s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/d0ec3db61ecd42a1ac167b3a3ff80500 : SUCCESS in 57m 30s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/e74753630fb34d488e8a93495d27e0d6 : SUCCESS in 1h 19m 49s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/a2d56c2289b94131b41fe8b943f2d719 : SUCCESS in 1h 23m 00s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/9ce0802e9b744d6486e6ad356869e3ab : SUCCESS in 2h 19m 38s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/994237b29035430f9f0e9e1c45735a69 : SUCCESS in 59m 43s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a01afe3a38fb4865a0a5f65a40723b6f : SUCCESS in 2h 00m 45s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/f15449fe22404743b2230b8b87aa1cc0 : FAILURE in 1h 10m 35s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/0439953bcc81401b8c9b4916b17cabd7 : SUCCESS in 1h 07m 00s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/41e7216f38824bf78f4a849376793a7c : SUCCESS in 1h 08m 13s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/ba484ee133f14c13a678dea12f95188b : SUCCESS in 1h 05m 41s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/2f822cc7771d40f0952dc7be9426e6c7 : SUCCESS in 57m 16s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/f5bb21e76cf142d0941eebce1e2c9110 : FAILURE in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f3f57887e645c57a9d9f4a2d48dbccd38c9258e5","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-02 00:12:48.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7f14ef55013875a53a99df8d6219725cc5569f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-02 00:29:04.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"40551cbbd73ca3df168830775e6e76003afa1d76","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-03-02 00:45:49.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b59020c21d4cf70509584e5fd184c24f0012fc82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 02:54:38.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84fc7e310f6648dfb6ee3d60e452c7b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8954e212cfdd431ea142d72113aef635 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7c5e4ac8e041e28a5cb2daf567b893 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56947d8004b6413f97b8c23838af7cee : SUCCESS in 8m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b150f2fb6cf4b1882783bfc2c0dab3b : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6d0f9910564019ade7a50ab45893ba : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d15f9dac469249cf9d5d18e486da40fd : SUCCESS in 3m 56s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/b50fd1f5862e4bee9ffee45c3ef40623 : SUCCESS in 1h 33m 06s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/39197a9085d241a0a2d7ce4d26878597 : SUCCESS in 2h 07m 20s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/5041477ab0bc4972ada54cc893dd97bb : SUCCESS in 1h 01m 03s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/44aa9b0458794ff5af5c20c4892b6a58 : SUCCESS in 1h 30m 03s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/f4d7ed944068474ebe0c9fbf8b13afbc : SUCCESS in 1h 23m 06s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/0dfcbb5e2ea24aff8d4f6b0395a57bd7 : SUCCESS in 2h 09m 49s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/e997f404710f4bac862bbf978db6eeef : SUCCESS in 1h 24m 16s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/180beedb78934aa7b4aaa2bae4418dd9 : FAILURE in 1h 57m 50s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/9207eb4210b1409eb5c800ff08ad3a98 : SUCCESS in 1h 36m 07s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/40a20a40724f411b8fcb5ddae8940672 : SUCCESS in 51m 18s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/6b89c9d0555149b08259ad36a3970810 : SUCCESS in 1h 24m 25s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/1acbf68908ba4db7bb7cc8a1a5f88e92 : SUCCESS in 1h 21m 07s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/5d604f62e11b4a2487a0cf7deb8888b2 : SUCCESS in 49m 20s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/ae1b3221a17d49658f1597de584b7720 : FAILURE in 1h 07m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8a6b43ea1e8a326953b156ff28abfb99f57ae25c","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-02 03:39:40.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2e8283f0a4c0c76806731831c1640c6337a965bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-02 04:43:55.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c285bd9583104824eae8cb7dd581ce515455c9bc","author":{"_account_id":31668,"deleted":true},"date":"2023-03-02 05:24:25.000000000","message":"Patch Set 11: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"37968124d73dccc0a6f33002c2b630878f5a6e18","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-03-02 05:29:35.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"59c0b097af0ece3cf730b730992d9910c91b86b4","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-02 06:37:29.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"87b0bdc342e3e0ffbb8c79b48264847c1887aa90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"date":"2023-03-02 07:08:38.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9956592698f6d797198d0cafdc71aae40d0097be","author":{"_account_id":31668,"deleted":true},"date":"2023-03-02 07:29:42.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"5d9d352fe78fc51b56e55c92ccded37b52bb8b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 09:32:32.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37360cfb42ea401ca3594cb1a8ae44b5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b17243d80c2f464fb64018175c5b4ede : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/639eb71f9dad4c288ec5dfa661a700c4 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f318aa750e44266aa930d2057ae07fa : SUCCESS in 8m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24784125381b4b33a9194ab06bae6fd5 : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd70a33237e4322a1cfb2e164b57a92 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd56e818e6734814962573a48e4dcfb3 : SUCCESS in 3m 33s\n- tacker-functional-devstack-multinode-legacy https://zuul.opendev.org/t/openstack/build/855171f8474749e389c160ea5a6212ac : SUCCESS in 1h 41m 21s (non-voting)\n- tacker-functional-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/3a0d3143940f439581b7aaff63b96211 : SUCCESS in 2h 13m 53s\n- tacker-functional-devstack-multinode-sol-separated-nfvo https://zuul.opendev.org/t/openstack/build/fefa9286ec404bf68762f1ae4cb0cc4f : SUCCESS in 1h 05m 12s\n- tacker-functional-devstack-multinode-sol-kubernetes https://zuul.opendev.org/t/openstack/build/9f28dcfcc1f0449e9e4937bab6c713a7 : SUCCESS in 1h 31m 51s\n- tacker-functional-devstack-multinode-libs-master https://zuul.opendev.org/t/openstack/build/d984dbc18ee64110a6e740029637ee9d : SUCCESS in 1h 31m 48s (non-voting)\n- tacker-functional-devstack-multinode-sol-v2 https://zuul.opendev.org/t/openstack/build/c0d961500c2a4dd6b4732ae566d9a224 : SUCCESS in 2h 15m 55s\n- tacker-functional-devstack-multinode-sol-separated-nfvo-v2 https://zuul.opendev.org/t/openstack/build/8ddd3cb263754197a812026814352d49 : SUCCESS in 1h 14m 43s\n- tacker-functional-devstack-multinode-sol-v2-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b0cc9c2da11406493be9c79a1f295f3 : SUCCESS in 2h 14m 55s\n- tacker-functional-devstack-multinode-sol-kubernetes-v2 https://zuul.opendev.org/t/openstack/build/c82a3ad01cd94aadafe7893ac40c6154 : SUCCESS in 1h 23m 39s\n- tacker-functional-devstack-multinode-sol-multi-tenant https://zuul.opendev.org/t/openstack/build/5f52371f49b840b5aa5f32f1d62962e6 : SUCCESS in 56m 12s\n- tacker-functional-devstack-multinode-sol-kubernetes-multi-tenant https://zuul.opendev.org/t/openstack/build/3075b4ed7e71435080191ce383bf3f3b : SUCCESS in 1h 11m 29s\n- tacker-functional-devstack-kubernetes-oidc-auth https://zuul.opendev.org/t/openstack/build/d92b8946487042ffaa2c86efbe54dc1e : SUCCESS in 1h 09m 19s\n- tacker-functional-devstack-multinode-sol-v2-az-retry https://zuul.opendev.org/t/openstack/build/6c6ca015c7fa40f7b9400d102a1d05f4 : SUCCESS in 59m 49s\n- tacker-compliance-devstack-multinode-sol https://zuul.opendev.org/t/openstack/build/b6c49d8890c84e188c64905fa493add7 : FAILURE in 50m 56s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"42ac1ac0b3cac4708d79cbf1ca35f05e87c4f4af","author":{"_account_id":31668,"deleted":true},"date":"2023-03-02 09:48:23.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"59e6ed90a7d2277c897e00926bfe7b5fe19ecc1e","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-03-02 14:07:26.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"1fe671e5b4efab4c0d154b2460b7811803d20ad9","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2023-03-02 23:53:44.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"090c849dc276380838365622f184cd14517956e2","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-03-03 00:02:15.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"ec8ab559b9fff91a7d9b5ea5d2557dfb65277874","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:02:35.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"35654a27bb8cfe9c97ed94ff1d1d202d3715f0ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:20:02.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a6fbee6b89e46178c80038b0017630b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2744eb6785d747fd8cfac39b15c3aefe : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ac635f6d9d347db898f4b10adbcd93c : SUCCESS in 9m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b75a55f58f4d49f7be527c5bac6813da : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a90fcb40e2451199775e44e27beccd : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1db7c01ac2341768f35bda7884a1ca5 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":12},{"id":"1389141b5869cf6b7f0bd18c408811781afd4a94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:20:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"06ca68da4882ce226e5c43ec6634f342610a602f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:22:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb11490b6ca84fd7b71850883b08c6cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce8b6e99b88446486e248140970efb9 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7cb6097fd0447faa97b6af3e22bb8c6 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"a423efca837ed5b4914fc56a114e3f342e794662","revisions":{"685b3f448d25058c98f3b676a403584d356439f4":{"kind":"REWORK","_number":1,"created":"2023-01-31 03:41:21.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/1"}}},"commit":{"parents":[{"commit":"cb069e5dbbcb3b6df6ecb35c43bcc732c40d1453","subject":"Placement enhancement enables to AZ reselection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/cb069e5dbbcb3b6df6ecb35c43bcc732c40d1453"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"subject":"[WIP] Support Coordination API in change_vnfpkg","message":"[WIP] Support Coordination API in change_vnfpkg\n\nThis patch supports Coordination API when VNFc are updated by\nchange_vnfpkg v2 API.\n\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/685b3f448d25058c98f3b676a403584d356439f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/685b3f448d25058c98f3b676a403584d356439f4"}]},"branch":"refs/heads/master"},"402ff6f2f7c8f54366cfa750dab136f3947b2dec":{"kind":"REWORK","_number":2,"created":"2023-01-31 06:23:49.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/2"}}},"commit":{"parents":[{"commit":"cb069e5dbbcb3b6df6ecb35c43bcc732c40d1453","subject":"Placement enhancement enables to AZ reselection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/cb069e5dbbcb3b6df6ecb35c43bcc732c40d1453"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 06:20:53.000000000","tz":0},"subject":"[WIP] Support Coordination API in change_vnfpkg","message":"[WIP] Support Coordination API in change_vnfpkg\n\nThis patch supports Coordination API when VNFc are updated by\nchange_vnfpkg v2 API.\n\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/402ff6f2f7c8f54366cfa750dab136f3947b2dec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/402ff6f2f7c8f54366cfa750dab136f3947b2dec"}]},"branch":"refs/heads/master"},"8787dc1a19a9a905045c5e7c58ffccac6ce9e508":{"kind":"REWORK","_number":3,"created":"2023-02-08 01:27:53.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/3"}}},"commit":{"parents":[{"commit":"0309368d3ab774d2370fef8e74f9c0c0a3c44140","subject":"Placement enhancement enables to AZ reselection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/0309368d3ab774d2370fef8e74f9c0c0a3c44140"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-02-08 01:19:10.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8787dc1a19a9a905045c5e7c58ffccac6ce9e508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8787dc1a19a9a905045c5e7c58ffccac6ce9e508"}]},"branch":"refs/heads/master"},"547968064b2a34fc9f4c857d0f66cc3c65d35318":{"kind":"REWORK","_number":4,"created":"2023-02-16 07:21:16.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/4"}}},"commit":{"parents":[{"commit":"25459c257161ab2c408553052e85019d28caa3df","subject":"Placement enhancement enables to AZ reselection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/25459c257161ab2c408553052e85019d28caa3df"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-02-16 04:52:36.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/547968064b2a34fc9f4c857d0f66cc3c65d35318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/547968064b2a34fc9f4c857d0f66cc3c65d35318"}]},"branch":"refs/heads/master"},"e8020c79e7c87254c0a191589532edb12d8ff0f5":{"kind":"REWORK","_number":5,"created":"2023-02-21 08:56:17.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/5"}}},"commit":{"parents":[{"commit":"1cb3077ace88554923a8452fad1b2842e864e996","subject":"Merge \"Restore overwritten documents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/1cb3077ace88554923a8452fad1b2842e864e996"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-02-21 08:43:42.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e8020c79e7c87254c0a191589532edb12d8ff0f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e8020c79e7c87254c0a191589532edb12d8ff0f5"}]},"branch":"refs/heads/master"},"4134dd06e90a8f4065d7d1b260f8fc3a7ca89ad1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-02-23 22:25:55.000000000","uploader":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"ref":"refs/changes/52/872252/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/6"}}},"commit":{"parents":[{"commit":"844de7fb20a954bdcee540c4b3f81c27d376271c","subject":"Merge \"Fix oidc\u0027s vim not applied to FT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/844de7fb20a954bdcee540c4b3f81c27d376271c"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Itsuro Oda","email":"oda@valinux.co.jp","date":"2023-02-23 22:25:55.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4134dd06e90a8f4065d7d1b260f8fc3a7ca89ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/4134dd06e90a8f4065d7d1b260f8fc3a7ca89ad1"}]},"branch":"refs/heads/master","description":"Rebase"},"00fba8e0eb28b2273ef886cde67dd450de85fbfc":{"kind":"REWORK","_number":7,"created":"2023-02-27 05:38:28.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/7"}}},"commit":{"parents":[{"commit":"0dda8d07b68f7335d0963b3e92afdc899ed3d8d3","subject":"Merge \"OAuth 2.0 Mutual-TLS Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/0dda8d07b68f7335d0963b3e92afdc899ed3d8d3"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-02-27 05:34:06.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/00fba8e0eb28b2273ef886cde67dd450de85fbfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/00fba8e0eb28b2273ef886cde67dd450de85fbfc"}]},"branch":"refs/heads/master"},"8beff93ebafc365f6e3d84de60fd48e520ace86b":{"kind":"REWORK","_number":8,"created":"2023-03-01 05:16:58.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/8"}}},"commit":{"parents":[{"commit":"d457da69d96bfe0c3586abfb59797d51462a5806","subject":"Merge \"Extend vim register function for helm in v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/d457da69d96bfe0c3586abfb59797d51462a5806"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-03-01 05:08:55.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8beff93ebafc365f6e3d84de60fd48e520ace86b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/8beff93ebafc365f6e3d84de60fd48e520ace86b"}]},"branch":"refs/heads/master"},"485f8e8224ddfadecb5d8638da89d8bc49ec89b9":{"kind":"REWORK","_number":9,"created":"2023-03-01 08:37:50.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/9"}}},"commit":{"parents":[{"commit":"d457da69d96bfe0c3586abfb59797d51462a5806","subject":"Merge \"Extend vim register function for helm in v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/d457da69d96bfe0c3586abfb59797d51462a5806"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-03-01 08:32:13.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/485f8e8224ddfadecb5d8638da89d8bc49ec89b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/485f8e8224ddfadecb5d8638da89d8bc49ec89b9"}]},"branch":"refs/heads/master"},"034ef80b4d35456869458a3fbbd1fccb26478b31":{"kind":"REWORK","_number":10,"created":"2023-03-02 00:29:04.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/10"}}},"commit":{"parents":[{"commit":"d457da69d96bfe0c3586abfb59797d51462a5806","subject":"Merge \"Extend vim register function for helm in v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/d457da69d96bfe0c3586abfb59797d51462a5806"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-03-02 00:26:52.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/034ef80b4d35456869458a3fbbd1fccb26478b31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/034ef80b4d35456869458a3fbbd1fccb26478b31"}]},"branch":"refs/heads/master"},"e9516936fe3c8424c5d88dfe9f62708789835441":{"kind":"REWORK","_number":11,"created":"2023-03-02 04:43:55.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/11"}}},"commit":{"parents":[{"commit":"d457da69d96bfe0c3586abfb59797d51462a5806","subject":"Merge \"Extend vim register function for helm in v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/d457da69d96bfe0c3586abfb59797d51462a5806"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-03-02 04:42:46.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e9516936fe3c8424c5d88dfe9f62708789835441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/e9516936fe3c8424c5d88dfe9f62708789835441"}]},"branch":"refs/heads/master"},"a423efca837ed5b4914fc56a114e3f342e794662":{"kind":"REWORK","_number":12,"created":"2023-03-02 07:08:38.000000000","uploader":{"_account_id":34226,"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","username":"k-fujimoto"},"ref":"refs/changes/52/872252/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/52/872252/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/52/872252/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/52/872252/12"}}},"commit":{"parents":[{"commit":"d457da69d96bfe0c3586abfb59797d51462a5806","subject":"Merge \"Extend vim register function for helm in v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/d457da69d96bfe0c3586abfb59797d51462a5806"}]}],"author":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-01-31 03:30:16.000000000","tz":0},"committer":{"name":"Ken Fujimoto","email":"fujimoto@valinux.co.jp","date":"2023-03-02 07:07:24.000000000","tz":0},"subject":"Support VNF LCM Coordination IF in change_vnfpkg","message":"Support VNF LCM Coordination IF in change_vnfpkg\n\nThis patch supports Coordination IF when VNFc is updated by\nchange_vnfpkg v2 API.\n\n* Modified the coordinateVNF script in the sample package\n  to execute the coordination IF.\n* Modified change_vnfpkg operation in Functional test\n  to execute coordination IF.\n\nImplements: blueprint add-sample-coordinate-script\nChange-Id: Ie0c8b30df9d6e8c9ae4a6ba9e894203561475022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/a423efca837ed5b4914fc56a114e3f342e794662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/a423efca837ed5b4914fc56a114e3f342e794662"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[]}
