)]}'
{"id":"openstack%2Ftacker~987813","triplet_id":"openstack%2Ftacker~master~I112f9455a2b73d7acb3f933e866581f21242f83a","project":"openstack/tacker","branch":"master","attention_set":{},"removed_from_attention_set":{"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2026-09-04 22:16:40.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2026-09-02 00:19:59.000000000","reason":"\u003cGERRIT_ACCOUNT_17255\u003e replied on the change","reason_account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-04 22:32:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I112f9455a2b73d7acb3f933e866581f21242f83a","subject":"Fix typo","status":"MERGED","created":"2026-05-08 08:17:18.000000000","updated":"2026-09-04 22:33:44.000000000","submitted":"2026-09-04 22:32:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987813","meta_rev_id":"70311e25fbe8c99c915c60b6bd94b9b278b992ca","_number":987813,"virtual_id_number":987813,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-04 22:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2026-09-04 22:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2026-09-02 00:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":1,"date":"2026-09-04 22:16:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-08 11:07:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 19:24:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2026-09-02 00:20:17.000000000","updated_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"769e0acc2d5a9de0637a1568e9d02c072780590a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-08 08:17:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"953b08801f3e01ea02ae98675b27f7b93c85308c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-08 08:17:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5de461d8793fd6fd03cf2c6ea43987c8fc46e04e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-08 08:18:02.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"6a6006d5a72582f1255650214ecc799f14c1a9bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 11:07:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b591ba8c3b8c4dd984fa7f12706d9e78\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6668c25d337442bd8121571538781b80 : FAILURE in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd1aca307ed4741a420c44b809bbe5e : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a3d964e4bc04b36b3c214694f6c8cb0 : FAILURE in 7m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/334324f13dc14db781775b6adf607b91 : FAILURE in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a058032d2ef46ad81afbddb45a18e2c : FAILURE in 11m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304c254e05ef4622a97adc72305e2223 : SUCCESS in 6m 46s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/5da5b782f495462cb13610509d8c8487 : RETRY_LIMIT in 44m 46s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/f84997ef117745f78d6a8cbcc95aeb73 : RETRY_LIMIT in 33m 43s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/b7cbb40567434baa9b09de1ee7a306de : RETRY_LIMIT in 45m 02s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/5746af261bc449d58dc8895926d13b71 : RETRY_LIMIT in 25m 08s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/e8d48188e8714fbf9b71c3cf43b280e4 : RETRY_LIMIT in 49m 54s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/72a839af714a4a69b392da2dd635f5fe : RETRY_LIMIT in 24m 21s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/95cfe63f25824cfca26ea298ba42e32f : RETRY_LIMIT in 44m 17s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/13f9fcf905c044d0a78b3190babe5e42 : RETRY_LIMIT in 36m 41s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/a756131f84df48d0917569667d918f4c : RETRY_LIMIT in 58m 54s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/e8549f1149c34d0cbc2f506e0ce4fd27 : RETRY_LIMIT in 20m 10s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/a8594bd4cafe4f47887d89727cd5aaae : RETRY_LIMIT in 32m 17s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/24c072b16f7f436d8dc49769c16a419b : RETRY_LIMIT in 9m 55s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/cb95d3d78fed4c988cdc14b19babc575 : RETRY_LIMIT in 30m 12s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/fa11b1b549d441c58eabfb1698ce9056 : RETRY_LIMIT in 42m 53s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/cd4203f287074483b60b2f8663c38938 : RETRY_LIMIT in 24m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3547dd3da953f9e7a127792b15358c41a3c7b98a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:24:22.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1747ed501fdd0fa8b9d6e91bfba3d11693cce987","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:24:30.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"91421cbbbf877db0fd77559f48a42d28686d2799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:47:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ef06d7684ce4e19b896a648b0afae70\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d29b21fa81a040a8a96d4dda47fe4257 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a14d66fdbd5844a98047878185fa10f9 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf34632916f0440f91430ecb5c235074 : SUCCESS in 10m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4392afa1f4754fa9ac9fe72126f7abd5 : SUCCESS in 9m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c2e9ee1cff0428a824486954759493b : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20af29920cde4948a4a9dca5980c6021 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcbf9a9230eb48398ebbdde2557abdc0 : SUCCESS in 6m 53s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/bcf073c0bcaa4b348c2ba9da0245b17b : SUCCESS in 1h 23m 10s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/baa2eea459a24d49a341fb08b79bbfc7 : SUCCESS in 29m 14s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/c8a915a9ba3e433da42f845ee87136c5 : SUCCESS in 51m 28s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/44c0460e32a042069ae7717ae936ea99 : SUCCESS in 1h 05m 57s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/9f681dd576224beeb98c038319222751 : SUCCESS in 1h 00m 13s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/8b14bd6c4aca4720b12f82a8525ae98e : SUCCESS in 1h 01m 04s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/6be1c6f71e0b4f3390563d096e1eba91 : SUCCESS in 1h 06m 52s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/782cc800cfe742f6b6df714e29058f10 : SUCCESS in 44m 53s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/7849daeee68c4c7f9913f6e531e7ade2 : SUCCESS in 33m 01s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/bf84bc256ebc4348900d6a637dfc7abb : FAILURE in 53m 33s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/2343ca06f928442f89953ebe36027b20 : SUCCESS in 50m 12s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/3e9b3dd490b74522bf44a790a3cfc7dd : RETRY_LIMIT in 8m 33s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/662c5af68a6448e4b25671e1015ef3b1 : SUCCESS in 54m 46s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/8e00a38bc4d04d33adc8e6a8da23e2af : SUCCESS in 1h 14m 22s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/777442cd34214692a20d7e221fc92f53 : FAILURE in 2h 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"45e9b387565e8781a574f64f1d8709f6ca89c3a0","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2026-09-02 00:19:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"988c2a7aaf062cf18e63e4fe11d45cf7597673bb","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-09-04 22:16:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5b99bb40c08f128095cd9d1362e3b09557798cfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 22:16:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e05dc4fe2943b0d43ca3fa205d5d20bf02f7b74b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 22:31:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a9f66c4c5f342ab8eb941a667a31faa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd35747e1ad43d9a566ac2ce3bc182a : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c0a5e2c91944f9eb4b50b4f6b314ff1 : SUCCESS in 9m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77bc36a490e7471394620cfc7247a489 : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6223b6c88d41309b0dbdc6e57fef47 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b85cd92238a453e81a794b2ded4afe9 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":4},{"id":"7d21a0150b84395b292f9e718af73c8ffe110d3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 22:32:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"70311e25fbe8c99c915c60b6bd94b9b278b992ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 22:33:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52be0ba62f7c499dae1f3a20e4ed146e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83b52d4ce6bc4d7aae2803cb4d3cbed9 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5c7f6552eb01ead268cf14f1c4f06aa4df004b55","revisions":{"27b1c09d9b1b4474854bc8b3d5aec13c462ba3a2":{"kind":"REWORK","_number":1,"created":"2026-05-08 08:17:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/13/987813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/13/987813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/13/987813/1"}}},"commit":{"parents":[{"commit":"310e86944b5c5751876f986425a2e2f4486784a2","subject":"Merge \"Add periodic job to validate Tacker documentation links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/310e86944b5c5751876f986425a2e2f4486784a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-08 08:17:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-08 08:17:18.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I112f9455a2b73d7acb3f933e866581f21242f83a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/27b1c09d9b1b4474854bc8b3d5aec13c462ba3a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/27b1c09d9b1b4474854bc8b3d5aec13c462ba3a2"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"067e01563ae5c88b3e29d16c679f4143d2131587":{"kind":"REWORK","_number":2,"created":"2026-05-08 08:17:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/13/987813/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/13/987813/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/13/987813/2"}}},"commit":{"parents":[{"commit":"310e86944b5c5751876f986425a2e2f4486784a2","subject":"Merge \"Add periodic job to validate Tacker documentation links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/310e86944b5c5751876f986425a2e2f4486784a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-08 08:17:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-08 08:17:58.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I112f9455a2b73d7acb3f933e866581f21242f83a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/067e01563ae5c88b3e29d16c679f4143d2131587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/067e01563ae5c88b3e29d16c679f4143d2131587"}]},"branch":"refs/heads/master"},"fb2b8ed6936677015f909c4f6ca3be2d945c6a72":{"kind":"REWORK","_number":3,"created":"2026-09-01 19:24:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/13/987813/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/13/987813/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/13/987813/3"}}},"commit":{"parents":[{"commit":"310e86944b5c5751876f986425a2e2f4486784a2","subject":"Merge \"Add periodic job to validate Tacker documentation links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/310e86944b5c5751876f986425a2e2f4486784a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-08 08:17:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:24:22.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I112f9455a2b73d7acb3f933e866581f21242f83a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/fb2b8ed6936677015f909c4f6ca3be2d945c6a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/fb2b8ed6936677015f909c4f6ca3be2d945c6a72"}]},"branch":"refs/heads/master"},"5c7f6552eb01ead268cf14f1c4f06aa4df004b55":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-01 19:24:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/13/987813/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/13/987813/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/13/987813/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/13/987813/4"}}},"commit":{"parents":[{"commit":"3f9a82be95b0c094961aa3a549d440e468f9b248","subject":"Merge \"Replace eventlet green subprocess with stdlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/3f9a82be95b0c094961aa3a549d440e468f9b248"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-08 08:17:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:24:30.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I112f9455a2b73d7acb3f933e866581f21242f83a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/5c7f6552eb01ead268cf14f1c4f06aa4df004b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/5c7f6552eb01ead268cf14f1c4f06aa4df004b55"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"310e86944b5c5751876f986425a2e2f4486784a2","ours":"fb2b8ed6936677015f909c4f6ca3be2d945c6a72","theirs":"3f9a82be95b0c094961aa3a549d440e468f9b248","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
