)]}'
{"id":"openstack%2Ftacker~989906","triplet_id":"openstack%2Ftacker~master~Id0a338f747501177e9f73b37048f64d166b899c5","project":"openstack/tacker","branch":"master","attention_set":{},"removed_from_attention_set":{"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2026-07-24 04:24:15.000000000","reason":"\u003cGERRIT_ACCOUNT_17255\u003e replied on the change","reason_account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-10-03 01:22:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id0a338f747501177e9f73b37048f64d166b899c5","subject":"Replace OVN Metadata agent with OVN agent","status":"MERGED","created":"2026-05-24 16:23:22.000000000","updated":"2026-10-03 01:24:27.000000000","submitted":"2026-10-03 01:22:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989906","meta_rev_id":"e1238a6480c451119fc90dfc4d8a392580e13ed9","_number":989906,"virtual_id_number":989906,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-03 01:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":2,"date":"2026-10-03 01:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2026-07-24 04:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":1,"date":"2026-05-28 02:13:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":1,"date":"2026-10-03 01:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-24 17:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-28 02:13:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-06-05 05:24:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2026-10-03 01:09:30.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"d397e993985f19e91bc41a92e8faf6d0bda6dbb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-24 16:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eadb50c505292f766dc0adb233ae398466083446","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-24 16:25:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"950c68f3b182c59a664cfc32208c1342d4b75f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 17:45:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/202f4e76cf6144959ce6e188735796f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55341017dc8c41dcb4dd7329714efb66 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e682c046474824a9135da793d88ed5 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84b00b78b12147739f03bffb64461ffc : SUCCESS in 8m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ccdb4232dd83420688a2c311a8b39a6a : SUCCESS in 8m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f3b4523f30d4bde81759aa6ef1509a2 : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9f0e1d1d99464a84f83630a43a5f41 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/540b279ef29a410f96ca202265aa69b8 : SUCCESS in 5m 26s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/2a8deab5a3a140bdb8bed2fbde8346d4 : SUCCESS in 1h 11m 16s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/9fa2a6f2601b47e8afeee30f2a605a79 : SUCCESS in 53m 32s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/dc218cadb2c241a09726840ec63a4359 : SUCCESS in 1h 01m 37s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/83254fae820b4a86969a548cffdb1465 : SUCCESS in 1h 07m 58s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/865cafcc26c24063976f9474e8f2db24 : SUCCESS in 1h 03m 00s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/7f35f76726dc421aaf64cb42cacf0fe6 : SUCCESS in 1h 00m 21s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/0dcec55082084777a5adf5b5d59b74de : SUCCESS in 1h 06m 21s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/ec01d7b97eff4888993d9ad9f307df7a : SUCCESS in 54m 33s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/06182b075d2b493386e0159bcb079a19 : SUCCESS in 1h 06m 39s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/cd41add318ff4f389c9d1c013228540a : FAILURE in 50m 26s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/42dd1ea77bf647ea839c1ab8d505b2df : SUCCESS in 1h 14m 10s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/0b2fc2d497f940c48baaf05b11f18337 : RETRY_LIMIT in 16m 09s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/2a7debbe6e7d44f58089582c6f73c628 : SUCCESS in 58m 46s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/c7efb72d777646519a1eae331859cac1 : SUCCESS in 1h 16m 41s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/fb7d7cf603e340688debc3778cd97192 : FAILURE in 48m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c33d6e8e42f59e317b09668df4ac2ceef984b38","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-28 02:13:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"16b00ee98b6920880d7c897c3826e2fda43732b1","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2026-07-24 04:24:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"da39def580a30c291acd00afc23192ad28b185c2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-21 17:59:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7bdece09d48eff08a7f089cb692f9470b45bf1d4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-30 15:19:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d1c8bed39956deff136ab572833d808c2c9bf29","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-10-03 01:09:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3249ec630cb31be355a2cbd9075335cead08119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 01:09:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1c5e0817c795c5e17d631e906fe6990b3cb4b52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 01:22:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37181b4f15824d66a82813b33e4fd535\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebed7d7e0eaf4fa998af2a3c2b5b53cf : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/742f47204be74fecaed0e55caa0d8e9a : SUCCESS in 8m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/379cd2f115e142d9a28564d4f6a6e217 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41dac60aa58a4fb1a3013ccae9a23129 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be388e8fac0f46f6a31e3c9d10af7c96 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":2},{"id":"3cc26be983a3f60bd07cd9bad42e3405debe6761","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 01:22:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e1238a6480c451119fc90dfc4d8a392580e13ed9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 01:24:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd59794f010b4d90b2f2f50286fc13ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c416e24836c4222ba75089c84a75285 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c07413057fbb75db04236018e481f555744acf29","revisions":{"05fb991086fd01f102ada667137cfd4298b61c21":{"kind":"REWORK","_number":1,"created":"2026-05-24 16:23:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/989906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/06/989906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/989906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/989906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/989906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/06/989906/1"}}},"commit":{"parents":[{"commit":"310e86944b5c5751876f986425a2e2f4486784a2","subject":"Merge \"Add periodic job to validate Tacker documentation links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/310e86944b5c5751876f986425a2e2f4486784a2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 16:22:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 16:22:47.000000000","tz":540},"subject":"Replace OVN Metadata agent with OVN agent","message":"Replace OVN Metadata agent with OVN agent\n\n... following the change in devstack[1].\n\n[1] 6180e73702cfef2011c32f315cde97128a4b7eec\n\nChange-Id: Id0a338f747501177e9f73b37048f64d166b899c5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/05fb991086fd01f102ada667137cfd4298b61c21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/05fb991086fd01f102ada667137cfd4298b61c21"}]},"branch":"refs/heads/master"},"c07413057fbb75db04236018e481f555744acf29":{"kind":"REWORK","_number":2,"created":"2026-05-24 16:25:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/989906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/06/989906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/989906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/989906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/06/989906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/06/989906/2"}}},"commit":{"parents":[{"commit":"310e86944b5c5751876f986425a2e2f4486784a2","subject":"Merge \"Add periodic job to validate Tacker documentation links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/310e86944b5c5751876f986425a2e2f4486784a2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 16:22:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 16:25:18.000000000","tz":540},"subject":"Replace OVN Metadata agent with OVN agent","message":"Replace OVN Metadata agent with OVN agent\n\n... following the change in devstack[1].\n\n[1] 6180e73702cfef2011c32f315cde97128a4b7eec\n\nChange-Id: Id0a338f747501177e9f73b37048f64d166b899c5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/c07413057fbb75db04236018e481f555744acf29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/c07413057fbb75db04236018e481f555744acf29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
