)]}'
{"id":"openstack%2Ftacker~992245","triplet_id":"openstack%2Ftacker~master~Icbcc51922e1fe37231e9498f2e398839667ff2dd","project":"openstack/tacker","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"36672":{"account":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},"last_update":"2026-08-25 18:29:06.000000000","reason":"\u003cGERRIT_ACCOUNT_36672\u003e replied on the change","reason_account":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"}},"37108":{"account":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"last_update":"2026-08-31 18:05:47.000000000","reason":"Change was submitted"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2026-08-12 12:34:37.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2026-08-31 17:45:56.000000000","reason":"\u003cGERRIT_ACCOUNT_17255\u003e replied on the change","reason_account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}}},"hashtags":[],"change_id":"Icbcc51922e1fe37231e9498f2e398839667ff2dd","subject":"Replace eventlet green subprocess with stdlib","status":"MERGED","created":"2026-06-08 13:37:08.000000000","updated":"2026-08-31 18:06:56.000000000","submitted":"2026-08-31 18:05:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"992245-eventlet-removal","meta_rev_id":"c016b402f39e760f0ac6b333300ef1927a48da69","_number":992245,"virtual_id_number":992245,"owner":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-31 18:05:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"recommended":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},"all":[{"value":1,"date":"2026-08-25 18:29:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},{"value":2,"date":"2026-08-26 13:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2026-08-31 17:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"all":[{"value":0,"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2026-08-31 17:45:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 16:54:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-29 11:42:06.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2026-08-12 12:34:37.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2026-08-12 12:34:37.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},"state":"REVIEWER"}],"messages":[{"id":"8c237fb5450ea70ca40bb87241ab40f2c85409de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-06-08 13:37:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4c9ace5c5b2be4ef6492fba5da91c86caf8a59e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-06-08 13:38:53.000000000","message":"Topic remove-eventlet-green-subprocess-agent removed","accounts_in_message":[],"_revision_number":1},{"id":"80745b877aae1c25d791f35b45d91d08e33b1c9c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-06-08 13:39:10.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"21534733300fae0a1992683846a42fa1fab4886d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 16:54:40.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c3e64552f5c41bb9b5492caee1e7c10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7115a40b0ba244d685ef4d3db11e7093 : FAILURE in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4340a624c341d1a5c0e2cc50959b73 : FAILURE in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9a13ed778ef4316bd1569c9468d1aa5 : FAILURE in 9m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e54524511cd4f69bccb69afc603a273 : FAILURE in 9m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9309658a52124bb0966a97fbbc3d1731 : FAILURE in 12m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aeeb38e6d2447c1a65bcc49c2b536eb : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7736de9359524fb8b070b38366e1b5c8 : SUCCESS in 6m 57s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/e94a13c901e042ba976feb7bf5bed84b : RETRY_LIMIT in 56m 22s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/8cd9138932124f84874e3fbfaf5bda6f : SUCCESS in 1h 02m 20s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/b91d02524e474033ba22c73bd8ea4362 : SUCCESS in 33m 39s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/92cb7954c5c84a98bb49f700e206481e : SUCCESS in 39m 33s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/913cb85cb3e64eed8b8785305847d32d : SUCCESS in 1h 05m 27s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/17f4050c21854bf784ccad5dae431c90 : SUCCESS in 40m 47s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/e38a428fa6fa41ad814b3ec3943467e9 : SUCCESS in 1h 09m 52s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/ca96e34bd59a4a84b6be9b5bb50f77fa : SUCCESS in 51m 40s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/edb104c6075f40fd81c26eb9f9d1e1a8 : SUCCESS in 58m 47s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/e7e789d6af8b44afb99a6c251faf86b8 : FAILURE in 52m 49s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/9634292b45de472181a258da90185a2c : SUCCESS in 45m 56s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/6947ac2a8353459da9731bfe01c01bb2 : RETRY_LIMIT in 16m 44s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/feaad238daa343d5b6f4be15496e3dbe : SUCCESS in 48m 10s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/7652617569c243a5abae9b31574bf43d : RETRY_LIMIT in 31m 21s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/149d3296539a452eac8c1133d8dabfc2 : FAILURE in 59m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8a0af86027b1c4a314a07a97def10c8145bad4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-06-10 09:19:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8c8e991caa879a8801ac2b34db1f49b342491f5c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-06-10 09:20:11.000000000","message":"Topic remove-eventlet-green-subprocess-agent removed","accounts_in_message":[],"_revision_number":2},{"id":"61a264981b23c819f83a557a8311f14c7f9ee3a8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-06-10 09:20:14.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":2},{"id":"ab163dac86953b81c4d403b88c625e9d8e5ddd7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 12:07:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc4c179813c8486dae2afb02f8723c1c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bacdba90930b416c9a572f8b35f359bf : FAILURE in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42af3201b10a4595a4cc425d248ecc81 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/accf1eee76f949419a4f628ad64a22ba : FAILURE in 7m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cad01465d0f8451ea63eea2630f0eb2a : FAILURE in 8m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1cee1037e9c4f488409975d2032f951 : FAILURE in 11m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59dc8545f37241488d73432824fe999b : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7834feae540c443a9a5cd40a579ab5df : SUCCESS in 5m 40s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/9dca766cce31424e902d9698193a3826 : RETRY_LIMIT in 54m 45s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/0550f19e5b81434ebc84df5d68f7c983 : SUCCESS in 57m 52s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/fdfb6e3f8d1745eababa983a5064d73a : SUCCESS in 1h 03m 30s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/27559c309e754c1a91a79f1b757b824e : SUCCESS in 1h 10m 32s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/e2a0709ab05c4179bc14102c67374974 : SUCCESS in 1h 03m 57s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/b6661f81aaa74ad6b6cfe0269bb7a247 : SUCCESS in 1h 04m 13s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/2df2e99b0ca64b098224bdbb0912ba3c : SUCCESS in 1h 13m 43s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/90844342c85842f3a56e0b9d14490058 : SUCCESS in 28m 22s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/d0fecadefd764947a0de857bfb34437e : SUCCESS in 57m 11s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/8ebd2beee66f4f1b9c6034c4061b16e8 : FAILURE in 52m 35s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/b41530f2719f43008ee749dab75040e5 : SUCCESS in 1h 06m 33s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/272a3afe746442c2bbf24153a584bc63 : RETRY_LIMIT in 7m 34s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/9daa8a0477be46e1afbfe60e98938f50 : SUCCESS in 57m 03s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/43be936501384d858580870d46621754 : RETRY_LIMIT in 33m 02s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/ef91151fa99843fe9568956caddb491c : FAILURE in 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9214cadd33af0e406fc5e627b928e39b62b0c0f1","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-06-11 09:09:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7203d86631815b40e168f065d7a233af517fd9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 11:45:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41d67677017e484b925ce19d17093f69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58a8f512ab014bdbb6ffa5585034f575 : FAILURE in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2244235a6a2c49f4a47ec495e089c4a9 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65290ed90a524f55b77ffbf1f7f9c359 : FAILURE in 8m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ad84018910548d2be18a4e4cec81beb : FAILURE in 8m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eee6a6744c29469bb716bcb86f4e95b0 : FAILURE in 11m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a0fac94a75446084a6e77de627d8b6 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e934462fe834de694b3568ddf78a8e7 : SUCCESS in 6m 11s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/bb115b6441c74959b3a5512a61f897e7 : RETRY_LIMIT in 57m 22s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/39a5a104c1614f7ab0bd35660b102f10 : SUCCESS in 1h 00m 18s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/20e135436b364fddac2934dce0f3574b : SUCCESS in 52m 05s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/0f1a23b1ba394a189ef59f567e84bb07 : SUCCESS in 1h 20m 38s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/93eb5bc319ad422480f2018f0d095998 : SUCCESS in 1h 09m 36s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/dac4a1292572411d98985f7637bb3314 : SUCCESS in 1h 02m 36s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/d6e5804d1108421581015cb9b67eb600 : SUCCESS in 55m 41s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/8918e84150e940089ff24109fad9f80e : SUCCESS in 57m 36s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/d519250f71c342579c0a5337c3163189 : SUCCESS in 1h 10m 54s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/40e5c63ee5fb403699b3f53ad97752e4 : FAILURE in 58m 52s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/bc897d2ef50046d3bf13d538d8d0b65d : SUCCESS in 48m 26s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/b437bfa2e10e46fcbf22141f6aef2642 : RETRY_LIMIT in 18m 22s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/217317fa947f4b88915bee2ef27976f3 : SUCCESS in 54m 56s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/25ed3ac839d44de3b373d2b5f39e9fe6 : RETRY_LIMIT in 46m 03s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/d091eb23c5464bc4a403b8ebb432c5d6 : FAILURE in 27m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16d60585de700acba32c8224a4fb0a1899aa4a73","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-06-25 07:27:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03c48d8e63024cd2fe8a1ec2d12211f9d0bd8878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 10:11:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca7607e3d99c42df868a2a686e26d580\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d160e4d8d4204adeb1e31bb3c5a081ba : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e7d73eb12824bb58241c7e74a3e8390 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cc56b644d2d42c4ad87db115b354d59 : SUCCESS in 6m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb570ed338924a3da013035fbcf199ae : SUCCESS in 7m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/505b0408d3f54d6e95fdbccaf8da80b2 : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc22e3f2236e49aba6e9ee3743973442 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeb6f983c0ac47cc96ef0ef2e6c3e931 : SUCCESS in 9m 30s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/76ca3e91105c4b4b9a2d246fc0dbb41b : SUCCESS in 1h 17m 35s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/c3e87c82c15d477b8834e1fefab299ce : SUCCESS in 56m 04s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/92f46dcace0c45009060473840504649 : SUCCESS in 1h 02m 27s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/4d2d2cc581c243c58e86e395f31da624 : SUCCESS in 1h 12m 37s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/2a50db470f8d466989eb3deb2e71f623 : FAILURE in 1h 04m 58s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/897b107b68814ae2989d0be7d23ec16e : SUCCESS in 32m 08s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/b3c867c605dd498aa488d914eec5bab0 : SUCCESS in 59m 52s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/d8c59dc254c9494ea4977748fd2a73b1 : SUCCESS in 32m 14s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/ec55b0b372f34b2ca7b6bc083a395cf8 : SUCCESS in 37m 45s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/6f07c0656fc4481388c3d06747139043 : FAILURE in 49m 35s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/7a41f001ee0b4641bdda675f0e852569 : SUCCESS in 1h 17m 33s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/ad01b58ed03e4f0dae11615fe64b0f9b : RETRY_LIMIT in 7m 04s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/d5b10d65350c4ec2b87bbc838680820c : SUCCESS in 36m 12s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/bdf713da5a7c4a15a44858be130ada80 : SUCCESS in 1h 08m 04s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/708df5ae57a94f0e85a25a36ef476330 : FAILURE in 2h 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d2abbfae8095df85f0aaa3e9069abdb1e84dec9","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-07-15 08:02:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ed2e099c3e9c9fa10817e2108f6c66ce0cd9bc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-07-20 11:27:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f2a0b57c335fd1907e1425b1993c94d453af8178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 13:16:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97ac2099423848e9964180015148a6f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f6818afd44e45f89aa2d238edd242df : FAILURE in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eced0c68da54c43b83d8a434c986ab7 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e897006052c24a809f69d31b9ffe3499 : FAILURE in 6m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c03c7fc8aa54699a884942fce92883c : FAILURE in 6m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50c7d15e56444b7ca69074f1d331cafe : FAILURE in 11m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5061dcf71d4bec9773e6f85eb8370d : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ead7687c574748c9bce0cbaf071ec40b : SUCCESS in 3m 18s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/774e9b7157894de7991204c78673699d : SUCCESS in 1h 15m 08s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/bea2c9c78b1c40b49296f0be6a686c4b : SUCCESS in 30m 36s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/5433fccda07b4ebbb839d0bfb87064e3 : SUCCESS in 59m 56s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/f8fa7af02b9344098c344ef5bd859f53 : SUCCESS in 1h 11m 07s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/25f150642bc84273bcda23bdd25f332f : SUCCESS in 32m 25s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/18bcdd42d0084d499fe7efd99d7eab29 : SUCCESS in 50m 56s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/bd29202264d04e67bf3dbbce55478f34 : SUCCESS in 1h 02m 20s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/bd4e8a22c0624b39acb5d6bfd62b3e6b : SUCCESS in 1h 01m 09s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/ef40f30174604ff68c70f500be0d5efa : SUCCESS in 54m 50s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/6b2c98d09ae04205b29b69da589f3c17 : FAILURE in 49m 51s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/299954df32004aa9abac555cab2fa288 : SUCCESS in 54m 03s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/c1268533653b4a4a811a5c1e807dd02e : RETRY_LIMIT in 17m 40s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/e8e0a035b9664709852603b99b4aad4d : SUCCESS in 1h 00m 43s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/a1c5d05ed64e405c819af27318c69bc6 : FAILURE in 1h 34m 23s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/a14c03af202c4454a437b2e6d5d25132 : FAILURE in 1h 47m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3158ce0deba375bc8be4947dcfe152e8c0b27750","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-07-22 07:37:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e0c5332f761b90d532aaa84abc677d357d1b859d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 09:25:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/582a34ef057f4faba3731e6cfc902730\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47b82b89bf1640009303c5dee632906b : FAILURE in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660ddef5664946358e2aaab989e5e7a2 : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be4ea02c3f75452ea745a975e423d832 : FAILURE in 8m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b43e9260817f4ddba51a3b3f05dbbd47 : FAILURE in 8m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e8d073e3f2947aaa33909428293ce98 : FAILURE in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f06d55d3e34ddeaf4762acb7799b65 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c4327e2a5f74fbd9fa3ca2a999622a2 : SUCCESS in 5m 24s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/89980c70906f4c97a3121e6dd88ebe98 : SUCCESS in 1h 09m 29s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/2e8235f79dfd4054905ed6fb4ca0e430 : SUCCESS in 29m 23s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/705f2aff4eb3410cbe15ee10cdeb69aa : SUCCESS in 28m 05s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/2f74c69637874afda294258a41f69f00 : SUCCESS in 1h 08m 03s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/dfe310c21b3b477bb731c59d5fb2e8f9 : SUCCESS in 59m 58s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/62e44f714cc346c0a3be563eaeb9be9c : SUCCESS in 57m 26s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/01a32968f9bb474983538e5a497b7998 : SUCCESS in 1h 00m 48s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/bb6963b3114849adbf70a23220ab588b : SUCCESS in 56m 39s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/80bd95bb6d2c4294965d20150ef4f69a : SUCCESS in 1h 06m 14s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/c4ad66750b7e418eaeb5ff36e38d7e5a : FAILURE in 52m 30s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/91b96f3c384f484897d236677cabcfbe : SUCCESS in 39m 20s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/797da216b50e429d80708d7f50664119 : RETRY_LIMIT in 18m 20s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/3bfb042ae1a14191b067c3b6ec898d98 : SUCCESS in 24m 42s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/c6919e864ef945ad9f76e0bab5a2ce61 : SUCCESS in 54m 33s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/5032331e20484890a6714565f05cf3e8 : FAILURE in 1h 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b33e038a59b1df1f73d8247a901e11da63b761e9","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-07-22 09:59:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b579e5f72115d6e3142e9ed18ac6bdeb66eaa93d","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-08-03 08:21:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82de6655b96de9bbb4726e56bfc4f77a5102b1cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 10:38:27.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea6ed98baf8a4b9383cc61b7e2e1ad2f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77224e9a85d84a7980e65c390e87e55f : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edae205380c44dc2a0e404f4d59da993 : FAILURE in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36b928dc383a4b9292919db4c5015725 : SUCCESS in 9m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e07ed842c9554c8596e0753e5de9dfb4 : SUCCESS in 8m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a420a713c85499185235e3659df6cc6 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1abd6bfbb44392838df80f111791bd : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b294bdc91a54d34b915708e07d5660f : SUCCESS in 5m 33s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/55f185a73432420c8cba7314b3eb74fb : SUCCESS in 1h 11m 17s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/febc3a04624443ae9f1d19462e65c782 : SUCCESS in 54m 01s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/8249c68273ac47f9ae36d04611977df0 : SUCCESS in 56m 08s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/8bb45f5d128045448fdfbc760ca729b9 : SUCCESS in 1h 02m 29s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/d15600164b104a09afda0e368e3d4e9a : SUCCESS in 43m 34s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/ff0c9d73baa54f5692f4292758c36d73 : SUCCESS in 1h 05m 04s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/994c76485d764bae9a1c7e74e9fbc975 : SUCCESS in 31m 52s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/30ad1de1eec44723a4c7fff9ad8ef9ab : SUCCESS in 49m 59s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/9ba6e98397294e2e9310ddf48d86c51e : SUCCESS in 46m 39s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/3eaa9105247e48efb56d1d8b41177f43 : FAILURE in 50m 28s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/19f82aa92cf84c9198b9cb92d66f7af1 : SUCCESS in 36m 42s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/995437ad45b941ad8668b3c9114aba07 : RETRY_LIMIT in 14m 35s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/fcc3b6e5d0234ecfaa71e030a78be993 : SUCCESS in 23m 36s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/6601a28ff24a48d5887e5b0df850c350 : SUCCESS in 1h 05m 22s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/ec50706d190048f8b45439a245e39fa3 : FAILURE in 2h 09m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6be6ae2be2c15e43e9448c021ba35bdcee4318d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-08-04 05:40:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"fc2da39f5c3e0986342e9ae19637d517bea64647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 08:03:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d2ea016bb8943f58a75cd8d880c16e5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/276b5c1e9ed54c6babdaca0bbe83f1ee : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1e32b83146423ca44577181005b8b8 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f886d21a42494a549190363f88752b79 : FAILURE in 9m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de1898ae6a984d4eb67db719fdd5b479 : FAILURE in 5m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/240ce362acf0493d9d6bf319d575ca38 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd97837dc30a451193b87ffc93dd2f8d : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2e5ab53caf94877bfb3f400de488631 : SUCCESS in 3m 23s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/a3e9f98ac72c455a9ab5f8edfc191d34 : SUCCESS in 1h 11m 04s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/3e1849bc52f24d8398566ebbd30203f6 : SUCCESS in 53m 01s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/7a60857d2e584ead8996372ce450c6a2 : SUCCESS in 29m 08s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/c6904695fd91424cab33dc93e7f35870 : SUCCESS in 46m 53s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/aeee6219f5a44aa6a70dff3d430c0088 : SUCCESS in 57m 58s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/5379f3490ce44bd2b4e8504e2b3cc8d3 : SUCCESS in 54m 27s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/97b916c2f6bb438aa5feb0b7575b4e19 : SUCCESS in 1h 03m 05s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/58a22acb614c40ba90c530192538e981 : SUCCESS in 55m 43s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/854a380e1b0a42d5abb214392d0143a8 : SUCCESS in 1h 06m 13s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/d4b60a9bb5c8424485305442862c3daf : FAILURE in 31m 58s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/9a4250419dda41edafd53d097ec96da9 : SUCCESS in 1h 09m 48s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/39b1039e3f314a6482bbeb5dec1a5acd : RETRY_LIMIT in 14m 36s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/e9d6849c93b241af9eab4a2afe0f56a5 : SUCCESS in 37m 23s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/7691e36c3e8d4b2ab2d06edb90960fad : SUCCESS in 1h 12m 25s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/1d363e4e73c84cff8b16edff4f8f29b8 : FAILURE in 2h 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5eb1dabe6e6e17e3376a569754a150d232b0d7ce","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-08-04 20:04:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e1210cdf125f33aa09dc3f37c33e296cfcb58024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 22:30:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29bcf8900ebd4721924294f8fbd53b8a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b94a9445f9c9445686fe52bc542b9e99 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf5f5035dc341b3a3db0da5e52c9dd9 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37e003e4378247e39f0846d7cfdd757a : SUCCESS in 7m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33d39d0b80bf4e869b9ddee706796858 : SUCCESS in 5m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c63994930fa4d8197479c6d1504cba3 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb08a060dc6a490686529178e7672a5b : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d75d5b69cb14ae3aa1fe41e98d56c8c : SUCCESS in 5m 26s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/6e9a0e66c11d4f0a9ec05cf07ac6ba6a : SUCCESS in 45m 01s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/a69a961611384885902c70ca436c331d : SUCCESS in 25m 02s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/1507f9ebea91454d8e54ce697f94d653 : SUCCESS in 59m 47s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/ddc977e7035d4209805208fdd933b190 : SUCCESS in 34m 46s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/5187986d57fe43ad99e3465f6a101c48 : SUCCESS in 57m 50s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/55f1a44e61a7490094553e0a9ab9a778 : SUCCESS in 57m 11s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/88667acb15f3479eb2d2650995f55cec : SUCCESS in 46m 02s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/752cd864db034d91a4c94edcd37ef343 : SUCCESS in 54m 37s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/b2e994e9a9f94082b602b4b2116112db : SUCCESS in 1h 01m 02s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/f72330d25e3c4672ae7201d82f1efbc6 : FAILURE in 23m 42s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/276e869c58c94a5e91bd498ef584b95a : SUCCESS in 1h 12m 11s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/718231ee3fda422794e56eea6f39c917 : RETRY_LIMIT in 16m 44s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/cf434840936140568cac3970acbfcdf9 : SUCCESS in 46m 16s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/ac2bde32ffa649b28b95f75421ec4bc6 : SUCCESS in 1h 24m 13s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/795431e6f2c94db084f99c0f54d843e0 : FAILURE in 2h 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b4f4f0ac09ba1209f7b884cf7e971c3a720528a7","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-08-05 12:34:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c58a30f5964dd2e06c8e8212460e2143d3a02767","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-08-05 12:56:07.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6a2449fcb8468ac697357e8cc9f5f6a42e012d77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-08-10 08:01:51.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"31ab7194a43e4f6a1f980bea0318e0cb060843af","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-08-10 08:07:55.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a0c430bc789118b08054f0380bbf17e722f6fc3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 10:15:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0f85df8510441e5853fd28885605557\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1383fba4a80645aa9371cff4b2b1d884 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83dbbe391ae8487f99bbfd84a003693c : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd797c5d95db4dddb1bfefe3f438507f : POST_FAILURE in 9m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3891ca3d8d35440bb8c7c9b2dfd6dc43 : SUCCESS in 7m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f267405857a94f1e949df2f5c708ccce : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a9e16d99794f2e8cc8f828f76bfe39 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1701f0ffa8d40f9af788ac8abeac941 : SUCCESS in 5m 18s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/3672cb6d402946f5b96171c046e3a1ef : SUCCESS in 1h 19m 15s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/4e5cc8ff53c541388b1b5612b618c69a : SUCCESS in 54m 24s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/00947e61de364869be77e46f0cf87f96 : SUCCESS in 46m 32s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/41613ade405c4e30bccbeb0e503f9235 : SUCCESS in 1h 02m 25s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/261d6d2f45534664a63577a93c2d2a2e : SUCCESS in 42m 30s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/60b648671fff4f43a726e5422b81cf37 : SUCCESS in 59m 45s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/e38b435dfcbd481292753c27d8181442 : SUCCESS in 1h 08m 58s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/37e41aa481c4470abc39a83872b3782e : SUCCESS in 56m 15s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/127a800102ae456891aabad37a04a83c : SUCCESS in 1h 02m 22s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/942d05d4e4924f0a960235630d05e477 : FAILURE in 51m 16s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/a6cdfc5f514446d9b9f8e8041a75c0f3 : SUCCESS in 1h 06m 01s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/0b4972511a7545aea2dcb169847183a2 : RETRY_LIMIT in 6m 58s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/f91793bd96ec412fb81bb36fbf812bb9 : SUCCESS in 51m 47s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/b0d3f974ffeb4c9d935ab52e1c02bec5 : SUCCESS in 46m 47s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/b5fd32f1146d4f8aa84354bd67c9d470 : FAILURE in 2h 07m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a1b6af44c1c80bed5b151805125c51d4521ec8a0","author":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"date":"2026-08-12 08:30:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"256e233442bd0eca488e4271d3e3f3f0c381b001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 10:58:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4440618e8d344d4186caa074df027018\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be43e55631c54689b82699765bb517ec : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce58157fbb0e4b6d87291b3833c35885 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d887c88deed44c9913377bcba939f30 : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1930b7dc084544ceb8ecd90941f6d646 : SUCCESS in 8m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/996eb10d53914dd28a82977db4a967bd : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/044aad421ad047a08ed68834463f1926 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aff08eb376643f8b425736903daded8 : SUCCESS in 2m 55s\n- tacker-ft-v2-k8s https://zuul.opendev.org/t/openstack/build/efc532e5db12496c8858f4ae8f317c90 : SUCCESS in 1h 01m 20s\n- tacker-ft-v2-df-userdata-basic-min https://zuul.opendev.org/t/openstack/build/5b7c9d0a7f6248ee8aaf34b34b5ea147 : SUCCESS in 53m 43s\n- tacker-ft-v2-df-userdata-update https://zuul.opendev.org/t/openstack/build/8128d10884b64fd9a6d62f55fcab9e3b : SUCCESS in 54m 26s\n- tacker-ft-v2-df-userdata-scale https://zuul.opendev.org/t/openstack/build/992c983a11e442a3a4fb1b9ba711a986 : SUCCESS in 1h 00m 51s\n- tacker-ft-v2-df-userdata-ccvp https://zuul.opendev.org/t/openstack/build/5a22e081f545448ba27b41deecaa605e : SUCCESS in 55m 53s\n- tacker-ft-v2-df-userdata-err-handling https://zuul.opendev.org/t/openstack/build/2367f0060b2c40de8b1997f82d34d5a9 : SUCCESS in 59m 28s\n- tacker-ft-v2-st-userdata-basic https://zuul.opendev.org/t/openstack/build/3c9fd6fd9d2c4a50b2ad67640f26bea0 : SUCCESS in 59m 56s\n- tacker-ft-v2-st-userdata-basic-min https://zuul.opendev.org/t/openstack/build/6bbe507ad31249449fd50398bab7b673 : SUCCESS in 52m 23s\n- tacker-ft-v2-st-userdata-ccvp https://zuul.opendev.org/t/openstack/build/2f5cbba571b54b2ba85afb4734227c7d : SUCCESS in 58m 05s\n- tacker-ft-v2-df-userdata-https https://zuul.opendev.org/t/openstack/build/c0217e944c1242a3a9b2dd0406d8829e : FAILURE in 37m 42s (non-voting)\n- tacker-ft-v2-df-userdata-notification https://zuul.opendev.org/t/openstack/build/aa710d54b637473991b1acb3868c818a : SUCCESS in 1h 02m 26s (non-voting)\n- tacker-ft-v2-terraform https://zuul.opendev.org/t/openstack/build/dc906fbca23e4cd7bf5e2cc016e48f18 : RETRY_LIMIT in 17m 11s (non-voting)\n- tacker-ft-v1-vnfpkgm https://zuul.opendev.org/t/openstack/build/5d5e5446b6864ee0be0def914d94f594 : SUCCESS in 53m 55s\n- tacker-ft-v1-k8s https://zuul.opendev.org/t/openstack/build/d125db6f87b04f4c903f28a06bca2b24 : SUCCESS in 57m 12s\n- tacker-ft-v1-compliance-sol https://zuul.opendev.org/t/openstack/build/e2815df70d7a409b854116b5f9cc1269 : FAILURE in 2h 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9a891b5f6c32cb00054318d3cb6d73b1c429d6fd","author":{"_account_id":36672,"name":"Shivam Shukla","display_name":"Shivam Shukla","email":"shivam.shukla3@india.nec.com","username":"shivam.shukla3"},"date":"2026-08-25 18:29:06.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ae96736d444c413947094854519d519c8bee022c","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2026-08-26 13:05:39.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1b004c6fa00a3b1699d951681da27ca327f704c7","author":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"date":"2026-08-31 17:45:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bac4453de453beb1d784590c3a18689d1c49cdd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 17:46:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"57c81b0a1b6bd5e0318c239e16a7e4d2b409e8f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:05:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc7337fd9c8e430cbddc5d7c33007422\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ebf2739953a4321af10c8f414a81f84 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b824c8d27c2643e1b0fbcbf023066d10 : SUCCESS in 9m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01b0c903e7404fdcba2a888aefc64f07 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e63aef3bba4e9ca5188bd19068dfac : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b78ec9f179e04cacb30efe5992b49980 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":5},{"id":"3ffcb7361a9f6c902227a2af73a00946c2b6ce88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:05:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c016b402f39e760f0ac6b333300ef1927a48da69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:06:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7419a5adba429e8ed505ea80c41703\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/690a3f169bfa45acad03a2f2a94184b5 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b074bf9779bb343003c820b6800f5e1edb28be92","revisions":{"742c8a5a8a96a05197c14160f7fc1b65b2a147e0":{"kind":"REWORK","_number":1,"created":"2026-06-08 13:37:08.000000000","uploader":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"ref":"refs/changes/45/992245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/45/992245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/45/992245/1"}}},"commit":{"parents":[{"commit":"fea54ab7db0bc62865fbdc03cf8317337b73b3f4","subject":"Merge \"sqlalchemy: Use built-in declarative\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/fea54ab7db0bc62865fbdc03cf8317337b73b3f4"}]}],"author":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-06-08 13:31:04.000000000","tz":540},"committer":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-06-08 13:31:42.000000000","tz":540},"subject":"Replace eventlet green subprocess with stdlib","message":"Replace eventlet green subprocess with stdlib\n\nThis patch replaces eventlet.green.subprocess with the standard\nsubprocess module in tacker.agent.linux.utils and tacker.common.utils.\n\nIt also updates the corresponding unit test mock target.\n\nThis is part of the eventlet removal work.\n\nChange-Id: Icbcc51922e1fe37231e9498f2e398839667ff2dd\nSigned-off-by: Hitomi Koba \u003chi-koba@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/742c8a5a8a96a05197c14160f7fc1b65b2a147e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/742c8a5a8a96a05197c14160f7fc1b65b2a147e0"}]},"branch":"refs/heads/master"},"ce6b8b0ba1f1566cf3df701e288e67189ed7d134":{"kind":"REWORK","_number":2,"created":"2026-06-10 09:19:43.000000000","uploader":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"ref":"refs/changes/45/992245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/45/992245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/45/992245/2"}}},"commit":{"parents":[{"commit":"fea54ab7db0bc62865fbdc03cf8317337b73b3f4","subject":"Merge \"sqlalchemy: Use built-in declarative\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/fea54ab7db0bc62865fbdc03cf8317337b73b3f4"}]}],"author":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-06-08 13:31:04.000000000","tz":540},"committer":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-06-10 09:18:53.000000000","tz":540},"subject":"Replace eventlet green subprocess with stdlib","message":"Replace eventlet green subprocess with stdlib\n\nThis patch replaces eventlet.green.subprocess with the standard\nsubprocess module in tacker.agent.linux.utils and tacker.common.utils.\n\nIt also updates the corresponding unit test mock target.\n\nThis is part of the eventlet removal work.\n\nChange-Id: Icbcc51922e1fe37231e9498f2e398839667ff2dd\nSigned-off-by: Hitomi Koba \u003chi-koba@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ce6b8b0ba1f1566cf3df701e288e67189ed7d134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ce6b8b0ba1f1566cf3df701e288e67189ed7d134"}]},"branch":"refs/heads/master"},"05be0f5394c027c5b14faa62ccdaacdaf568873d":{"kind":"REWORK","_number":3,"created":"2026-07-20 11:27:08.000000000","uploader":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"ref":"refs/changes/45/992245/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/45/992245/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/45/992245/3"}}},"commit":{"parents":[{"commit":"f3615fce08fa37af57636b5346cd42df607f3d38","subject":"Merge \"Refactor access to hash object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/f3615fce08fa37af57636b5346cd42df607f3d38"}]}],"author":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-06-08 13:31:04.000000000","tz":540},"committer":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-07-20 11:26:16.000000000","tz":540},"subject":"Replace eventlet green subprocess with stdlib","message":"Replace eventlet green subprocess with stdlib\n\nThis patch replaces eventlet.green.subprocess with the standard\nsubprocess module in tacker.agent.linux.utils and tacker.common.utils.\n\nIt also updates the corresponding unit test mock target.\n\nThis is part of the eventlet removal work.\n\nChange-Id: Icbcc51922e1fe37231e9498f2e398839667ff2dd\nSigned-off-by: Hitomi Koba \u003chi-koba@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/05be0f5394c027c5b14faa62ccdaacdaf568873d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/05be0f5394c027c5b14faa62ccdaacdaf568873d"}]},"branch":"refs/heads/master"},"ede1a55dd557f923265bcc8d10776889317ed5d9":{"kind":"REWORK","_number":4,"created":"2026-08-04 05:40:09.000000000","uploader":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"ref":"refs/changes/45/992245/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/45/992245/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/45/992245/4"}}},"commit":{"parents":[{"commit":"7b7fe811630ab2d55d2d5bfc3c340cb864c7e261","subject":"Merge \"Replace eventlet with stdlib equivalents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/7b7fe811630ab2d55d2d5bfc3c340cb864c7e261"}]}],"author":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-06-08 13:31:04.000000000","tz":540},"committer":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-08-04 05:39:57.000000000","tz":540},"subject":"Replace eventlet green subprocess with stdlib","message":"Replace eventlet green subprocess with stdlib\n\nThis patch replaces eventlet.green.subprocess with the standard\nsubprocess module in tacker.agent.linux.utils and tacker.common.utils.\n\nIt also updates the corresponding unit test mock target.\n\nThis is part of the eventlet removal work.\n\nChange-Id: Icbcc51922e1fe37231e9498f2e398839667ff2dd\nSigned-off-by: Hitomi Koba \u003chi-koba@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ede1a55dd557f923265bcc8d10776889317ed5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/ede1a55dd557f923265bcc8d10776889317ed5d9"}]},"branch":"refs/heads/master"},"b074bf9779bb343003c820b6800f5e1edb28be92":{"kind":"REWORK","_number":5,"created":"2026-08-10 08:01:51.000000000","uploader":{"_account_id":37108,"name":"Hitomi Koba","email":"hi-koba@kddi.com","username":"hitomikoba"},"ref":"refs/changes/45/992245/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker","ref":"refs/changes/45/992245/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker refs/changes/45/992245/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker refs/changes/45/992245/5"}}},"commit":{"parents":[{"commit":"7b7fe811630ab2d55d2d5bfc3c340cb864c7e261","subject":"Merge \"Replace eventlet with stdlib equivalents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/7b7fe811630ab2d55d2d5bfc3c340cb864c7e261"}]}],"author":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-06-08 13:31:04.000000000","tz":540},"committer":{"name":"Hitomi Koba","email":"hi-koba@kddi.com","date":"2026-08-10 08:01:31.000000000","tz":540},"subject":"Replace eventlet green subprocess with stdlib","message":"Replace eventlet green subprocess with stdlib\n\nThis patch replaces eventlet.green.subprocess with the standard\nsubprocess module in tacker.agent.linux.utils and tacker.common.utils.\n\nIt also updates the corresponding unit test mock target.\n\nThis is part of the eventlet removal work.\n\nChange-Id: Icbcc51922e1fe37231e9498f2e398839667ff2dd\nSigned-off-by: Hitomi Koba \u003chi-koba@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/b074bf9779bb343003c820b6800f5e1edb28be92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker/commit/b074bf9779bb343003c820b6800f5e1edb28be92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
