)]}'
{"id":"openstack%2Ftaskflow~1003732","triplet_id":"openstack%2Ftaskflow~stable%2F2026.2~If5ecb46230c15f5aadbbe1b4d3595a8327aeaaba","project":"openstack/taskflow","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"If5ecb46230c15f5aadbbe1b4d3595a8327aeaaba","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","status":"MERGED","created":"2026-09-03 10:57:48.000000000","updated":"2026-09-03 12:41:58.000000000","submitted":"2026-09-03 12:39:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003732-create-2026.2","meta_rev_id":"73f459e41132b19f47a29fbc174313e6b39e45c1","_number":1003732,"virtual_id_number":1003732,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 12:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-03 12:24:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-03 12:24:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 11:56:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 12:24:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b65607622cb6ce485b84ed27d14ae3e154eaa444","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-03 10:57:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fe2738a8681e07bd040d2ed3906018e5f87fb6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 11:56:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c5f8c5112934b0d8239ceaa0ab5e6dd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ded8201b5771454e8a0a17b92463dcae : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6902542a1f4b419887190d36cf8453b4 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52dc5e1b9e904e8296074a0338c5eda4 : SUCCESS in 10m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c4e6050590b4d8d80962f00a69feb6b : SUCCESS in 10m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f5561fc71854b22804ef6254398811f : SUCCESS in 13m 37s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ae3414bcea774de59d726a16eeab191f : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca325ce833594be0a4fb7e515acd3043 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92c879ce02a1409e8c81fa5d29ad7079 : SUCCESS in 5m 55s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/cfd7c154f5b0461998ef9b0ef67a31a6 : SUCCESS in 7m 14s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/3ac73b22d6114be3941515e9c2fcceba : SUCCESS in 7m 06s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/39590bb51265484cab204c0e28895ff2 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a51928b48472fbc069d214a4f879e83397c92ad5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 12:24:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43c1c110bead3842cc621790922028cc11356c47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:24:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47032daac71d3c449359951622414c70a3b03e72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:39:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2179f2be79fd429dba1f8f3756274286\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbad088f3d5d4e94a6012acbabbf5240 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5020f0a24614e36be38b1cd61c550f0 : SUCCESS in 9m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7126a8979f6e47ee873705189abc1e0f : SUCCESS in 8m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/de4f6bc012b84edeb8f371a121011a73 : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8fa709006194713ab9daa97d9621082 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/373085e4a96c4d669ed2632100956b82 : SUCCESS in 5m 18s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/885a44fc6d174f8d9270736e65940172 : SUCCESS in 5m 52s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/efebddadd07042f3aaac82fe66fecef3 : SUCCESS in 7m 04s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/92442189576a476080b190a396d4243e : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"622f5f8611a6e85e3ff18ab4b9a07355d96de59f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:39:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73f459e41132b19f47a29fbc174313e6b39e45c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:41:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f11558beb74ceb86fa7bc39e295655\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc6b64a8c8149cfb114bc0b62071f1f : SUCCESS in 1m 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f44f35f6af4d46378cd7e2f7dd90b53b : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05af19d7054e21fd02215ca7921ca038d875629a","revisions":{"05af19d7054e21fd02215ca7921ca038d875629a":{"kind":"REWORK","_number":1,"created":"2026-09-03 10:57:48.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/32/1003732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/32/1003732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/32/1003732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/32/1003732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/32/1003732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/32/1003732/1"}}},"commit":{"parents":[{"commit":"37c674cc89ab219b66d23fbb12e5a3f05480d07b","subject":"Update .gitreview for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/37c674cc89ab219b66d23fbb12e5a3f05480d07b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 10:57:47.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 10:57:47.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: If5ecb46230c15f5aadbbe1b4d3595a8327aeaaba\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/05af19d7054e21fd02215ca7921ca038d875629a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/05af19d7054e21fd02215ca7921ca038d875629a"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
