)]}'
{"id":"openstack%2Ftaskflow~1004449","triplet_id":"openstack%2Ftaskflow~master~I6547f0eb0645306280a987fd1f90088694c0c85e","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 21:57:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-07 23:05:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6547f0eb0645306280a987fd1f90088694c0c85e","subject":"Resolve setuptools deprecation warnings","status":"MERGED","created":"2026-09-07 16:03:07.000000000","updated":"2026-09-07 23:06:11.000000000","submitted":"2026-09-07 23:05:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004449","meta_rev_id":"8a5627323866e2f89974b8d067f21a0780f566cd","_number":1004449,"virtual_id_number":1004449,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 23:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-07 21:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 21:57:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 17:11:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 17:32:38.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b5d97daa0725f725ee5118d857ce9a8c08d08a80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 16:03:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9fbcdc55c479330b73e36a5f8fc2a1039bd87bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 17:11:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea70c26dc8414785823b44a2066b9522\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f444db5e50634b9ca49ab3476763e1e0 : SUCCESS in 2m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c1070d624ac5480f8a709ecb45947e89 : SUCCESS in 3m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/922004bc1edf4f81a89ca69c2c0638e4 : SUCCESS in 1h 07m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b47eae538cb4d53a033c9b2b20a8fd1 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe44c4488c14023bd26cc536014c434 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78f8fcf2c98b40f5a41fe4b42c84af88 : SUCCESS in 10m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b256c9a66a724123a037a5f46c1d3006 : SUCCESS in 9m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b323dbb83ccc4861ac9d0d6d412d81a1 : SUCCESS in 12m 37s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/1554bf0801d1418ca62efd9845843ad9 : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f68c30f4df4250aac16d2c2d386aa6 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebf7cfb3f17245c89ade37939f90faa3 : SUCCESS in 4m 39s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/a9e28ee0ede64a5faa05c55a8c42016c : SUCCESS in 5m 20s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/92f4db5ac2424c33a72f355e4084dd0e : SUCCESS in 6m 24s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/503567f1752a4adcbf1356127f8574da : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9e57a9bc4021cdfc3b0b551e54d949563172c679","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 17:32:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf07cb7940895c8be89f19cbbef7ba752f33bfeb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 17:35:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"846a0ecb793e26d66c36c5763741f49c4c329ca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 17:35:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"296d821bed2b9cb452ce24a1000c101e2fcdf195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 19:00:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38dbb3f75cf842279f8bdeab665947a5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c962489a3bd4f1d92649a83b4efbdf9 : SUCCESS in 2m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/edb8786da0554f5181a9a1badeb7e88d : SUCCESS in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/438adfb1eb424b43a8aa5b730317ab04 : SUCCESS in 1h 22m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb7067ad1e034ff3ab37b42b2c023333 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/454ef3c60aa24fa593e0bf135ff51245 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5536772347d47ea988a23c6584111ee : SUCCESS in 8m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cf6c1715e4d4b6cad7c57e969001d79 : SUCCESS in 9m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d4e394347ed54d75be6139495d8a1227 : SUCCESS in 12m 27s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/34cb59089e2c4690a34069012bd0d1b6 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c82f2c3d8954b7ba4765be440ac493b : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e5607da7d434c0aa490efa368438d3b : SUCCESS in 4m 57s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/d3b4864f3fcf449a825f43aa31976068 : SUCCESS in 9m 34s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/5b09b046e14e406287067532362f7fcf : SUCCESS in 5m 07s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/9031f38d8dbd46b3bec9d1775cc25073 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":2},{"id":"5cca1facebf3b14e7929b65a63284867a9fe7a33","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 21:57:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0c0ec582cdbe89275f384d164e72132213fe9a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 21:58:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"625421eb75c84e9f5432ec806b863ec8cc7bd990","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:05:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90ce52fdd1f425aa3e498ed517bf87f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b6872e12dc8e4d73a5b0352e2d195516 : SUCCESS in 2m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0a9f0656ea36469a83d702190d35014d : SUCCESS in 2m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/016a125002cf469dab17d90087ca6323 : SUCCESS in 1h 06m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f4c4d2fe52645a4a595d3370bd8aa50 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37b7902cfaed4710a72f9a6ea007ca2e : SUCCESS in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f99c56d2cd494792afaf04d5663e739e : SUCCESS in 9m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/653cb1178f30478db824806a76097718 : SUCCESS in 12m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67528d380272402186fcfed559f6a15b : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f05a5dc255b4326b0d963c678555d9e : SUCCESS in 4m 35s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/3a9884423afc4749a7c359ba7ec1ae7b : SUCCESS in 8m 54s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/b57251812cbe4a2e81abf9e1d3626b30 : SUCCESS in 6m 45s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/878fe2572e104ef89209f844ac5d6ee0 : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":2},{"id":"8977d618461a9a7ca7311b5ca5b167a9763813cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:05:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8a5627323866e2f89974b8d067f21a0780f566cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:06:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/948d5ff756bd4f238a616e6ceb949648\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89f09042000f412f922e9bb6c48a3195 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a00e0d7f5f76e6ba1a1bc6f03b19b93a8fbcafba","revisions":{"ee52aac2da667b949bd7b019b7472fd9516624b3":{"kind":"REWORK","_number":1,"created":"2026-09-07 16:03:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/1004449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/49/1004449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/49/1004449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/49/1004449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/49/1004449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/49/1004449/1"}}},"commit":{"parents":[{"commit":"8c978e8d46ba3931ef36ef0c28ca105d5f0754f4","subject":"Merge \"Remove duplicated optional dependencies from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8c978e8d46ba3931ef36ef0c28ca105d5f0754f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 16:02:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 16:02:57.000000000","tz":540},"subject":"Resolve setuptools deprecation warnings","message":"Resolve setuptools deprecation warnings\n\nChange-Id: I6547f0eb0645306280a987fd1f90088694c0c85e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/ee52aac2da667b949bd7b019b7472fd9516624b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/ee52aac2da667b949bd7b019b7472fd9516624b3"}]},"branch":"refs/heads/master"},"a00e0d7f5f76e6ba1a1bc6f03b19b93a8fbcafba":{"kind":"REWORK","_number":2,"created":"2026-09-07 17:35:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/1004449/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/49/1004449/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/49/1004449/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/49/1004449/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/49/1004449/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/49/1004449/2"}}},"commit":{"parents":[{"commit":"8c978e8d46ba3931ef36ef0c28ca105d5f0754f4","subject":"Merge \"Remove duplicated optional dependencies from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8c978e8d46ba3931ef36ef0c28ca105d5f0754f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 16:02:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 17:35:47.000000000","tz":540},"subject":"Resolve setuptools deprecation warnings","message":"Resolve setuptools deprecation warnings\n\nChange-Id: I6547f0eb0645306280a987fd1f90088694c0c85e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/a00e0d7f5f76e6ba1a1bc6f03b19b93a8fbcafba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/a00e0d7f5f76e6ba1a1bc6f03b19b93a8fbcafba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
