)]}'
{"id":"openstack%2Ftaskflow~1004914","triplet_id":"openstack%2Ftaskflow~master~I671c6310a54c47815eb25ce7b1b335949d30c3fd","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 21:58:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I671c6310a54c47815eb25ce7b1b335949d30c3fd","subject":"Use built-in StrEnum","status":"MERGED","created":"2026-09-09 17:59:57.000000000","updated":"2026-09-14 22:00:06.000000000","submitted":"2026-09-14 21:58:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004914","meta_rev_id":"e8eabc888a5dc44b3d90b3577e98d0786ab29d45","_number":1004914,"virtual_id_number":1004914,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 21:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 20:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 20:04:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 19:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 20:04:01.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"5a02eba9753f788482f928b85f5f6b8306e36d83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 17:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22fc0b5dab4f826271b5443554994f8351d34546","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 18:00:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dcea7c3b091598166bbe4d854a3ff02bbb8eecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 19:48:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6493f920bb9042c09f6f7a207e64cf61\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c7cecaa22f64d00b0a99292a7ae78cc : SUCCESS in 1h 41m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca18779a0ab649cb95b37d64dede2e44 : FAILURE in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff9284344164d66ac1689c1ab7557bd : FAILURE in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d93ac205e3646a3b01af6f7c0d1fd0a : FAILURE in 9m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96e94a01f8424a8b91a0178340281f1d : FAILURE in 9m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2ed295fcce154c3c9a2320be649c065d : FAILURE in 12m 40s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/439c85a351f8419facb153e00ab94128 : FAILURE in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58093ad754a547ad837c7a2204442126 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a996970c87d848008bc4c383b0035621 : SUCCESS in 5m 59s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/66a788777b3642b591d6ed8d7802ca55 : FAILURE in 7m 24s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/4e70c1be76114f11b0f16848bc47cc04 : FAILURE in 6m 40s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/a196fe47bb4e49b291e42c5862f57403 : FAILURE in 6m 47s","accounts_in_message":[],"_revision_number":2},{"id":"61d1977d425acb7bb4f7eef03b3705765be34a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 01:26:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c3c6c2eb6819da90a861ceeca525be3808a152e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 02:17:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c85b68d3ae34d5db584d3e87df617cd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6830f4c72dbb45f78410955db5c534bc : SUCCESS in 49m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cb5f46b931944a283c68845575fdd3c : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92b55a66169e4a87a37cbd9f4a229a94 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de60170d59d643338d573f48d9468af9 : SUCCESS in 9m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15ffa790ca9345ffbd80d7a37cd3ffec : SUCCESS in 8m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae0fc9daafa849368fe9c20f842581d8 : SUCCESS in 12m 16s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/e1a502ae4db44ef1b44b97fed7776a0f : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6a547226bb4073af65bcaf2b3e356b : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55537e41651f4ebd87d683f2d3cb101a : SUCCESS in 7m 36s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/fa62b37c1d1241689dcf8a3f61fb667f : SUCCESS in 5m 10s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/d8de33f3a4dd4e10b188b8ab6125f08d : SUCCESS in 6m 06s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/d7554ca49a6346cdbd417bedce934425 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":3},{"id":"60ab98b99fb79bf3bceecb8261d2d50304d6622a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 20:04:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ffaa45a74fc85f5300fbcedca71791d729c918ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 20:05:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4c64952e7f0ada80954b814629313dee908e7c7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 21:58:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77bed91a4f844314b63f6c2dca7482dc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5212b97c78834897a00322c0b294dc1b : SUCCESS in 1h 48m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378e6178fa1c45fcb313a684f9c4e479 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1b3569e2e37495187534158a96dca84 : SUCCESS in 9m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3c24a17cef149ad9c7370dde3d697f0 : SUCCESS in 10m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db90a7fcb67c458aa8d29f2ca7b5441d : SUCCESS in 13m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ab00937bc24951b47b2aecc2b17855 : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65637fd0e5e749729174c876382a8b1e : SUCCESS in 5m 57s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/2543b3ec6ffa4a16856189aebceb9a22 : SUCCESS in 7m 14s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/5d9c980922c74f5fbca202044933471e : SUCCESS in 6m 35s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/4ac1f76fecf34047b4ded3b48dc0d5c8 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"822fba69c573c6c441203ed928094d816cb55af8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 21:58:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e8eabc888a5dc44b3d90b3577e98d0786ab29d45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:00:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3882ee8aebd942cd9caad76810dfd65f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54c07a88c484354b648a5ce6db0a478 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fe890f8f71b8cd861900759ac915739c294f322c","revisions":{"3e8f08a70a253b5bb2ce6f2381cf268a719adc83":{"kind":"REWORK","_number":1,"created":"2026-09-09 17:59:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/1004914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/14/1004914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/1"}}},"commit":{"parents":[{"commit":"8c978e8d46ba3931ef36ef0c28ca105d5f0754f4","subject":"Merge \"Remove duplicated optional dependencies from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8c978e8d46ba3931ef36ef0c28ca105d5f0754f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 17:58:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 17:59:45.000000000","tz":540},"subject":"Use built-in StrEnum","message":"Use built-in StrEnum\n\nIt was added in Python 3.11 .\n\nChange-Id: I671c6310a54c47815eb25ce7b1b335949d30c3fd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/3e8f08a70a253b5bb2ce6f2381cf268a719adc83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/3e8f08a70a253b5bb2ce6f2381cf268a719adc83"}]},"branch":"refs/heads/master"},"0bb74a40086d3e91063d30f161053ceadb8653f0":{"kind":"REWORK","_number":2,"created":"2026-09-09 18:00:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/1004914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/14/1004914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/2"}}},"commit":{"parents":[{"commit":"8c978e8d46ba3931ef36ef0c28ca105d5f0754f4","subject":"Merge \"Remove duplicated optional dependencies from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8c978e8d46ba3931ef36ef0c28ca105d5f0754f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 17:58:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 18:00:30.000000000","tz":540},"subject":"Use built-in StrEnum","message":"Use built-in StrEnum\n\nIt was added in Python 3.11 .\n\nChange-Id: I671c6310a54c47815eb25ce7b1b335949d30c3fd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/0bb74a40086d3e91063d30f161053ceadb8653f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/0bb74a40086d3e91063d30f161053ceadb8653f0"}]},"branch":"refs/heads/master"},"fe890f8f71b8cd861900759ac915739c294f322c":{"kind":"REWORK","_number":3,"created":"2026-09-10 01:26:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/1004914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/14/1004914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/14/1004914/3"}}},"commit":{"parents":[{"commit":"8c978e8d46ba3931ef36ef0c28ca105d5f0754f4","subject":"Merge \"Remove duplicated optional dependencies from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8c978e8d46ba3931ef36ef0c28ca105d5f0754f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 17:58:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 01:26:15.000000000","tz":540},"subject":"Use built-in StrEnum","message":"Use built-in StrEnum\n\nIt was added in Python 3.11 .\n\nChange-Id: I671c6310a54c47815eb25ce7b1b335949d30c3fd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/fe890f8f71b8cd861900759ac915739c294f322c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/fe890f8f71b8cd861900759ac915739c294f322c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
