)]}'
{"id":"openstack%2Ftaskflow~1004953","triplet_id":"openstack%2Ftaskflow~master~I86f2a483a21a9b392cf382687c68e034ca5273c0","project":"openstack/taskflow","branch":"master","hashtags":[],"change_id":"I86f2a483a21a9b392cf382687c68e034ca5273c0","subject":"Remove old note about test requirements","status":"MERGED","created":"2026-09-10 02:27:34.000000000","updated":"2026-09-10 18:23:41.000000000","submitted":"2026-09-10 18:22:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004953","meta_rev_id":"37b712b01c655d2028e3e713198bbae1599aa56f","_number":1004953,"virtual_id_number":1004953,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 18:22:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-10 16:48:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-10 16:48:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 04:21:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ebe0bd4b2061cf45051ee0a09080786b1eb5397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 02:27:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"731fb6978fbc4001b3a1ee395f82ff3cf7bc8270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 04:21:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b838fdea12424ff08c8bda732b7b58e4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b248fed2bc594d829621c6369767d261 : SUCCESS in 4m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d285b157bca640ef954ad7149d8691d7 : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d29a87b8b8d54692854fc901684f34c9 : SUCCESS in 1h 23m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36007f540bd94da58eb8845be2874669 : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dceb683e5df642fc9924a76ff0bd730a : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8df7dd4335f24ed888b9f695231c0d25 : SUCCESS in 12m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0c5ece910474935bf07f0d51ae4b0b8 : SUCCESS in 8m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fcd7e9c220f74fdeb9401a551fa83fb0 : SUCCESS in 11m 57s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/be6c07e7f84345eba8ce47c6e5d66427 : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2057ea4143c04e0e9793d4bd4e2b236b : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cc85cedbbd64920a13884a2fedcb828 : SUCCESS in 3m 36s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/02a99e1586f04b6c97e451fd9d75093b : SUCCESS in 6m 15s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/07825c8d42a14989ba59137cd04d0631 : SUCCESS in 8m 36s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/4f47ac4169df47b8aea369db8079832a : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":1},{"id":"67f1a30ddf2ccc426dec3722e44b300dbb26addb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 16:48:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"313e4b053ae0fe4bca50a1544e4a215f53ec5e2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:48:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b091cd367486f6563a6392927cb24b712e3cce65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 18:22:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/114f705c593e44318192208c2e6c04d7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/274da1b7853746b0891e05834129f67c : SUCCESS in 4m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4009cc7e30df424e8e8dac4bf31b1ee7 : SUCCESS in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eedea16eaf79444992fe28a6b58c0292 : SUCCESS in 1h 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b36835d7e74f34a5e94337a7440bee : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d1f018a96b146ed9f8d30a24d925076 : SUCCESS in 10m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e702a3bcd1974bd9b3cdec24d0501249 : SUCCESS in 10m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d96dfe2ec69a4fdf8e84fb65c2421191 : SUCCESS in 13m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a02e816ad4cb48138337947d6179203b : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8e17c1f329c441bab667040ea5eb732 : SUCCESS in 2m 42s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/98a07703d3dc4dee8f6fc99c66a1489f : SUCCESS in 7m 26s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/eaa065eeb1e541fa814014575f95848f : SUCCESS in 5m 44s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/245ae0bfa9cb48528f9e4a8e19b9ab5b : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e156bb822868fea3000d55f39bc7eac05ae8ca40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 18:22:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"37b712b01c655d2028e3e713198bbae1599aa56f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 18:23:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aea03f37c41d417caf4a35a461cc93ea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49199acb57d44a0eac6e8a0072426ff5 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4bdb97504c3cf606489c5eeb7fa5c7f7d008329","revisions":{"e4bdb97504c3cf606489c5eeb7fa5c7f7d008329":{"kind":"REWORK","_number":1,"created":"2026-09-10 02:27:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/1004953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/53/1004953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/53/1004953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/53/1004953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/53/1004953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/53/1004953/1"}}},"commit":{"parents":[{"commit":"8c978e8d46ba3931ef36ef0c28ca105d5f0754f4","subject":"Merge \"Remove duplicated optional dependencies from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8c978e8d46ba3931ef36ef0c28ca105d5f0754f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 02:27:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 02:27:01.000000000","tz":540},"subject":"Remove old note about test requirements","message":"Remove old note about test requirements\n\nWe no longer need to maintain these in test requirements since\n2be4fa1a7742cecd5a864c931aadd8f99cba839c was merged.\n\nChange-Id: I86f2a483a21a9b392cf382687c68e034ca5273c0\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/e4bdb97504c3cf606489c5eeb7fa5c7f7d008329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/e4bdb97504c3cf606489c5eeb7fa5c7f7d008329"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
