)]}'
{"id":"openstack%2Ftaskflow~1004954","triplet_id":"openstack%2Ftaskflow~master~Ia8b2ebf15e400b9e80b611255213f19884b5b9c2","project":"openstack/taskflow","branch":"master","hashtags":[],"change_id":"Ia8b2ebf15e400b9e80b611255213f19884b5b9c2","subject":"Remove remaining ETCD_VERSION","status":"MERGED","created":"2026-09-10 02:29:21.000000000","updated":"2026-09-10 06:26:07.000000000","submitted":"2026-09-10 06:25:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004954","meta_rev_id":"c7dbcf0c0f34100ee82a6d7b697431e1aafaa827","_number":1004954,"virtual_id_number":1004954,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 06:25:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-10 04:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-10 04:52:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 04:13:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"652412ad5449bf73f96bff43a3bd81aa95981723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 02:29:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f01ad1af39331d2e4449b56388e9d7e4238a555c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 04:13:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/461924eed9ad431b86efbfd24bebee4f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fef3d971d5c43e0ab0dd03d1542045f : SUCCESS in 1h 12m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17e96299e1a347b6a0aada59752831c9 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dbd516e72454b5489971187e79c91b5 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/854114ef3eb1461fb10be7a3d0c81267 : SUCCESS in 11m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46d239c252fa4f71a0265f32c1dcaac4 : SUCCESS in 10m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/555117e13e4d4757a913e5f2ae44bfb7 : SUCCESS in 12m 30s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b4a189ad53db44db82bf36f189ad3ebc : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98023f14ecd547be81c3aaf82c269e77 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdb12376df5a49a0bf784d55ce5b7657 : SUCCESS in 5m 28s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/f76e05369c9e499cb792c897892957a9 : SUCCESS in 7m 56s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/051bf4eccba04380b194e72b6956cdfa : SUCCESS in 7m 15s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/a08bd4c5551d40f38378a273fe5ad90a : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"103a2bf596f77ecbf04a4552ee133853c196618f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 04:52:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"798ecb878069d1bf7ea09119ae62fcb2e69d3f1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 04:53:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"664fe7a4d3a10e5e259dd29e2ef006dfb39bb39c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 06:25:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13e9fee0723849cabee3831eeae27d69\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31235c2879ab45779a608d3b4131f7c6 : SUCCESS in 1h 26m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e53f20c5ba1246c2bcbf1d47e444b288 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f8094081294403fa173d01c7ea453cd : SUCCESS in 14m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0335580689745faa5c9bf4bd4c9b06a : SUCCESS in 10m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/88ab33bb032843558e51ae60975fec3f : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09bb8970dead464c8ff0de4a15e51c8c : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e43110b944243caa76ec4e88e1fd7ee : SUCCESS in 4m 55s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/e8324512e433431bb7ec21655ae77dee : SUCCESS in 6m 12s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/af58c0315e6b4dcbacbcb152252f9264 : SUCCESS in 6m 04s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/066ae4cb0d1f41868990ce5727bcdc7e : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2645fc961785540151e1d9b0869c0830897d9d59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 06:25:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c7dbcf0c0f34100ee82a6d7b697431e1aafaa827","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 06:26:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2792d441ecae4ea59f6ddbed1c42c937\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac2e7702f8442f796fff0e3e7c51f7e : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc71fc1187b44c3382c85d991fd28668 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc1d7c9ecd011ca791b2f04c005469aa0cfa5ed5","revisions":{"bc1d7c9ecd011ca791b2f04c005469aa0cfa5ed5":{"kind":"REWORK","_number":1,"created":"2026-09-10 02:29:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/1004954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/54/1004954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/54/1004954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/54/1004954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/54/1004954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/54/1004954/1"}}},"commit":{"parents":[{"commit":"8c978e8d46ba3931ef36ef0c28ca105d5f0754f4","subject":"Merge \"Remove duplicated optional dependencies from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8c978e8d46ba3931ef36ef0c28ca105d5f0754f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 02:28:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 02:29:14.000000000","tz":540},"subject":"Remove remaining ETCD_VERSION","message":"Remove remaining ETCD_VERSION\n\nIt has been unused since 0f27caeec138ec5b24021ac612577b596b48a699 was\nmerged.\n\nChange-Id: Ia8b2ebf15e400b9e80b611255213f19884b5b9c2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/bc1d7c9ecd011ca791b2f04c005469aa0cfa5ed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/bc1d7c9ecd011ca791b2f04c005469aa0cfa5ed5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
