)]}'
{"id":"openstack%2Ftaskflow~1004967","triplet_id":"openstack%2Ftaskflow~master~I209d7bf02fc3ae96d45218e9b578c1cb1a0d78f5","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-10 16:30:40.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 16:48:04.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I209d7bf02fc3ae96d45218e9b578c1cb1a0d78f5","subject":"Use close method from base class","status":"MERGED","created":"2026-09-10 07:18:30.000000000","updated":"2026-09-15 05:13:10.000000000","submitted":"2026-09-15 05:12:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004967","meta_rev_id":"6381cc8f11681f8ccd2639fdf3c1c44eadbb9eda","_number":1004967,"virtual_id_number":1004967,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 05:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-10 16:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-09-15 03:32:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 09:01:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 16:05:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-09-10 16:30:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"04b27b10895ed9fcadb85f5fe9334210727f5974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 07:18:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f031a691bd4a518e6bfb10d0565953260b13af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:01:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f7f65a6cc9c494295c272efb09dbd2a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ab4908bf398464f85b7cc01d0d13e1a : SUCCESS in 1h 37m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4743456d16dd465e8f1d95b94297fb9b : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f524eca79b4bd8b1dba1b0afc6c9e4 : FAILURE in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66d8bea0ad6e4ab5ab3842a714e3f875 : SUCCESS in 9m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f11bf4492a54cdb854a1b50deba3387 : SUCCESS in 13m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd692a972b234214a9adeb04dfbb6179 : SUCCESS in 12m 34s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b3029b162e8f437bac1806a1a2e3b05e : SUCCESS in 16m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef643387682441a3b4009bb29dd399e1 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ce46f9e9baa4a4c98dbc79a845c3d9b : SUCCESS in 4m 36s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/50459b1a9faa439cad05ab83a986df62 : SUCCESS in 11m 37s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/59c560ddd8084497ab96cf7a6d03fd96 : SUCCESS in 8m 41s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/203bb1e315404ff6b031caf23c1cc42b : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ece68ee58820a717aaf3a2d8ce2828ab00b494eb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-10 16:05:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff3128ba3fc6d7f4fd48e60d782338f6dead2fa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 16:16:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"990144053354f19f09e7ef4a5ea8f251ec9e92fa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 16:18:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"decbf3c5e1dfa1b98fcea50fe18fc7ddf09246cf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-10 16:30:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d33ebc2c4aaf874b6cc434e04a5f843bf21a07b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-10 16:30:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"32d577153f71675301561956e509b7d317a4aa55","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-10 16:32:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4cbdb383b221fdc582a0a39e9055b975f559c90e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 16:48:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d5393369dbb9b1ed6915d2acbdb64a8dbaadbfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 17:50:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec6814a84d0a493da417a797b0ef21e3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0837fac9b0594a5bbc9c6c00c4807ef4 : SUCCESS in 1h 29m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d380b96fbce452e8641ecd7ac6559a7 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae610b442814c6aae54ba2fd4546673 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd07814440684e9dab92c72619c595b0 : SUCCESS in 10m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c76f499c2bf746e4b79cb8c064dc4a9f : SUCCESS in 11m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b0abeb1e00cc43bdb1f27e35097a5382 : SUCCESS in 12m 46s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/13fd59b7e8d74f35be4a7c9d0a1c9d65 : SUCCESS in 13m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88638451ea6646e59ad1a9ed99541802 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63fa852072ae4cd38e658aec04aaead3 : SUCCESS in 5m 30s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/635ccf2f6f28400c8947a36af1e39136 : SUCCESS in 7m 23s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/58e69577ba3c4192b281248145bc65fe : SUCCESS in 7m 22s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/11f77db839ce4c62a6ca288221be1993 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"e3035f3ec8b24d8a693ab1dc42d626e5c7fedd6c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-15 03:32:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef9844453c5c19c3179e98753f9ce25c4ecc01b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 03:32:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7efa2a0c8b9dbc7f9cb3613f84ee2963d3503bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 05:12:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b11892a77d4e298345a3e71ce6d61c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7e51ded45d14e5aab64169501e89a7a : SUCCESS in 1h 34m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78786408cb6f45aea0dd513aa801142b : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02c965df0ea34ce8bd89e20c18791e9e : SUCCESS in 10m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8698be5ec5648e482e76ec0e3fe441b : SUCCESS in 10m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8e9a0ad5d5854cf6a52665aa23c52d41 : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e967e6f08ac142aa80505058f382f085 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e38a5f7a8b447a69826789d596b8497 : SUCCESS in 4m 28s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/ef343e843abb420a8eda4233a1af3780 : SUCCESS in 5m 58s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/851c1467f4d04ea0af5e0a7b72eee6e9 : SUCCESS in 4m 54s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/2134268299804459a32d0b118d93e69c : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"237ce787065e6b6cd205cd7bd4e68ba967e4eb0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 05:12:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6381cc8f11681f8ccd2639fdf3c1c44eadbb9eda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 05:13:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f077a1403a54d5b887b82c9f4c2bbe9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce064ea0a024b3a9f688b5c1d214362 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e7ddaafda2dbc2bf16e1584efed43d94fc255157","revisions":{"01daa3ffd99740c4b0bd1c4db8dfee69d578a422":{"kind":"REWORK","_number":1,"created":"2026-09-10 07:18:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/1004967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/67/1004967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/67/1004967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/67/1004967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/67/1004967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/67/1004967/1"}}},"commit":{"parents":[{"commit":"bc1d7c9ecd011ca791b2f04c005469aa0cfa5ed5","subject":"Remove remaining ETCD_VERSION","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/bc1d7c9ecd011ca791b2f04c005469aa0cfa5ed5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 07:15:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 07:17:51.000000000","tz":540},"subject":"Use close method from base class","message":"Use close method from base class\n\nRedis client provides its own close method to close its connection or\nconnection pool.\n\nAlso the mentioned proposal was rejected by the redis-py community\nand may not be accepted in near future.\n\n[1] https://github.com/redis/redis-py/commit/0984b10226\n\nChange-Id: I209d7bf02fc3ae96d45218e9b578c1cb1a0d78f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/01daa3ffd99740c4b0bd1c4db8dfee69d578a422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/01daa3ffd99740c4b0bd1c4db8dfee69d578a422"}]},"branch":"refs/heads/master"},"e7ddaafda2dbc2bf16e1584efed43d94fc255157":{"kind":"REWORK","_number":2,"created":"2026-09-10 16:16:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/1004967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/67/1004967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/67/1004967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/67/1004967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/67/1004967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/67/1004967/2"}}},"commit":{"parents":[{"commit":"c34aeff03d9ad208856ab388a49bb6969538e1e0","subject":"Merge \"Remove remaining ETCD_VERSION\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/c34aeff03d9ad208856ab388a49bb6969538e1e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 07:15:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 16:16:29.000000000","tz":540},"subject":"Use close method from base class","message":"Use close method from base class\n\nRedis client provides its own close method to close its connection or\nconnection pool.\n\nAlso the mentioned proposal was rejected by the redis-py community\nand may not be accepted in near future.\n\n[1] https://github.com/redis/redis-py/commit/0984b10226\n\nChange-Id: I209d7bf02fc3ae96d45218e9b578c1cb1a0d78f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/e7ddaafda2dbc2bf16e1584efed43d94fc255157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/e7ddaafda2dbc2bf16e1584efed43d94fc255157"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
