)]}'
{"id":"openstack%2Ftaskflow~1005095","triplet_id":"openstack%2Ftaskflow~master~I166cbca0379c89bf43abfdc3a328dd2c450f4170","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-15 22:58:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I166cbca0379c89bf43abfdc3a328dd2c450f4170","subject":"redis: Deprecate retry_on_timeout","status":"MERGED","created":"2026-09-10 17:28:03.000000000","updated":"2026-09-15 22:59:30.000000000","submitted":"2026-09-15 22:58:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005095","meta_rev_id":"eb453beb0667f9d7614313589201310cd53ec9ca","_number":1005095,"virtual_id_number":1005095,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 22:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 20:03:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-15 21:09:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 18:57:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 20:03:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"06e24ed54da3da133e46374eb1dca914c6998cdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 17:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee5e125730fa730983340f2d149100d4360c31b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 17:29:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"723b728a6280784d62ebfa7664236c05787a8886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 18:57:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea0fc04f2aee4ebc834d46fc0118345c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a6ce0f50947463fa9bcfb1cf65cc7e4 : SUCCESS in 2m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b97efda3e9d94a13b4e4f7c53f7c97bd : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8296cd13088640d0bcf4d517f9ed0fd4 : SUCCESS in 1h 21m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a68a44e80223478992009de5fb07556f : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0aff5d1a91d4e61adfe987aef18c4d4 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96715cc77d8646c3ac641c0f2e49d602 : SUCCESS in 8m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d55a09c983b1476d8f7b5774f829a44c : SUCCESS in 8m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5c787662134c4d49b2db3b98e9898207 : SUCCESS in 13m 08s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/1415baf32e364028b07a1c3609d7c1f1 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fe96829a1b491f91e1fc7faea39b6b : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11bea99e359041b4a858491d24f229c8 : SUCCESS in 4m 51s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/65801fe336e547fabe58de7d7b828acf : SUCCESS in 7m 36s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/26d3199594544e5298dd52942328dce1 : SUCCESS in 5m 35s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/396ff7c53e714395bc97f62da6bbac51 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"7769b065ca595c711a5390ad565d13c63c5afbb0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 20:03:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3fa4132838d68622875add34f1ce2f29b8f8db2d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 21:09:57.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b67ae8526449165da8fd51910e27450f79ad7cbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 21:10:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7fc90af95f068c4722ec9001453b3d4c6e67fb50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:58:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d79de7e551848eeb6eeeb61b354ac57\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/89e23cbfb6054e72baac1897defdb6bb : SUCCESS in 2m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/12fcf3c1df924369bb8c0a6753eecc2d : SUCCESS in 3m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c54408c93fe464f852ee1d790ad8d43 : SUCCESS in 1h 42m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89bb96cafd7f4959b736be42ab6a7657 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb7f921a9eb1475fa1af120fcde1d6fe : SUCCESS in 10m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1a7835c49004436812db3d8b0d57b64 : SUCCESS in 10m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c912ceafb8eb405bbc162f0acc617fcf : SUCCESS in 12m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bba7a0e02d348f2bff4f16c72e4ede4 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10df970cc92d46a8980af8aa0f214af2 : SUCCESS in 5m 43s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/c42c90828678433cbaee3c54723adae9 : SUCCESS in 7m 21s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/74133baf3dff420f8eab7582c64c7a78 : SUCCESS in 8m 32s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/26490813d1e04ee09c4ba18e907d80fa : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a94e93735de3bb6d85bd453c41ff860ebc136507","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:58:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"eb453beb0667f9d7614313589201310cd53ec9ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:59:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34b03f3fc0a449918dd7c2d081c36369\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3282f2dcf143fba4210a0f823a15bc : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f368c67d572140218e42b57a8ee8c188 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94956140ab48ed296adac6490d61e3d96aa09cee","revisions":{"76fee598b81cddc67d7d4b5ff11599b0c7ea835d":{"kind":"REWORK","_number":1,"created":"2026-09-10 17:28:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/1005095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/95/1005095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/95/1005095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/95/1005095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/95/1005095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/95/1005095/1"}}},"commit":{"parents":[{"commit":"c34aeff03d9ad208856ab388a49bb6969538e1e0","subject":"Merge \"Remove remaining ETCD_VERSION\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/c34aeff03d9ad208856ab388a49bb6969538e1e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 17:23:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 17:27:55.000000000","tz":540},"subject":"redis: Deprecate retry_on_timeout","message":"redis: Deprecate retry_on_timeout\n\nThis option in redis-py was deprecated in redis-py 6.0.0 and now redis\nclients retry operations on timeout by default[1].\n\n[1] https://github.com/redis/redis-py/commit/f32a1ca4f6\n\nNote that warning message is not implemented in the driver because\nthe same warning is already implemented in redis-py.\n\nChange-Id: I166cbca0379c89bf43abfdc3a328dd2c450f4170\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/76fee598b81cddc67d7d4b5ff11599b0c7ea835d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/76fee598b81cddc67d7d4b5ff11599b0c7ea835d"}]},"branch":"refs/heads/master"},"94956140ab48ed296adac6490d61e3d96aa09cee":{"kind":"REWORK","_number":2,"created":"2026-09-10 17:29:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/1005095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/95/1005095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/95/1005095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/95/1005095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/95/1005095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/95/1005095/2"}}},"commit":{"parents":[{"commit":"c34aeff03d9ad208856ab388a49bb6969538e1e0","subject":"Merge \"Remove remaining ETCD_VERSION\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/c34aeff03d9ad208856ab388a49bb6969538e1e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 17:23:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 17:29:45.000000000","tz":540},"subject":"redis: Deprecate retry_on_timeout","message":"redis: Deprecate retry_on_timeout\n\nThis option in redis-py was deprecated in redis-py 6.0.0 and now redis\nclients retry operations on timeout by default[1].\n\n[1] https://github.com/redis/redis-py/commit/f32a1ca4f6\n\nNote that warning message is not implemented in the driver because\nthe same warning is already implemented in redis-py.\n\nChange-Id: I166cbca0379c89bf43abfdc3a328dd2c450f4170\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/94956140ab48ed296adac6490d61e3d96aa09cee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/94956140ab48ed296adac6490d61e3d96aa09cee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
